)]}'
{"id":"openstack%2Fopenstacksdk~929560","triplet_id":"openstack%2Fopenstacksdk~master~Ief5578335cb1a40bc2ac7e8329ed2c11ba5c9ed0","project":"openstack/openstacksdk","branch":"master","topic":"quota-functional-tests","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-09-20 08:51:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief5578335cb1a40bc2ac7e8329ed2c11ba5c9ed0","subject":"tests: Add helpers to create, delete temp projects","status":"MERGED","created":"2024-09-16 17:08:04.000000000","updated":"2024-09-20 08:53:04.000000000","submitted":"2024-09-20 08:51:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929560-quota-functional-tests","meta_rev_id":"dbf484e674b8e4f350c339ae3c7ca66763a8776c","_number":929560,"virtual_id_number":929560,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 08:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-09-17 04:56:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-09-20 07:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-20 07:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-09-20 07:29:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 18:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 04:56:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-09-20 07:06:56.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-09-20 07:29:22.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"a69aee29e70ab5c5794c47f906ee98011ee4c23d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-16 17:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8de97b1ec0b644d20215d8f4c6a993c3e2ea43f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 18:08:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5176de01de464b8e8963fab23ce111ea\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3636465aaca4470fb8884e7b7cdf07e3 : SUCCESS in 48m 11s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/0647693d75934fc1b2722011ccc0a6e1 : SUCCESS in 38m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb65f2a3a6424e1a8781a5bebb369b2e : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/374cc28f840443a0a536c52e00a015c1 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb4aa9afa51049dca1848d433811aeed : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af04fd7dd2614f2fade33a91347f641b : SUCCESS in 5m 02s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/34d87f940ba54ef680b7296aa3604358 : SUCCESS in 26m 50s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f73b72921a7c454e9df365bd738063b4 : SUCCESS in 3m 37s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/20606b1152c1485682d5773b3f6633ee : SUCCESS in 7m 13s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/fa41ad80878e425482fcdada351c841d : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65201f4f6f941629ecad2a6dacce771 : SUCCESS in 9m 08s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0807801ff74c447bab96d338dc39da82 : SUCCESS in 31m 50s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/48775169b1f2487a9e053df39c90bd9a : SUCCESS in 10m 13s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/69bd6aee5b4d44319bc28b905f11abe8 : FAILURE in 17m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1257e93f8144c7c96afd259c01a6e7d : SUCCESS in 18m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/213f22327a8044769467b8e3bf11a0ab : SUCCESS in 19m 54s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/59fc726a37db4b1081d31c6513a42dbe : SUCCESS in 36m 47s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2de5fb2f7a924be3bc1495ba96dd9251 : SUCCESS in 19m 36s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/6aeb41f27a244fbeb4f873d1f34cdabe : SUCCESS in 15m 55s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/201cbb69390342f68746440c7d7ae573 : POST_FAILURE in 16m 27s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/66c283ef5c2a4b1f96d74b9197939749 : SUCCESS in 35m 49s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b816c40f6bf24686b9232d013f27f344 : SUCCESS in 53m 08s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/91398ee6e2d347e29dbd163b4688aa28 : FAILURE in 51m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"854cdf2b1f4a21bb02a822caa89ce6c19e225b97","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-17 04:56:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2a8c8bf34daaee2cf837fd6e74d7d80ea50279f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-09-20 07:06:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9802922d0a59bbb12482d5f3475e3f3aa64c205b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-20 07:29:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c3d064777fd790829e5d9fedf61fe1b92aab24a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 07:30:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1696be439b68d56a76ea71aba2838c2cfaf879b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 08:51:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e7f80172fc4d1c86b3b1fc8886a4fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea966308f7b4e22bbea44142e0d5a47 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3495a3ab3c294053bf9254e0dcef0ea1 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70e48320cb5747e4a33eb6a374506d45 : SUCCESS in 5m 35s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b4b44e7aa01c4484964a34e03738820f : SUCCESS in 33m 40s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5aac9786e5164fac8dc3d1f24d3bc55e : SUCCESS in 3m 36s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/82fb21f154a04cb9a16d2b57063ac1df : SUCCESS in 6m 24s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/8b5ff09d73384513836956f18037b4ae : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6196efbf6454563a56bf0388f10c40f : SUCCESS in 8m 06s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6b7e00c6bef845e7809e7e8d9b325146 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06c2bca937164abb976f3d9da8142469 : SUCCESS in 26m 39s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/89434ef52ea84b6fbb395f4c1277ae53 : SUCCESS in 33m 23s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/96425838a8c54476bf8ae00a2fef6ab7 : SUCCESS in 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e6e8eef94164a60ba8c0d91b75bffe62aabd6d98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 08:51:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dbf484e674b8e4f350c339ae3c7ca66763a8776c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 08:53:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2e17c5fa6c8418abbb335cf49909cfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc7a9ac90b74a28ab39185d98fa4847 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7f14b09d71783521dad565b7bbd37e732063753","revisions":{"e7f14b09d71783521dad565b7bbd37e732063753":{"kind":"REWORK","_number":1,"created":"2024-09-16 17:08:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/929560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/60/929560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/60/929560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/60/929560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/60/929560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/60/929560/1"}}},"commit":{"parents":[{"commit":"1523ceb2e64f3ef2da67bdc51ebf47ac11e6bd51","subject":"Merge \"Fix volume summary passing non-existent all_projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1523ceb2e64f3ef2da67bdc51ebf47ac11e6bd51"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-16 10:27:47.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-16 17:07:48.000000000","tz":60},"subject":"tests: Add helpers to create, delete temp projects","message":"tests: Add helpers to create, delete temp projects\n\nThis should help us avoid quota issues going forward.\n\nChange-Id: Ief5578335cb1a40bc2ac7e8329ed2c11ba5c9ed0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e7f14b09d71783521dad565b7bbd37e732063753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e7f14b09d71783521dad565b7bbd37e732063753"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
