)]}'
{"id":"openstack%2Fopenstacksdk~934226","triplet_id":"openstack%2Fopenstacksdk~master~Ic5248c57ac2e83a522426e9a2dc4a6b8ceb89d67","project":"openstack/openstacksdk","branch":"master","topic":"identity-gaps","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-11-26 19:32:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic5248c57ac2e83a522426e9a2dc4a6b8ceb89d67","subject":"identity: Add functional tests for projects","status":"MERGED","created":"2024-11-06 11:06:34.000000000","updated":"2024-11-26 19:32:32.000000000","submitted":"2024-11-26 19:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934226-identity-gaps","meta_rev_id":"1074bd10e50e4712e4f15bbd2bd44c8767152431","_number":934226,"virtual_id_number":934226,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-26 19:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2024-11-26 16:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2024-11-26 16:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-06 12:46:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 16:05:03.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"ca4e488a71d761e3bff8b66ad491a3612dec93b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-11-06 11:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ef7e970faa64d0f08fadb78943a5af2fe029d14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 12:46:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4899c3256b374114b362f1eb87710c12\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a05154a4e672405c8e4d5107961ce8c2 : SUCCESS in 50m 01s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b6e1b55fd4a246f69dc3e577ff738876 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1772c55a4b6d42e2a316939e4f4fd0f7 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b673aacd5a6f4ae19ddbc1f148b9f3c7 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9289814ea104babbb510f8d26d7c10b : SUCCESS in 5m 43s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/75985cc62ee34dcba1ddf2458a55acb8 : SUCCESS in 53m 15s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/701932da8e0e46028616a12031af62bb : SUCCESS in 3m 27s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/31fa73ce8d4e49b88feaab4e9928dc16 : SUCCESS in 4m 52s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/261a2f733007463fa87e8a2a81f9cf26 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b766c4ad844ec3b9c98fad4e514b3f : SUCCESS in 9m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/de7d39c1c4174be3a86b4ecb49a2fb7f : SUCCESS in 56m 51s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/1a346daf653c49968fea57b71e3ecea2 : SUCCESS in 16m 09s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/f746aabc5aa6483a860d55ae7b12fe60 : SUCCESS in 34m 29s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25fd3c70e2f44e409ce22a5ae5b1c6bc : SUCCESS in 1h 00m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/25fd86d9667048a88b3bf94c75738f52 : SUCCESS in 1h 17m 47s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d881e5db1a0944ba88d4c1f2655f0b3e : SUCCESS in 1h 31m 14s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/cf6fea9d3c6b4d9ba7eedcc582675479 : FAILURE in 50m 52s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/37e3ac1b57c845038f03469523995428 : SUCCESS in 24m 19s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/3a8d4320561d411387b995872afa8259 : POST_FAILURE in 29m 07s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/19dc91c9ae104441a8487d0f24ad08fc : FAILURE in 29m 26s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fc5184d51c884be1a35edd1c5cf7323d : SUCCESS in 54m 25s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0b61ec64b45f47868b15e24b6e970281 : SUCCESS in 1h 05m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c0a3614e8d52780388c78c4fc1335ac1b2947c3","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-26 16:05:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86ec2b640fa22f852cccf085cbde3228d9d91edb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:06:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b3649d19d2528acbcb6e1ba71098cb3c772c380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:32:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91e307e3efc742b7aaf4a6715577c145\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfeeb3d6741848ca9989e6f87aa9f4bf : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/befc16d1c1af4fab861d0edd90836ed7 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06cddcc3915349e8bc4308338e1714ad : SUCCESS in 4m 25s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/43ee07ebdb74485d9d28da6178ffe118 : SUCCESS in 1h 03m 48s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/365ef417da0242d1bce7e6eb7bc2011f : SUCCESS in 3m 50s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ebb7592651c440ac81041a4b169eb3df : SUCCESS in 4m 47s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/9fc2996ea21c4a23817e3360ce6cba71 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1aea7d064c841fd83c11d318857a8a0 : SUCCESS in 9m 21s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4381370973634d06b2c2d72222eaf973 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db5c78fc4d58400a88bd1d9ae5ef1707 : SUCCESS in 1h 09m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/da528a7b881049c9b66933f3021162ee : SUCCESS in 1h 26m 50s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/55c4d0dcdcae4f25941b8d5d96b378dd : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1074bd10e50e4712e4f15bbd2bd44c8767152431","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 19:32:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"677192bf82319380d8b5eff34f2ca6470f2fad9d","revisions":{"677192bf82319380d8b5eff34f2ca6470f2fad9d":{"kind":"REWORK","_number":1,"created":"2024-11-06 11:06:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/934226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/934226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/934226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/934226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/934226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/934226/1"}}},"commit":{"parents":[{"commit":"d5ba0c42a9d2ed489aca3a6e431da59be3cf39c8","subject":"Merge \"compute: Add locked filter for Server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d5ba0c42a9d2ed489aca3a6e431da59be3cf39c8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-11-06 10:38:36.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-11-06 10:55:04.000000000","tz":0},"subject":"identity: Add functional tests for projects","message":"identity: Add functional tests for projects\n\nChange-Id: Ic5248c57ac2e83a522426e9a2dc4a6b8ceb89d67\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/677192bf82319380d8b5eff34f2ca6470f2fad9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/677192bf82319380d8b5eff34f2ca6470f2fad9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
