)]}'
{"id":"openstack%2Fopenstacksdk~934364","triplet_id":"openstack%2Fopenstacksdk~master~I9aa39810fe94f7ee9b68d34050f4adb9dbdfccb8","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"35548":{"account":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"last_update":"2024-12-16 19:52:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9aa39810fe94f7ee9b68d34050f4adb9dbdfccb8","subject":"identity: Add support for project endpoints","status":"MERGED","created":"2024-11-07 19:08:46.000000000","updated":"2024-12-16 19:53:34.000000000","submitted":"2024-12-16 19:52:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934364","meta_rev_id":"0de3a809bb6a63a08d8560407f57ce45075c62e7","_number":934364,"virtual_id_number":934364,"owner":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-16 19:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-12-16 18:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-12-16 18:30:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-07 20:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-16 18:30:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d53f22101a4770f3dbb097a3286c60b66833a647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-11-07 19:08:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abf7d21d18b3b2a90993c677a5d0088eab9cf77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 20:42:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19555d65d8bc47558ebf7092fc0437ab\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/731d67bb831a4ef5a4e735d3e1c45a0e : SUCCESS in 40m 36s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a155b026887b4ec5988828576144b8ae : SUCCESS in 31m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/257a7df8a40847d997a0f94b4fd2f902 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f721dfa6c1eb4d36af5e8e4daad63d72 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/820c8849f777426ab8d18d2f73362ef8 : SUCCESS in 4m 35s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/890da8288ab44094a8d0575a8bdfbdfa : SUCCESS in 1h 09m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1c65e88271394185bcecc2695b81b97b : SUCCESS in 3m 52s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0f097f5a984a4dd88c54714d82c7aad7 : SUCCESS in 5m 14s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/7538da7c78bc45519b1410dcde83bc9d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393493beb14c41caae960c29d71e1f7d : FAILURE in 4m 39s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ff91f40c03864d77818b8d288c3220c5 : SUCCESS in 32m 31s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/9575cd59db95490fa8c4da6ac6c7a148 : SUCCESS in 8m 12s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/ea4df8f56e574548b9c419319f91c3a2 : SUCCESS in 20m 30s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59e2b5cd84f641c0bb42414b732dd980 : SUCCESS in 1h 04m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bd8ce5d67b7b4525adca586d85101e9b : SUCCESS in 1h 29m 34s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e6253142dedc46e6ab5acd03c9e04ccd : POST_FAILURE in 22m 19s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/972848cd503f4d7e8996ae81851c8818 : FAILURE in 58m 10s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/29b3e933b69d45ccbe87d47de04ef73d : SUCCESS in 31m 58s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/fb64c6b1f6f74a2da5be122637c94ff0 : POST_FAILURE in 22m 40s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/00ca9daddb4b4200a4ebcb0687a615a2 : FAILURE in 40m 02s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0af741c70bf34f74b62c674fd45afcbd : SUCCESS in 1h 05m 37s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/461d376a185941229eba7e55fa974eb8 : SUCCESS in 1h 06m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45643b7aa907c9b6ff3fe39273bfb9b8eb7d0f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"date":"2024-12-02 23:47:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"85f580aef734f03d16f36c23d0344c80fa782ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 01:07:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0404f2fc30449a99e507aafa70c862\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/8b9748ed42a14a549ea308b0d05d6dc7 : SUCCESS in 38m 52s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/8e63e1d75fbb492b831094c6f42b83a6 : SUCCESS in 33m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e216e4c9f434ec9a0b3fb6b1a77eb05 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d6a39a855504591b55cc61a900f2bb4 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0da43c5b62ed47f6819d37783b3388e2 : SUCCESS in 3m 16s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/30a7f1a2501345ec91c7c468dbe3c6dd : SUCCESS in 1h 03m 42s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ce491641cbf445c7975092f86a7e247c : SUCCESS in 3m 55s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3b3302484cc24b55aaae6368671e38b3 : SUCCESS in 2m 58s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/ea2da16a46a64c2a8e7512a7b3f624c5 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa81507aeacb440e8a340dde0d7474c2 : SUCCESS in 5m 40s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/38ee26edd0634f7c818e9177df3bb30d : SUCCESS in 46m 13s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/57ed1c8c7ff04c12a5eb79059ff8f46c : SUCCESS in 8m 10s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/3d3c6561db5542e58b73ed786ec7a62d : SUCCESS in 18m 14s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/554bbfc5b52b4078a7df34abb936767f : SUCCESS in 40m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/73320aef1c0c4063a0111583dce3dbc8 : SUCCESS in 1h 02m 46s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/800ec0bf6c5b4fac9d398dfaaef32556 : SUCCESS in 1h 12m 41s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9162ae70ac0c4d7cbcb900ca72c5d5e4 : FAILURE in 43m 16s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/040f479487ba41c78e58c2b31fb6e314 : SUCCESS in 23m 42s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b956fc8462634a9cadc891f8565932f4 : POST_FAILURE in 29m 31s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d5abbc72380c488e981843dd3568f6e8 : FAILURE in 41m 37s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/be704c27c66b4513b60a73d5e367ae61 : SUCCESS in 1h 13m 57s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/21fe15afb6da491b8a37d2040db4ff8c : FAILURE in 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47661d9f0613a754bf97337a2cd1113ef9da4420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-12-11 15:04:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_35548\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"}],"_revision_number":3},{"id":"4da69f6e4d66135c63d832915a3c9bc440d3d079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 17:12:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a21894270d44a16a8e1fb15f33d92ac\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1f050bcb43b34762acd10f2dc210b9c0 : SUCCESS in 52m 17s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/955f11db8051475995ebf49d0301cd2d : SUCCESS in 37m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25244afec1ed4316bb774421f01dd375 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6867372d0ce3480a9dd80efa21fff209 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eaafd56fb16b48a48a5029be1da9a33b : SUCCESS in 2m 55s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dee624aaab2a4453aae9d8aac336efac : SUCCESS in 1h 07m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5a35402228b447dea7741979728c57e1 : SUCCESS in 3m 32s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/21fb53f12b7f4025997af52395abdc09 : SUCCESS in 4m 22s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/badd1fe2f94e4356b8fd926dd5af5d30 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85fc8c02d4b41afb76f362b13eeb3ef : SUCCESS in 8m 56s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5a362d9331594b628ac0e8e14c4b081d : SUCCESS in 1h 15m 23s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/f97282a2dba64b6290adec9a0adc9b2c : SUCCESS in 13m 26s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/ba0e9fb6bdb34a0a9e4c72ab5c8703ce : SUCCESS in 32m 08s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d8aafb3e79c49798895e89adb14ec3a : SUCCESS in 1h 06m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/840a9681a03d4c6db37dd4982d53fc93 : SUCCESS in 1h 29m 16s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/cd0bdb39466443c2b4930bf939b2acab : SUCCESS in 1h 33m 03s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9bd43fbf0c884050997929ed6c5f8258 : FAILURE in 1h 02m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/5074bb9014e144a983301e574b4057e1 : SUCCESS in 25m 54s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1e9ddc99dd3e44b7bbaf985132df0272 : POST_FAILURE in 31m 19s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8ed67d5b20d54c67b99bae2ae5cd7502 : FAILURE in 38m 42s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9d19b94b7b0043b4964ca7fa5d135fcb : SUCCESS in 1h 19m 12s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b526b0703b9c477ab977a984e6b37e32 : SUCCESS in 1h 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d06bbb83f9beb43ef04f02e9a86a2603b85c93a0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-12-16 18:30:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7747973c4dd500e07e3e1b9d8782d1717b47c356","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-12-16 18:30:52.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c5e3b19923adbb92df66ff8bf453ad58d3d67785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 18:31:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3c32bedeb844b3ee579f1b740bc631559e8b7ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 19:52:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a948c99261242ad8936a335f8fae977\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9571b8e1d6432584f103131f863149 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b64cdddc18b405fbacb082f7d1fedd5 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d189191dac24a3b94943f67da645f38 : SUCCESS in 2m 57s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a8fbd07bef514982a5ad26f0b128d75f : SUCCESS in 51m 00s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c8b14adc546f4b7080e299fca180a097 : SUCCESS in 2m 58s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/361db4eb45814504896367248fbd0aaf : SUCCESS in 4m 25s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/f3da3a7e2edf49b2a54ac38820ad4a37 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0494f0df667a47d2a7eca446f51500d2 : SUCCESS in 8m 02s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/89827b6538694fe3870e9d89f7131d26 : SUCCESS in 2m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f047bc4ee3a45279d4e4b25292a286c : SUCCESS in 51m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cecbffefdbd44cafbeef3c2946251333 : SUCCESS in 1h 15m 50s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4a78887c2280482596d8986377d6b07b : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":3},{"id":"9d94ec2ca47887a96bcef24e51457c3485cb9a71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 19:52:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0de3a809bb6a63a08d8560407f57ce45075c62e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 19:53:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/153c5628c93d4cc0a99794e8a5b09e91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c60fd20e7f4c0883890385ea043b45 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"371efde4f0b83d4e92d13a16f888b9a33784b8e0","revisions":{"f11b38538358e8f2cdd343f0f947596561556781":{"kind":"REWORK","_number":1,"created":"2024-11-07 19:08:46.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/64/934364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/64/934364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/1"}}},"commit":{"parents":[{"commit":"3faf2cacb6b7aa67391a589f2325fb6dcb79832f","subject":"Merge \"Add functional test for tags in Neutron resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3faf2cacb6b7aa67391a589f2325fb6dcb79832f"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-11-07 19:07:27.000000000","tz":0},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-11-07 19:07:27.000000000","tz":0},"subject":"identity: Add support for project endpoints","message":"identity: Add support for project endpoints\n\nChange-Id: I9aa39810fe94f7ee9b68d34050f4adb9dbdfccb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f11b38538358e8f2cdd343f0f947596561556781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f11b38538358e8f2cdd343f0f947596561556781"}]},"branch":"refs/heads/master"},"162a8ebcb1a02b1cc03e0dbb9ff59c5512a581da":{"kind":"REWORK","_number":2,"created":"2024-12-02 23:47:45.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"ref":"refs/changes/64/934364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/64/934364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/2"}}},"commit":{"parents":[{"commit":"3faf2cacb6b7aa67391a589f2325fb6dcb79832f","subject":"Merge \"Add functional test for tags in Neutron resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3faf2cacb6b7aa67391a589f2325fb6dcb79832f"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-11-07 19:07:27.000000000","tz":0},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-12-02 23:47:38.000000000","tz":0},"subject":"identity: Add support for project endpoints","message":"identity: Add support for project endpoints\n\nChange-Id: I9aa39810fe94f7ee9b68d34050f4adb9dbdfccb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/162a8ebcb1a02b1cc03e0dbb9ff59c5512a581da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/162a8ebcb1a02b1cc03e0dbb9ff59c5512a581da"}]},"branch":"refs/heads/master"},"371efde4f0b83d4e92d13a16f888b9a33784b8e0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-11 15:04:26.000000000","uploader":{"_account_id":35548,"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","username":"antoniagaete"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/934364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/64/934364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/64/934364/3"}}},"commit":{"parents":[{"commit":"16f3e62dc4e77543d433af5992309006ec298cfd","subject":"Merge \"pre-commit: Migrate pyupgrade to ruff-format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/16f3e62dc4e77543d433af5992309006ec298cfd"}]}],"author":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-11-07 19:07:27.000000000","tz":0},"committer":{"name":"Antonia Gaete","email":"antoniagaete@osuosl.org","date":"2024-12-11 15:04:26.000000000","tz":0},"subject":"identity: Add support for project endpoints","message":"identity: Add support for project endpoints\n\nChange-Id: I9aa39810fe94f7ee9b68d34050f4adb9dbdfccb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/371efde4f0b83d4e92d13a16f888b9a33784b8e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/371efde4f0b83d4e92d13a16f888b9a33784b8e0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
