)]}'
{"id":"openstack%2Fopenstacksdk~939703","triplet_id":"openstack%2Fopenstacksdk~master~I58e9a8bde3dc9425c8ea698a6f4ec07ce21f1e5a","project":"openstack/openstacksdk","branch":"master","topic":"bug/1915151","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-02-14 13:23:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I58e9a8bde3dc9425c8ea698a6f4ec07ce21f1e5a","subject":"Add \"vlan_qinq\" attribute to the \"network\" resource","status":"MERGED","created":"2025-01-21 10:50:58.000000000","updated":"2025-02-14 13:25:40.000000000","submitted":"2025-02-14 13:23:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939703-bug/1915151","meta_rev_id":"5ce8c1907de5988f44c0938e583526646a5c0c42","_number":939703,"virtual_id_number":939703,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-14 13:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-14 11:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2025-01-28 23:16:48.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-13 12:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-14 11:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-21 13:43:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-23 08:21:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-01-28 23:16:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-02-13 12:05:23.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-02-14 11:07:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"c72a43da587cd83a88abeca242bb31761954a1a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-21 10:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb909d8321b98748dd53914d07a5e1475db04286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 13:43:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/936be433f798461ab09c0c1e6c793b11\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/e8be52cd0a5f440299ba406b1c132468 : SUCCESS in 1h 00m 17s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/38d2f1e8a3d84af2b5d9faca2f1a95f6 : SUCCESS in 33m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b871850de33b48aca34465615ea9b1b6 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/025f8dd064bf4ffebc54934465302a47 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b33eccba6c9042bbbab3d3a32742a7a0 : SUCCESS in 4m 04s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f39b553512534a359870974ec122abd0 : SUCCESS in 52m 30s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a1d66d3f2f2a475786526a6c10e71108 : SUCCESS in 3m 33s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f10e34fd932147728e14d8264a72968d : FAILURE in 4m 32s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/fa1f55b1f0d14694bfeedcbf34a30d94 : FAILURE in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2707facc39432693ce8f774c43241d : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8517c7e0f9c64f0db9dc304828489abe : SUCCESS in 3m 31s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/46ec301631aa4e02baf6717f9a30696b : SUCCESS in 1h 46m 08s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/8e0de72ec4a448f6b7810ff4e50ae308 : SUCCESS in 14m 43s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/87718e983f8f48638e62a383ee1a091b : SUCCESS in 30m 06s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6efad798c928426bbc99d65eda804644 : SUCCESS in 1h 09m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/39313ab0649d4cbdbc3ce6c31092b414 : SUCCESS in 1h 03m 19s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/92ad747aa39048fc937cd51ef574f986 : SUCCESS in 1h 07m 54s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d5794bd894394de1b326424050c0b4d1 : FAILURE in 51m 11s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/10eda3a937854e45b7482a15db25a240 : SUCCESS in 34m 11s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/23d117616c0d412491568289c6110ce9 : SUCCESS in 32m 25s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ae9f4d488f9c4e93b377fdb17fe6d01f : FAILURE in 33m 01s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ed00aaa4e24c428687669fef5be8b104 : SUCCESS in 51m 05s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/633300729f5042ca800d3ed6cdd74626 : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84ab35c8cb2ad63e0c89b86f7934d649f7c5b7e5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-23 08:21:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2947e4ca658946a5b8f5b777d004ddcb01654c69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-01-24 10:13:41.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8627a8c9bc1d2e8c6023437cf87b0cc57ee93bf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 11:47:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75cc9355d172435eae0c950ad006313a\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/edafa21209104698a8da6b49df33e75a : SUCCESS in 47m 55s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/6930ff773fd143da9eed537180f0d494 : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddddc350f2dc415398dd9f6804a045a3 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfd0363204db4f608fd85a3b3e40635c : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af504daf856d4c1ea13b7aea149ea347 : SUCCESS in 6m 32s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/790dd751d02c435284072a65c12fcc24 : SUCCESS in 1h 03m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0a0221bd28194fd9a2aad1362d38afe8 : SUCCESS in 4m 07s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b4392a60013a4f968d4c228b0dc58280 : FAILURE in 5m 54s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/9cad178611444114b4fe4e1108697b3c : FAILURE in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbff69eca26d467eaaae263dccb814da : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69264cb94ae1429b9d7939fe104199d6 : SUCCESS in 3m 14s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ab1cc85b00a64ce790021d67f82f3c66 : SUCCESS in 1h 15m 48s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/dae9b0eef8884d7b8ea16ac085a6bc4c : SUCCESS in 12m 38s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/e91af864f2c24ef9adab43397217500b : SUCCESS in 33m 56s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28b8e14252de430d8c90b925819f04b6 : SUCCESS in 59m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d5932026ef4a4c5faf7a9f895c0a58a7 : SUCCESS in 1h 23m 46s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/b462565abe7d43d3990cd505b6e30f59 : SUCCESS in 1h 28m 00s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c62ce43c604e43f18f438c79f1d0c04e : FAILURE in 55m 42s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e232b09f31d7424983b6c45b335b3735 : SUCCESS in 26m 20s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/d1ca647fc391494eb69fc62308899d8a : SUCCESS in 30m 22s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/827034dc5a1f4d8a8c43692cd2179a5f : FAILURE in 37m 31s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c0126eaa013b45ae8d6047219a63e1d3 : SUCCESS in 1h 11m 53s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a92a6f76877c478fa65dfe21d67bf6c8 : SUCCESS in 54m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d56a0db3020cdfaf5e645c634ab1b1497d9e992","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-28 23:16:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ae4c83553d7bc58b0e9890d9ce102fd98c6779bb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-28 23:16:48.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"0e8394f1f81327263ebe760e775c0f354c6bb9eb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-12 15:43:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76f61ffa3ad0609b79284308c43c72dbbc1d6e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:42:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9226681e49054c51b79d7419a27ac446\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/2c7bb8d98ccc4794bfff2b28e2abcdb5 : SUCCESS in 1h 05m 22s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3fed109342424b70b9c4aea4efb3447f : SUCCESS in 46m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b3774d055d14b73bcc3dced349d1649 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13eea2a8da1e4d7c8c339a73484f0fe8 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/983846a3b3744b898f92860ee4f735a1 : SUCCESS in 4m 41s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c68950fb04274521943c02ca3000bc78 : SUCCESS in 53m 31s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2b22ce740c1d45198f1993b0e003c101 : SUCCESS in 3m 47s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7b68ea674664421cb52b280598c2b448 : SUCCESS in 4m 42s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/be917f0500364498a93f6f9d526781b7 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b7ef4fea104e5e8d614e326c3bbdb1 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/737034c130c44982b7bf7eff40842cb8 : SUCCESS in 4m 59s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7ef4d5441295497ba96f8c238839b3ef : SUCCESS in 14m 30s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/490f13080aa2432fa1951b43f9f5f855 : FAILURE in 5m 20s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/adfdd4c4024145bea606442212435b54 : SKIPPED Skipped due to failed job nodepool-build-image-siblings (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12212ea2e3a840aa9186c1923223625e : SUCCESS in 41m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/daf6a5b4cff6481b8f6b55d543190ef4 : SUCCESS in 1h 51m 17s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4f5b8555e2c44002b5f4483988d9543b : SUCCESS in 1h 34m 17s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c0482e34171d4522be97ea801c0c469d : POST_FAILURE in 18m 32s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/31b18a6ab0b64f999a8492e1d0224749 : SUCCESS in 30m 02s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ea8bf0f20a0c4b888d22e38ed33217cc : SUCCESS in 23m 28s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/469338915aff4005aa55ed16ecaea8df : FAILURE in 30m 31s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e84633f4885e45179de204609cd65178 : SUCCESS in 54m 07s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8b8e1e837081423b8c837b46a3348a73 : SUCCESS in 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"350b0183017c21812fe0d4dead7035db57284cda","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-13 12:05:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8af2199230b7fe3fe9ec42dd4b557ab2ef02d69d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-14 11:07:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dcf632630d54c5774c499ad25b0c919367a7ed50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 11:07:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fcb9e3d765ecf4c565c80daa9cd248c81713a867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 13:23:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96aed0a7879f4913971fe274d98b4c4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43eabc807809447aa43ab60cfe502113 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5fa312c471e4ad5948feed383d2213b : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d9d993140f349c29e2ad546ae20f794 : SUCCESS in 4m 01s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7cfac91725e246d5935d1da3b7c00d3b : SUCCESS in 50m 21s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/52e70a43e0f748eab89fa608a1a8552f : SUCCESS in 3m 39s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e277cb6532464cca9a396026fe39cc82 : SUCCESS in 4m 38s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/1e852fd469d54ba28b76ecb2c9ea7810 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a777ae0f2c4bf6a6d261cbc70638a1 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/506209f3af2a486c8186f5bee5cd06f6 : SUCCESS in 3m 40s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/406b8656f1b94a90b993193a2103db19 : SUCCESS in 4m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdcf9da39b3547768775c15c2049f386 : SUCCESS in 1h 09m 58s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1c0ced1c01664a0488c95b0e18b1eb62 : SUCCESS in 1h 15m 24s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/69ba13050e7b4a218300ef1dd218fae3 : SUCCESS in 1h 32m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5d32e50d01376f09eb419fe14bd088746f5b3624","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 13:23:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5ce8c1907de5988f44c0938e583526646a5c0c42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 13:25:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e8c3cd5758449e838260ce2ea43041\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c433d835ab474bddaa64badf9682c41a : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/123e8aaffe684dd3b4190ed5db711e43 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf862933628e48d70cfee121e24833660bb80eea","revisions":{"6a13aa8bf4e137f614f0806a8b80de0e00f8f011":{"kind":"REWORK","_number":1,"created":"2025-01-21 10:50:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/03/939703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/939703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/1"}}},"commit":{"parents":[{"commit":"da17a2762ef432ad6ffdd59df481067aa02e8863","subject":"Fix TestKeypairAdmin tests and compute.*_keypair","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/da17a2762ef432ad6ffdd59df481067aa02e8863"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-21 10:50:04.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-21 10:50:54.000000000","tz":60},"subject":"Add \"vlan_qinq\" attribute to the \"network\" resource","message":"Add \"vlan_qinq\" attribute to the \"network\" resource\n\nRelated-Bug: #1915151\nChange-Id: I58e9a8bde3dc9425c8ea698a6f4ec07ce21f1e5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6a13aa8bf4e137f614f0806a8b80de0e00f8f011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6a13aa8bf4e137f614f0806a8b80de0e00f8f011"}]},"branch":"refs/heads/master"},"bf862933628e48d70cfee121e24833660bb80eea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-24 10:13:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/03/939703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/03/939703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/03/939703/2"}}},"commit":{"parents":[{"commit":"da17a2762ef432ad6ffdd59df481067aa02e8863","subject":"Fix TestKeypairAdmin tests and compute.*_keypair","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/da17a2762ef432ad6ffdd59df481067aa02e8863"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-21 10:50:04.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-01-24 10:12:57.000000000","tz":60},"subject":"Add \"vlan_qinq\" attribute to the \"network\" resource","message":"Add \"vlan_qinq\" attribute to the \"network\" resource\n\nDepends-On: https://review.opendev.org/c/openstack/python-openstackclient/+/940111\n\nRelated-Bug: #1915151\nChange-Id: I58e9a8bde3dc9425c8ea698a6f4ec07ce21f1e5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bf862933628e48d70cfee121e24833660bb80eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bf862933628e48d70cfee121e24833660bb80eea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
