)]}'
{"id":"openstack%2Fopenstacksdk~946199","triplet_id":"openstack%2Fopenstacksdk~master~Idb4e1e8c4b7606f4d309aade6322c923cfd64483","project":"openstack/openstacksdk","branch":"master","topic":"identity-gaps/user_groups","attention_set":{},"removed_from_attention_set":{"36482":{"account":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"last_update":"2025-04-08 10:47:56.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-04-08 10:47:31.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Idb4e1e8c4b7606f4d309aade6322c923cfd64483","subject":"Add identity.user_groups method","status":"ABANDONED","created":"2025-04-02 20:21:32.000000000","updated":"2025-04-08 10:47:56.000000000","total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"f5198bc6636c1aa4bcd1f7109d090ac6f22fd7aa","_number":946199,"virtual_id_number":946199,"owner":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-04 22:31:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-2,"date":"2025-04-08 10:47:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-02 22:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-04 13:14:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8147d2e1f0677caf39421c0d87bc36f11be7f595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-04-02 20:21:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f70d9ec544d749252cb9e898b580402934d8a14e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-04-02 20:34:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1615c0689211e23f2f9987811489358a12ff4ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 22:08:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/669b48d56b83482490ad360ac2aa5bb5\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/1dacf319fffd43a3afe66f89c503c559 : SUCCESS in 40m 25s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/394253438940411fb68a134251be3724 : SUCCESS in 21m 13s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7e6aabb7d8d847cf9cd53b2007f6c823 : SUCCESS in 42m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aec9eb30ebd417bb57ba23938e5750e : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c8fe4c697404c9681356da8e449ce5b : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cd693cfb9374cd6921dd9851ef6231b : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b1e51266ec24c558178822902732009 : SUCCESS in 5m 55s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/57a2cdb969464a64a9d621bfec4c8f2d : SUCCESS in 36m 17s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1d933b14dc7d4dd8aae9cda165b46e98 : SUCCESS in 3m 56s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/18a69b467e4d4f289ca471e29e387fc2 : SUCCESS in 4m 10s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/bce696a0dc054fb89ea7ba057821bc3f : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3aaf38ece5c469892325e201eb7f41d : FAILURE in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f77941e6c77946b4add1ee1df99260c9 : SUCCESS in 2m 34s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2e73264884ca4f2cb8872f128f314c3b : SUCCESS in 1h 05m 52s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0e6de3a6839240c3ad0046b3275fb261 : SUCCESS in 14m 23s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/45042698bb004c0681170539fb52b4c0 : SUCCESS in 38m 52s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7cc09c860e04326884fb6f3f3e91b3c : SUCCESS in 45m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b9be3beb1cb848a2b805d0777c9ff9a6 : SUCCESS in 53m 59s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d63c9e8e4d8d41b7bd83df374c2fbfc6 : SUCCESS in 1h 23m 38s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/cf21aaeadd4c4a17b64c5ce075e837aa : FAILURE in 47m 14s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/deefba2853934a3da937f6f8cc773717 : SUCCESS in 29m 17s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/380d6490c44743748542bb67300a91e5 : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/15a2f0af6920484ea82a273036f33cc2 : FAILURE in 27m 39s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b2f0853f3482435984962c6a3fa46a0f : SUCCESS in 1h 27m 39s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8d9820d230fe4a3e9e02bb2b630d7b2e : FAILURE in 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5605cc80ff206d106cf7e9e94c0eee39ed870f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-04-02 22:30:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"528d4b43bb8765e5b9ab53716b36f796bdc5ddbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 23:49:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29399d85fe9346458fa7694ed5b1ba02\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a944f2ba60c442f38bc6ff7bf20c3cbb : SUCCESS in 40m 41s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b2eb29aac3a7447d8e54e8e9d86a7e4b : SUCCESS in 35m 43s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/b5ee78a3fea447bcb581f2d67086dfdc : SUCCESS in 59m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/325f405a50a94fbcb1f2382996d50c0b : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/586549edb85d4edbb0037875c7f186aa : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1ec6e85d4cb4a659055315b20e26e07 : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e8643c644124f41baf18933f28c6802 : SUCCESS in 3m 48s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1fce13eac6ee4af6bf91997098ceb008 : SUCCESS in 32m 25s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/70cdd5431a8445afb1ebde86e992cf14 : SUCCESS in 1m 54s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2325315e9d384454a012f1039f484c65 : SUCCESS in 2m 45s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/98463bc71b1a41e6ad70b7061846ee6d : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4db5254b2e4e29a6c39d14b475bbd2 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8fe98460a1d4596bef8739d892e54c0 : SUCCESS in 1m 45s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/72524c9fa68548e48dffa08e6e28e366 : SUCCESS in 43m 08s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/7658c6e7a27d47f0ad20b8636660aed8 : SUCCESS in 11m 15s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/952c789463ed469e9389cf1e98bf8301 : SUCCESS in 27m 18s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53d1411fce9b4b0ca23da68b46205aff : SUCCESS in 49m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f599fa47843948f680ef49eb3da7c860 : SUCCESS in 1h 15m 54s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/9f3b5e2ea0cf47f0937d6d85ef3f29cf : SUCCESS in 1h 15m 23s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9e18ce161c7d45cc84423fa497db67cd : FAILURE in 46m 02s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/0da39dd95fa3493ca7aaec4f3221a748 : SUCCESS in 24m 40s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/52577ee3f3044f67b645ed8ed056682c : SUCCESS in 21m 52s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5ebff5feb3914c1c9c856636d7edd556 : FAILURE in 32m 39s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5a79f000d8af4c359555df939bd98894 : SUCCESS in 1h 01m 11s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/7556ccfaf3ea4b90be6575c9e23fb4c0 : FAILURE in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a6382218ef9b929512fcbc3a275a6fcad0b057f3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-04 13:14:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f028494d845342b1a0ad95737b34625538487fea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-04-04 20:48:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"efc8a89778511b9ac5bcd472828a81f46c32b76c","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-04-04 20:49:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6dd7bf8a186f5c159912bb86907de8148639cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 22:31:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eafede4fb4c64de2910b6fe22508bbcc\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7c39669fb4004b42a6a7a2b7425bd8d0 : SUCCESS in 30m 31s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/221cafde920041f5bae766a301642bc3 : SUCCESS in 29m 37s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/d3981a90c3a74f7a9c017ecd738719f3 : SUCCESS in 42m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68234880ef224ee08fd0a5dcf890b203 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f80a6220a1e7476baae29bcb52205060 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cc09e7bbf9644c1bcc6937f1b5d9dd5 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7841ffebce34511b8c7693a39f85948 : SUCCESS in 4m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/67658ed01e2044d680b0d3a03b96a760 : SUCCESS in 1h 05m 02s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6a5771e8afe14051a985c42b31107e11 : SUCCESS in 5m 01s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f53f3a6be8d14abe9c56fe4a7dc49996 : SUCCESS in 4m 02s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/89e7bbd5f7ea4afba9ed37a02750a191 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f4abd3a2aae425490c01ef0183dd84b : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c723ee6cbaa41c5acd84f01f8f3ff09 : SUCCESS in 1m 46s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/95b6613806ae47d29c395fd0bd8f25f6 : SUCCESS in 38m 28s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/580ec0d3c72e460689a60291c061f79c : SUCCESS in 10m 34s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/7cd97288e8f140d78873b29788b9a4da : SUCCESS in 24m 34s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30cda0a19e494310963b349fafb527b6 : SUCCESS in 44m 06s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e027247381ea49d58fd68c34b4116e2c : SUCCESS in 1h 29m 23s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/c15c2d6c629f41028722754328ed39f7 : SUCCESS in 1h 35m 13s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c52f7100d82243e594355de967ae67a7 : FAILURE in 43m 45s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f6388abb6a254775855013176192b8af : SUCCESS in 29m 16s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b038dd9e7bbc4e6c994f6274a3bc443e : SUCCESS in 28m 50s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/22d2ae9bebcc41d4898c307c5365c2cb : FAILURE in 26m 34s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c9af979b14f241dc9a149449cb44acd6 : SUCCESS in 41m 55s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2727d692b9014f138c71a5f53a46257b : FAILURE in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93d3fa602866b1090cfb815606738de276e5fffe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-08 10:47:31.000000000","message":"Patch Set 4: Code-Review-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f5198bc6636c1aa4bcd1f7109d090ac6f22fd7aa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-08 10:47:56.000000000","message":"Abandoned\n\nI\u0027m going to assume you\u0027re okay with this being abandoned 🤞","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4b42b27ff62091b0c295df4913a8ad2d5536b830","revisions":{"bc010daddf52edda8939140ce22ac407bd229fbb":{"kind":"REWORK","_number":1,"created":"2025-04-02 20:21:32.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/99/946199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/946199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/1"}}},"commit":{"parents":[{"commit":"79fbbbabff69e8e40e725ebbb8c605d2cc70ccae","subject":"Merge \"Support streaming download of the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/79fbbbabff69e8e40e725ebbb8c605d2cc70ccae"}]}],"author":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-02 20:18:36.000000000","tz":-420},"committer":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-02 20:21:11.000000000","tz":-420},"subject":"Add identity.user_groups method","message":"Add identity.user_groups method\n\nAdd method to list groups a user belongs to.\n\nChange-Id: Idb4e1e8c4b7606f4d309aade6322c923cfd64483\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bc010daddf52edda8939140ce22ac407bd229fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bc010daddf52edda8939140ce22ac407bd229fbb"}]},"branch":"refs/heads/master"},"b4da958dff2bf74df71556d9f951fd8b2d5a0ffe":{"kind":"REWORK","_number":2,"created":"2025-04-02 20:34:28.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/99/946199/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/946199/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/2"}}},"commit":{"parents":[{"commit":"79fbbbabff69e8e40e725ebbb8c605d2cc70ccae","subject":"Merge \"Support streaming download of the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/79fbbbabff69e8e40e725ebbb8c605d2cc70ccae"}]}],"author":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-02 20:18:36.000000000","tz":-420},"committer":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-02 20:34:21.000000000","tz":-420},"subject":"Add identity.user_groups method","message":"Add identity.user_groups method\n\nAdd method to list groups a user belongs to.\n\nChange-Id: Idb4e1e8c4b7606f4d309aade6322c923cfd64483\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b4da958dff2bf74df71556d9f951fd8b2d5a0ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b4da958dff2bf74df71556d9f951fd8b2d5a0ffe"}]},"branch":"refs/heads/master"},"01ac4eeb1ad42070a3a02c6a4b056dc212f9ebc0":{"kind":"REWORK","_number":3,"created":"2025-04-02 22:30:34.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/99/946199/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/946199/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/3"}}},"commit":{"parents":[{"commit":"79fbbbabff69e8e40e725ebbb8c605d2cc70ccae","subject":"Merge \"Support streaming download of the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/79fbbbabff69e8e40e725ebbb8c605d2cc70ccae"}]}],"author":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-02 20:18:36.000000000","tz":-420},"committer":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-02 22:30:15.000000000","tz":-420},"subject":"Add identity.user_groups method","message":"Add identity.user_groups method\n\nAdd method to list groups a user belongs to.\n\nChange-Id: Idb4e1e8c4b7606f4d309aade6322c923cfd64483\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/01ac4eeb1ad42070a3a02c6a4b056dc212f9ebc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/01ac4eeb1ad42070a3a02c6a4b056dc212f9ebc0"}]},"branch":"refs/heads/master"},"4b42b27ff62091b0c295df4913a8ad2d5536b830":{"kind":"REWORK","_number":4,"created":"2025-04-04 20:48:57.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/99/946199/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/99/946199/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/99/946199/4"}}},"commit":{"parents":[{"commit":"79fbbbabff69e8e40e725ebbb8c605d2cc70ccae","subject":"Merge \"Support streaming download of the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/79fbbbabff69e8e40e725ebbb8c605d2cc70ccae"}]}],"author":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-02 20:18:36.000000000","tz":-420},"committer":{"name":"ArtofBugs","email":"oweng@osuosl.org","date":"2025-04-04 20:48:50.000000000","tz":-420},"subject":"Add identity.user_groups method","message":"Add identity.user_groups method\n\nAdd method to list groups a user belongs to.\n\nChange-Id: Idb4e1e8c4b7606f4d309aade6322c923cfd64483\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4b42b27ff62091b0c295df4913a8ad2d5536b830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4b42b27ff62091b0c295df4913a8ad2d5536b830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
