)]}'
{"id":"openstack%2Fopenstacksdk~946350","triplet_id":"openstack%2Fopenstacksdk~master~Icc83eb5e70bba89ff2b496e9aa01c160cf9e4270","project":"openstack/openstacksdk","branch":"master","topic":"cinder-gaps","hashtags":[],"change_id":"Icc83eb5e70bba89ff2b496e9aa01c160cf9e4270","subject":"volume: Add Transfer to volume v2 API","status":"MERGED","created":"2025-04-04 16:35:02.000000000","updated":"2025-04-07 10:27:49.000000000","submitted":"2025-04-07 10:26:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946350-cinder-gaps","meta_rev_id":"5489dac2850b2f0a25b672c6ca576710d4ead1f0","_number":946350,"virtual_id_number":946350,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-07 10:26:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-04-07 09:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-04-07 09:05:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-04 18:12:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab32f1219a8d865880906fdcf60ae25c14e8da1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-04 16:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db2074b627a7258f2b88cc7abf05828c1facf0da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 18:12:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/976525a3ca42468299805d0d1db3dc6d\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/d3b7d390245647dbb69841ecb2aff59e : SUCCESS in 39m 06s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/ceeda23c61504b178526225b1776a3c1 : SUCCESS in 31m 01s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/62eb8078324b4100a92c6f440a0e4027 : SUCCESS in 56m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d7f41da9e1f41109ddb1e01a8d5fc10 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83ed8a2522fe412cb9cf4c55daa04e2f : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72b40d6fdbbd405882cc88b594751327 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5dbeeda73e274e1ba8dcc736d47056cb : SUCCESS in 7m 04s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4ec47841600c480796197ba46168725f : SUCCESS in 1h 01m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/09191cf6bd0347febf8093771be17984 : SUCCESS in 4m 20s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4df0faff32b74ab89ac8f7f34499e365 : SUCCESS in 4m 47s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/ad5a71352b264c439b07fdc744647a32 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c1bcc6a0be4ed396cd5c965bb0c195 : SUCCESS in 10m 35s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a0b4438d35fc4a10acaf9d6816493cf3 : SUCCESS in 52m 17s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/e087a15e029d4b53baec3408cd90f750 : SUCCESS in 12m 41s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/d7513f597b77428fb10edcc446f4a787 : SUCCESS in 31m 49s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddcf3c6fad84445992e6a4d0b5d65fdf : SUCCESS in 54m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/52fd4858531445b1ae4071e0f5bba6a3 : SUCCESS in 1h 32m 01s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a2b67f0310e746efb888c0ff1a56e8fc : SUCCESS in 1h 25m 49s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/84f0137dad4a412582b17a679f027c17 : FAILURE in 52m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/2374c473ed0241dfa771b2051fad3bd6 : SUCCESS in 29m 50s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9b1fb4f9f6914298a402a9858e14e749 : SUCCESS in 22m 17s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7b7159b06c3d419c8a07bb7811f67087 : FAILURE in 39m 18s (non-voting)\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/63ddf54de5944155819c57ac2e44ef50 : SUCCESS in 1h 12m 32s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8ce7ad0240dd47048f30d6cde261d2f2 : FAILURE in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a69bef9892e482378a00dbc4af3ae973027daf00","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-07 09:05:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fd9fbed3f308b3848febee9b6fcb713695de847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 09:06:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"caa95c657a7aaa64d14cd9112d8f2ddd4eda8340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 10:26:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/542d4c2e9c504c3f89edcb88d7983175\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e87009fdeb24b66a434fe24f05d9440 : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e40ba4e7531f40afaa65f1ab5f1f8f44 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/439f532215244691a8868681d30dbf82 : SUCCESS in 4m 06s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/574ccfc4c0e744ea9f96755932a5c39f : SUCCESS in 51m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cb62461cd92f4ce19dbfb7672ecab5b3 : SUCCESS in 3m 46s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dffbeed872354edb8de77c63938c2a79 : SUCCESS in 4m 39s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/b005761fc5e24a7793c83e45e3a15107 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0061b1dbf3b481293fa6f4c94c4e157 : SUCCESS in 11m 09s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e141f6c492c94d30b9d05205bf592c3b : SUCCESS in 2m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7bc568844304d5ba162a51b442c0766 : SUCCESS in 50m 58s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/28c625c0ec9f4b1697287e8ba409a9ab : SUCCESS in 1h 19m 44s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/cfb33220675f4adb9c473114752934f6 : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dcf8c6498b1dcd40c50e97afd87eabefb7d98d41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 10:26:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5489dac2850b2f0a25b672c6ca576710d4ead1f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 10:27:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80eba6c6d7a04775a1e91d507c9987bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60315aabd467480da31832fa4aa86f4b : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2440aa7e2dfb4a7e34165d316f5e5138f395468c","revisions":{"2440aa7e2dfb4a7e34165d316f5e5138f395468c":{"kind":"REWORK","_number":1,"created":"2025-04-04 16:35:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/946350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/50/946350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/50/946350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/50/946350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/50/946350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/50/946350/1"}}},"commit":{"parents":[{"commit":"43dd5d9041c78c70f41ebf07fbaa012fa6d7638a","subject":"Merge \"Bump compute max microversion to 2.100\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/43dd5d9041c78c70f41ebf07fbaa012fa6d7638a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-04-04 16:15:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-04-04 16:34:15.000000000","tz":60},"subject":"volume: Add Transfer to volume v2 API","message":"volume: Add Transfer to volume v2 API\n\nChange-Id: Icc83eb5e70bba89ff2b496e9aa01c160cf9e4270\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2440aa7e2dfb4a7e34165d316f5e5138f395468c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2440aa7e2dfb4a7e34165d316f5e5138f395468c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
