)]}'
{"id":"openstack%2Fopenstacksdk~953923","triplet_id":"openstack%2Fopenstacksdk~unmaintained%2Fzed~I4c87a61bca04ac12fd7bb284174f351df545a3f0","project":"openstack/openstacksdk","branch":"unmaintained/zed","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-07-18 15:57:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c87a61bca04ac12fd7bb284174f351df545a3f0","subject":"Remove un-needed nodepool testing","status":"MERGED","created":"2025-07-02 02:19:39.000000000","updated":"2025-07-18 15:59:02.000000000","submitted":"2025-07-18 15:57:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953923","meta_rev_id":"b377fa03c68d9e74940da3a366a70cdced092f4b","_number":953923,"virtual_id_number":953923,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-18 15:57:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-07-18 14:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-07-18 14:05:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-02 04:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 06:15:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-07-18 14:05:05.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c5e4aa57f24c37fb257deb1afa440cfe1115df3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-07-02 02:19:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e59a8b10ae50a0960e1abff9b156b930f2f2f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 04:28:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8310ef83cf1d4bde8c2734baf6eff12a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9d59832ab84c76ba921480750f240f : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36577acda3fb48c2ba760f84ed47536a : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98dba305b187474fa1428b50a7eb6dd8 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b30ee4e0c2b1436582f63ba9826853d3 : SUCCESS in 3m 03s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/78d5b522c8474863ad05a81050e32208 : POST_FAILURE in 17m 13s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/69ff6706f5d04373a3aeadbca78069d2 : SUCCESS in 2m 52s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/19315db13a0142229b826ff9777b890f : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cf7366951943248ae05b6d4f37d5a3 : SUCCESS in 5m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4bbf746e34d445e99be0a41326871c59 : SUCCESS in 2m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6c388a2a1394af997c5dde32a9d4926 : POST_FAILURE in 11m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cf20bc6437f94f868509191d53d6c203 : POST_FAILURE in 12m 05s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/225fc892c3b046de80f1b920a358e306 : POST_FAILURE in 7m 38s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7aa19b903da54035998dea5adcb5de64 : POST_FAILURE in 7m 15s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/5582b992339b415a8e08f3bd8c20384c : POST_FAILURE in 8m 58s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/18104951c50b4a03a7d9050375296230 : POST_FAILURE in 8m 47s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/743023b6df7f476b904710376cf4e580 : POST_FAILURE in 3m 23s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/3d7ff8c3a6a4483e9a28f66a1d534a1c : FAILURE in 13m 45s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66565d33614e4abfa26c7f2e656a4dcc : FAILURE in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0470b3218b61f6ad314a68b51280ed44007fd236","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-17 06:15:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7f9b1c2f5a018d42b76fc8de0b3c24d8f7579f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 07:26:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c65d017565442dfb3132901605c3432\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3e8d576fe24d7682f72ed332932b1d : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dff2ed2462fc4443ae289fd71f126496 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62a6049adaa1476bbbd11f7c0be6a7a2 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12c65a8a155848b1a38a673e85447920 : SUCCESS in 3m 06s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0eca48886b004317af07595025391f13 : SUCCESS in 35m 30s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c775762938334172919f0ffe39254e98 : SUCCESS in 5m 45s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ace94ade76734cfca3fe926b654cc03c : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bbf73e73686411e9d6dfb528b8ae4c0 : SUCCESS in 6m 15s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a142b45c1c2f410baca770682199d8dc : SUCCESS in 2m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5015599865674046997e455fca216e5d : SUCCESS in 47m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a6b3607e28704000b5e2f9eab374c523 : SUCCESS in 1h 03m 51s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b2b9a18339064046b66993d3e513bdb1 : SUCCESS in 52m 06s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/9d7804424ae241d799aa6b30ce476b24 : SUCCESS in 25m 22s (non-voting)\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/e87a6caca4174991ac4ee6ab052a4069 : POST_FAILURE in 22m 31s (non-voting)\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/2d5967bbd0dd4526aeab014b99111d68 : SUCCESS in 29m 19s (non-voting)\n- openstacksdk-functional-devstack-legacy https://zuul.opendev.org/t/openstack/build/c4f7cc671bb0458b8ce61083c6fe8ea4 : POST_FAILURE in 10m 11s (non-voting)\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/11dc3e527a654eb784abe65278732d8b : FAILURE in 18m 24s (non-voting)\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/084582b885eb4974a98ba228fd8f5d00 : FAILURE in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a2971e216f168821d9e063c295c5bd4482ac438","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-18 14:05:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a7f2bdba8452fb0ee4085c9fa0d25393dfc3847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 14:05:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c795e0138009b902c7ff6ec669df541f34da65f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 15:57:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5bf34b6f0c434da925157f1af3b754\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e9815c47c54e82a2d1649edf111c2d : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d13aa71339da49388c5061be687595b1 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53fc39fbc3fb46f193ab56fa4dc1d085 : SUCCESS in 4m 01s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/473e76a53b444fb08c711e050df6e7b4 : SUCCESS in 36m 32s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/1344a9f3c9ed449788e24ec2213f5f6e : SUCCESS in 3m 00s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c353f867f3c14b849ecd819cad5721b1 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7280fb4de19744e990f07044a1c09907 : SUCCESS in 6m 10s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f26251ce0f6f4dd2911644d4a53bc467 : SUCCESS in 4m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/528d81fb75a34821bededb8fd7eda6d3 : SUCCESS in 35m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e41703069dda44a9a68f4eea5de73557 : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5dac7d45755d013e6cd9a0beafcc32b0f8899ccf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 15:57:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b377fa03c68d9e74940da3a366a70cdced092f4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 15:59:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f73407e75604c6ea4608b50776cbd49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a49f1eef0f415dab9bd72667f60200 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c046fc90dab2ca5aafad50e9cc4263cc820aa19b","revisions":{"c046fc90dab2ca5aafad50e9cc4263cc820aa19b":{"kind":"REWORK","_number":1,"created":"2025-07-02 02:19:39.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/23/953923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/23/953923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/953923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/953923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/23/953923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/23/953923/1"}}},"commit":{"parents":[{"commit":"49ee948e6d026854ea3dab19a6b4c5942762de50","subject":"[CI][UM-only] Remove undefined job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/49ee948e6d026854ea3dab19a6b4c5942762de50"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-06-24 03:57:33.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-07-02 02:19:30.000000000","tz":600},"subject":"Remove un-needed nodepool testing","message":"Remove un-needed nodepool testing\n\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nChange-Id: I4c87a61bca04ac12fd7bb284174f351df545a3f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c046fc90dab2ca5aafad50e9cc4263cc820aa19b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c046fc90dab2ca5aafad50e9cc4263cc820aa19b"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
