)]}'
{"id":"openstack%2Fopenstacksdk~956275","triplet_id":"openstack%2Fopenstacksdk~master~Ic9c8c574cb973b3e8e87171cfc4b7e1eec7dd0e8","project":"openstack/openstacksdk","branch":"master","topic":"functional-test-cleanup","hashtags":[],"change_id":"Ic9c8c574cb973b3e8e87171cfc4b7e1eec7dd0e8","subject":"tests: Streamline flavor test","status":"MERGED","created":"2025-07-31 17:24:49.000000000","updated":"2025-08-01 19:24:42.000000000","submitted":"2025-08-01 19:23:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956275-functional-test-cleanup","meta_rev_id":"42f2b135ed431a74f298f933ec0656dc8f7ac8c6","_number":956275,"virtual_id_number":956275,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-01 19:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-08-01 17:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-08-01 17:33:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-31 19:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ba885ff915dd8a1439099cca6f63edd0e6303e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-31 17:24:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cfcb528af5449e23b32a333b772826347288688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 19:59:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db0bc54f6c64fd6858e68cea7627725\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/f6b7456068664b36a048020a38411b90 : SUCCESS in 47m 05s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/12c2a4c32b064e4e9b3a65b16120f7a9 : SUCCESS in 36m 37s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/668023eeea4649deb556974d600e8e03 : FAILURE in 29m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a587a2fb66134a7ab292354fae30e103 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d8981ea194944db98e76c167510f626 : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b096e788d6534a158e93adf89f882566 : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ccb10fbc6a1549e38a782b77dd7fb95e : SUCCESS in 6m 30s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/76101aefb950474b999917ed553ec201 : SUCCESS in 54m 02s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d1d9fd43d57b4a0aad269e59bd5ecb68 : SUCCESS in 3m 55s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/296393a070e14e4cbfa96af1c60f71d0 : SUCCESS in 4m 56s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/2e7ec37dcc24474bad448cc78582e263 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbee19b27a5c4d75a065d3008e0aa0fd : SUCCESS in 11m 46s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b20905a3e3d14ffca2deb734ce8ebd73 : SUCCESS in 57m 02s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/a3911c9688644704830ce80f89eea8b1 : SUCCESS in 11m 44s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/dd4b06afd89549b7a88cc6cc7bf21e9d : SUCCESS in 36m 32s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94aa3233a29d49988136ec220386e616 : SUCCESS in 51m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f9e5ed4f0cb4494f8e6dc018eb1ff417 : SUCCESS in 1h 12m 08s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/8d8a17aa2dc148b2b8ddea95d92e5e19 : SUCCESS in 1h 30m 13s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/1e94ee5e50004381984d14c8678ad655 : FAILURE in 44m 36s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/412fd390447a47f5baaa7eb125c1db1c : SUCCESS in 32m 19s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9769f07b48aa4ed986998995568c43e3 : SUCCESS in 27m 03s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/15b58861888b4d9e8bcba08fa9c45baa : SUCCESS in 39m 51s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ce6dee984a9d4af6b86fa91c8b63fd05 : SUCCESS in 1h 08m 32s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/13388357ab344fa2a080586a070a4dd9 : FAILURE in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f12afb1b5687d7e133a8915c37713acaf0c47ff7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-01 17:33:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dab56e50f2914e21347ccface4528f6f7d312a83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 17:33:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66804412741872945a74c867a4fa2ceedd3ec678","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 19:23:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629cbe73048e452abebd50ada93a6f86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e6c81aff404e70af1acb902e111245 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f01fe25898b4b9bbe9909ddc7e578ab : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af55647431ef4962a6d4290ad734e7d8 : SUCCESS in 2m 54s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/21e14980b98743539bea841205fa57e6 : SUCCESS in 39m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2317cf4adcbf4dd0afc2a9e44d185e26 : SUCCESS in 2m 34s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/3fe29a9e4c004d3ba5cdd40faa59d270 : SUCCESS in 3m 15s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/0ed6abff102e4107beda41e50ef2b44a : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8942f2624ac846aba49833d8b6ab75ea : SUCCESS in 10m 23s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/17efde2941ef4c4c80196356a3999f7d : SUCCESS in 3m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06f1e63904f949cfbdbef4b93dae8703 : SUCCESS in 58m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/711d4352e3394c0582085f94eecb0cf7 : SUCCESS in 1h 24m 00s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d834eecd808d49caa431e7380d2eee6c : SUCCESS in 1h 04m 49s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ce7938c49d7c4cf897186530db295d1f : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e571ae33a75517d078b3435cc4d640dc7690c397","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 19:23:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42f2b135ed431a74f298f933ec0656dc8f7ac8c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 19:24:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7204ca885bef4989bd0c3e3f315efc4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25fcfef56a7f4d21800d9cc2ea8d1ec1 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2788fd0cb0184ffc770625bf1f4eaa0f4ce9a4f","revisions":{"f2788fd0cb0184ffc770625bf1f4eaa0f4ce9a4f":{"kind":"REWORK","_number":1,"created":"2025-07-31 17:24:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/956275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/75/956275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/75/956275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/75/956275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/75/956275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/75/956275/1"}}},"commit":{"parents":[{"commit":"d649aada8d5f72af3537b8c413adc05da1706b53","subject":"resource: add max_items parameter to resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d649aada8d5f72af3537b8c413adc05da1706b53"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-26 13:30:02.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-07-31 17:23:47.000000000","tz":60},"subject":"tests: Streamline flavor test","message":"tests: Streamline flavor test\n\nChange-Id: Ic9c8c574cb973b3e8e87171cfc4b7e1eec7dd0e8\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f2788fd0cb0184ffc770625bf1f4eaa0f4ce9a4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f2788fd0cb0184ffc770625bf1f4eaa0f4ce9a4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
