)]}'
{"id":"openstack%2Fopenstacksdk~956369","triplet_id":"openstack%2Fopenstacksdk~master~I47d65383233e23ec5edf1e6735c39c66eeafb804","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-01 17:20:32.000000000","reason":"Change was submitted"}},"hashtags":["dco-signed-off-by"],"change_id":"I47d65383233e23ec5edf1e6735c39c66eeafb804","subject":"tests: Use binary mode to open files","status":"MERGED","created":"2025-08-01 13:34:16.000000000","updated":"2025-08-01 17:49:10.000000000","submitted":"2025-08-01 17:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956369","meta_rev_id":"9e9d0e9044e164a39e94fa59e4cabd28e28e4153","_number":956369,"virtual_id_number":956369,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-01 17:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2025-08-01 16:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-08-01 13:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-08-01 13:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-08-01 13:36:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-01 13:56:30.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-08-01 14:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-01 16:45:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"dddd2076b061cfd79179db146bfd41b0b2abb92d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-01 13:34:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e56506859fc602ea2cdc41312903eca2f893166a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-01 13:36:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f472e929e9af694e202742f854ffac7d8559475b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-01 13:56:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"38432c5d4d6ab2cf47bc80ecd2673d594f3c4145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 14:57:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094d396368084098aa6bab7d6c96fd18\n\n- ironic-inspector-tempest-openstacksdk-src https://zuul.opendev.org/t/openstack/build/e74825529bec4a2bbbe3d014ca947e47 : SUCCESS in 35m 25s (non-voting)\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/08d78e570d1748208134eb29adc93519 : SUCCESS in 36m 26s (non-voting)\n- metalsmith-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/64582190279240799fe72b3dd0a8abe4 : FAILURE in 26m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b439c1ac3b1348aebf85d25d628410d4 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6436c6530c4c4d1fb506f6a099191ef7 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b46132e6f6204950a397e303326cf860 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/274052e3c70347df881965d1458c67df : SUCCESS in 5m 17s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1145065949694b3e89ac80cba7393315 : SUCCESS in 47m 55s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6d9bc1cd38394bfa867de5e22587becc : SUCCESS in 3m 35s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/10e3e785f4f34c59a8055b971efe80a4 : SUCCESS in 4m 42s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/7b5d70e6112242babfb1348072dfe4a6 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5eae71d97654fa88b027a992f1135be : SUCCESS in 11m 20s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fe37c354f62c475f9b821aaa69675d57 : SUCCESS in 56m 48s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/1459910e3f234a2c94a21ef745854d5e : SUCCESS in 11m 15s (non-voting)\n- dib-nodepool-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/e3bace7f114a428185605c9fe9ea2a23 : SUCCESS in 32m 59s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d14cb68a98b8470594609c8b18fd9a9f : SUCCESS in 46m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d5a1fc46226c432f8dc0293d46cb82f0 : SUCCESS in 1h 12m 34s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/5a6e3bbab6b04bf591923058f949d77a : SUCCESS in 1h 21m 43s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5500ccd7e0c04031ba593b9f5c4db2d7 : FAILURE in 48m 44s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/d814f31d2e1d4d448205538f938c0675 : SUCCESS in 23m 18s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/b417c79aea1247858fe6c8d66951a724 : SUCCESS in 18m 27s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3f1fa60acfe046b997f52c89f8288526 : SUCCESS in 33m 18s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/45b2cbc224e742c4ad9d364df74b4e1c : SUCCESS in 1h 05m 00s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/85aef04e76114bf49d3d6ef3d24f100b : FAILURE in 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f004ce9f8f7c87f8a46e1b68879adbad858b350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 14:57:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39e6a6da598bfa9ed5c8159b94f5891c0304dadd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-01 16:45:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"20ea94db25816b827ecf878a82961d7b8fece8d8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-01 16:46:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b48e8d1976f04e5f69a0e421e26c5a0d809680f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 17:20:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8d64c06048a4edf87d76a23a09281ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b0791a1d82409087879c59d0e9abb8 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14ebdc8c244e4375b8f4fe77fb0bcd82 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04627689c1cd4ca09702648ee14be1cd : SUCCESS in 3m 48s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ab7b5cda89714559a8c14411f6e12c69 : SUCCESS in 1h 00m 56s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ed3951772e8e49909976ba718d1bb191 : SUCCESS in 4m 00s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/15ba35916d8244dd903f28d2539ec47c : SUCCESS in 3m 44s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/a9db750bc875491d9c046b4d39592669 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642901a034ec4c2d8922c80103620641 : SUCCESS in 8m 11s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/701ffc36a6dc4d709c49c62e0f74e1a3 : SUCCESS in 2m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c09e01a82644f9382cb28b04a71c492 : SUCCESS in 1h 00m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4f9a504458fc49ffad714e17d3edd59d : SUCCESS in 1h 25m 30s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/5f8a7e17631a4808b755e5da91146a2a : SUCCESS in 1h 30m 53s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e28ff3eab8bb4f5690a211fdfc3b62a0 : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8d51fbd5a6205a9e20e16af3b0950d68fc028052","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 17:20:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23a674d1d6553f3d73e086f797f31b3240595e89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 17:21:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07cf674f6e34dadbcd5d98df15feeed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c7c67c673a4053a5756e9861181db2 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"9e9d0e9044e164a39e94fa59e4cabd28e28e4153","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-08-01 17:49:10.000000000","message":"Hashtag added: dco-signed-off-by","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61323654ebc0a8c8bd64cc4adfef8fd695d311bf","revisions":{"61323654ebc0a8c8bd64cc4adfef8fd695d311bf":{"kind":"REWORK","_number":1,"created":"2025-08-01 13:34:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/956369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/69/956369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/69/956369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/69/956369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/69/956369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/69/956369/1"}}},"commit":{"parents":[{"commit":"1649ed0cdaec9bd408082a33dab0c333c1c3fd99","subject":"Replace CLA with DCO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1649ed0cdaec9bd408082a33dab0c333c1c3fd99"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-01 13:32:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-01 13:34:00.000000000","tz":60},"subject":"tests: Use binary mode to open files","message":"tests: Use binary mode to open files\n\nA recent change to the CONTRIBUTING.rst file introduced unicode\ncharacters, which requests does not handle correctly when the file is\nopen in text mode. requests 3.0 is removing support for opening files in\ntext mode so we kill two birds with one stone here by switching to\nopening the file in binary mode. We also remove the unicode character (I\ntested the fix separately), correct the project name in the document,\nand use the raise from syntax to get better error handling in these\ncases.\n\nChange-Id: I47d65383233e23ec5edf1e6735c39c66eeafb804\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/61323654ebc0a8c8bd64cc4adfef8fd695d311bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/61323654ebc0a8c8bd64cc4adfef8fd695d311bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
