)]}'
{"id":"openstack%2Fopenstacksdk~977226","triplet_id":"openstack%2Fopenstacksdk~master~I6d4d1d12853251d0713113568047648189916c31","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-09 22:28:23.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I6d4d1d12853251d0713113568047648189916c31","subject":"DNM: Test with dependent neutron change","status":"ABANDONED","created":"2026-02-18 14:43:06.000000000","updated":"2026-06-18 11:02:47.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9889bd9c4a9abde88f2f02302ab6b756dfe8d5f9","_number":977226,"virtual_id_number":977226,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-10 22:39:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-18 16:30:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e51565ad1f66a8cd87158efa6dc7f556bead33bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-18 14:43:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffa90887d62cf4af97c8b214052a023ac6191b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 16:30:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dadd321ab6a4d54b2c59630f70e5e02\n\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/4b412ce5c6124e82a41aa858ef691444 : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0c00e903594aee9bed3e37c6160388 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73ffb9ed342b475d950dd54a80a45a31 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b555cb238044826bc74402ed64dd030 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/426314a90daf4450be146d06f4d7d369 : SUCCESS in 2m 54s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a05cba76b7014b05b014d79bde19b250 : SUCCESS in 33m 20s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6083eeb99464484fad4680cc54851954 : SUCCESS in 3m 47s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/e4d0b69999b3451c9aeab1826d186246 : SUCCESS in 5m 06s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/d9ff9dabd89c4694837c8a47b2435fc4 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac133ffdd0f4eb8bd036d1bac91b553 : SUCCESS in 6m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8907d0a2dae74482995e087c0e5bf872 : SUCCESS in 1h 10m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/54241084f83d4337a4cbd24a1a88c396 : SUCCESS in 1h 25m 41s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/aca21b1d1aba441aa1ae477486b3732d : SUCCESS in 59m 30s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d1fd329327b04558a78e196593ba239d : FAILURE in 1h 02m 53s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/36cd0872332c45f893efbf57e5879b76 : SUCCESS in 26m 40s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/195b8e86ae894a56b6f1e8c8a63a9658 : SUCCESS in 29m 52s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/2b045d09f01a442dbb63d9c7d57953cd : SUCCESS in 50m 47s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/62e01e6f6afb4060be7adb701451124d : SUCCESS in 1h 17m 41s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/deb78a8e59b543f9b85a050bc731522f : SUCCESS in 1h 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"212bbac30c84a4f80edad9492dba202463e16746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-01 21:25:10.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"761af50d157ce6a8bd9e34dee10685de11d94810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 22:46:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa0872803a8e4854a6558c456fc348ad\n\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/7f6fa2b5dfac4ed7a8d30d7e0fa8337d : SUCCESS in 25m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebc87b955d949019febb2ca50874a4b : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7f8365d179b4c0993f8f2161438ad7c : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e60b96b3f9e42e5aef25b2585b90685 : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/baa2c2735db64b8fb9fe3cd293b6ebf9 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db50da6b538445f9b97ef6cf99a46e3c : SUCCESS in 6m 19s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/869406c43493444fa82c6d54fadc73d3 : SUCCESS in 32m 27s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7938cd8b86d7488cbde0b487f1d0dddc : SUCCESS in 3m 29s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/26a1b05e795e4b0ab8ebb6256f94c8e7 : SUCCESS in 3m 26s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/74582858f25b4208a76f4560889e7b92 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5b19466d5147f4b1a9ee90040fbfa6 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aedd3143fa7e4cb7bc7665e9716dc1b0 : SUCCESS in 43m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5b1e6354f27e4a70b669805332e097ef : SUCCESS in 1h 00m 44s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/fe4dde987ee542f380344540095d4ba9 : SUCCESS in 57m 24s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e85f2235e5f04df2983d74a9fcb3cdc8 : FAILURE in 56m 53s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/62f9c320cb9c4bf797a82e9adf307890 : SUCCESS in 23m 51s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/80e9dfb0185d49e996ca52a0d4a8935d : SUCCESS in 27m 41s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/412a2eded0fe46df885cf440bc77d743 : SUCCESS in 43m 09s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ffdd3bd206e34f9d983a93d553184a26 : SUCCESS in 1h 15m 18s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/95df46aebcbd40bba44cc7e2db01040a : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b77ad3fa7f4c18b0932d66a3799e2df781866a51","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-03 00:55:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3966c71428251ae102140d1be8395d41ddc6ab5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 02:28:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0978e78d2de40c2aa503bc576ff1bc8\n\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/a8a61cc2efa84c03ac867c539a54e8cc : SUCCESS in 38m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674282fc341a48a4a0c282e37285bbf1 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91f433076efd4804899e39131e644243 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/286e9d1da303435d99f17ed7945d92b4 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b096ef7bb074ae4ada894da5cd30b84 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34041b12be374813947ff62f0c9ce142 : SUCCESS in 6m 39s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bb04ba4558084accb53bf6542ea099ca : SUCCESS in 1h 02m 44s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3d147341e4054fed989aeca47535fa6e : SUCCESS in 3m 41s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/68e6dbe635d946c1947231126f4aa1a1 : SUCCESS in 4m 31s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/4783988f9cb34527a9709a143b7941e4 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1323dcf1613146b1812f91c47a34d61d : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8642cd1c79d94bd99a18649ab9bdf1cc : SUCCESS in 49m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b84c7880d85f48ffa86367bf277a7ad4 : SUCCESS in 1h 11m 09s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a51d0febbcd348d4a7c1208ee72ae1b8 : SUCCESS in 1h 15m 46s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/571d276965e042f7b8d13b143d2d880f : FAILURE in 1h 11m 27s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/bab692c5cc9947d181c064ea111ec818 : SUCCESS in 33m 23s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/72a31b9306d24bb594d9ab279b607d0f : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/00e77507cf9048fb909327113f4a83b0 : SUCCESS in 26m 18s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/774ff6280f674ca4ac6fbd9e6ece8ed7 : SUCCESS in 1h 25m 30s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/1db6bc15dd4c4a84935278352344ccd0 : SUCCESS in 1h 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79633314fa51ae47a5510bbec9dd0ff49e7cb26f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-04 20:46:47.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d1bc7c84f33023c555ec973c2369536b805db12b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 22:13:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9527c4d5ad924b6e91d1fee0250040dc\n\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/704f533b6e5e42ad8af6316070ba1727 : SUCCESS in 44m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c936e2b3e234d14b046f72a7bff560e : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5903590391a64c77a678ae681a4935b3 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b699f358e4214d3e84008741658428d6 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e5e1a06636d4cdb8a7e9926039cfbce : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1e750429d2548129b285e8bae6757ff : SUCCESS in 5m 59s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/858eac6abd314415917a885f8b813c02 : SUCCESS in 1h 01m 11s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b723a88f95e44f0d95f12ceec1ac8569 : SUCCESS in 4m 02s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/0ef607d081614774bf3e7323fb07d197 : SUCCESS in 5m 05s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/d84f0df57f1745d9ad3cb413da6c9c6f : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f463841fcc44d7090623d0101b9139f : SUCCESS in 8m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58e9c7bc84714f8aa4319261bc960ea4 : SUCCESS in 1h 04m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8e5873aac4734f24a8021aa9a7035bbb : SUCCESS in 1h 05m 18s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/616e1d533d1849efa7923f106db00f7a : SUCCESS in 1h 24m 47s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e00abdb7f92c4113943fe65b7a120e96 : FAILURE in 1h 04m 42s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/1bae5b52a3a34856968cc605e2f4739c : POST_FAILURE in 15m 43s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/86ddac37aa0e4f05816d9f25c2398d31 : SUCCESS in 30m 12s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f408de057da941b3967d2072d31c372b : SUCCESS in 50m 35s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6fa6afd9cb2046308da7505122545fb8 : SUCCESS in 1h 20m 26s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/00bb1e0576294150bb979a6419b959b4 : SUCCESS in 1h 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"75b87d0b5e55c1f6e059b93d5716343a76d892eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-03-29 23:35:05.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4bb211c6e06dd8e05d3b449badf9ca5540703b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 01:09:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88ec9317618a47b1a07e957724986f0e\n\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/91ec5ef3f8ef4b38bc22f139cf7ac451 : SUCCESS in 44m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc7574ef9ab487e903d7f6a8dda08ea : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae26f6a2bb3c495498c4eaa951620517 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c2b55bd226742afa3633ca3827509be : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/919a2801e62442049424d12105374cc5 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba0b2cf0be754fe4856748b6b862c57c : SUCCESS in 6m 41s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5d9eb5ed5a59448eae61143967ee5dfd : SUCCESS in 1h 01m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2708c3670b454371b85f0d7a00bd4adc : SUCCESS in 4m 00s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/e4425c45ac2a43d29ffd5123ea8fd39c : SUCCESS in 9m 13s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/9c785ba66ab44cab9b12f196f1244a33 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc8fd8d8ede488faab2f122ffbb5edc : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb5c5e957da64d0e9ccb0f6f0920e3a9 : SUCCESS in 34m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/67933adac07f42b6a27409d23762ca49 : SUCCESS in 1h 11m 11s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e5677012299943b9a98b260971e915bf : SUCCESS in 1h 29m 22s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/288b704585824be8993ca834f55a9e01 : FAILURE in 58m 28s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/75a595e16d544244aa145bd16068a747 : SUCCESS in 32m 44s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/d09bc539cf65455ebbe1df7473c286df : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3fd8464a11414cd4aa8b9c70262fc4c2 : POST_FAILURE in 13m 43s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8d837889da014592aac76b5e5b0366a9 : SUCCESS in 43m 36s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/57683f28a8a04cbb9170d2f5185cc3d9 : SUCCESS in 40m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b604f65b9a77f3b18d269f233caa5ae66c7619be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-04-12 17:06:51.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e407640f1e33c6d2447f60dadddcc1201d961953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 18:28:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1f0a1ea066438ea9406eb3dff602a6\n\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/c5021247d6e841ba88d91bb60f68c217 : SUCCESS in 51m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631326f52ab44850bfc3f35dc9a4ba42 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4373704a8bd44ef29061d2b24b70da2e : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8332a589e2c84005a1868ce98be8cea1 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19a1cd3686c34ab78f0455385a7b6c99 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e78e4a6b98b74747ac08819c259a0c8b : SUCCESS in 6m 04s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/674a186e832e4e2d8010316c1c09fde8 : SUCCESS in 1h 04m 49s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3948572432274c368afa8e9465d92e1f : SUCCESS in 3m 23s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/9d7d9bb50b40417190889f09b61728b7 : SUCCESS in 4m 29s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/a644035e3b3742d19fb1c07ed807773e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd8a44b20b245a1844f9adb79addf34 : SUCCESS in 6m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d9e0c3d08234748ba2214eef39dbfd9 : SUCCESS in 38m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ac4afc11ec6d4b3ca29f06acf8b5cd12 : SUCCESS in 1h 11m 19s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/98c66bcf9959495eb14a95f577bfcb3f : SUCCESS in 1h 13m 13s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e1fca37738fc4e2a9b0f8a04d08bdd04 : FAILURE in 35m 18s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/4fb1d8ca79704c4e8aa9fda54b9978c3 : SUCCESS in 31m 55s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/eeefa4b69e8f40ccaa972e61e41ce5f7 : SUCCESS in 14m 15s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/08219187bfad489d9683ace3e1f11a3e : SUCCESS in 16m 43s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7233ccf5181b465198f5585aa97b4545 : SUCCESS in 26m 52s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a48b877c0bbe4163b3a6b4e15d53496d : FAILURE in 58m 34s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/b31a76179fe948dc8606b948e80689c4 : SUCCESS in 1h 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"03e9c91c972c731e858381297ba6cf4af92adc21","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-07 12:10:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"7e7110169839f2f44cd164fdfe6b305c11571806","tag":"autogenerated:gerrit:restore","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 20:43:05.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"3ebc526a6d53a2ae568dbbe8da3d66f13e00e383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 20:43:41.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"48f105656d812b4da8a29083f24f54ae05705cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 22:25:42.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/986962 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"0bc6519f2071f3732639132d0b0129f589dd9079","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 22:28:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"462c26a15e27d65f205bf896c17d06d40edfab25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 00:06:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ad8a7482384708960ad4bdebaf2d7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c92d39d529a4097a33773a9e8c8f7a4 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ce012b623164c38bf3bf2e881bdff19 : SUCCESS in 7m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e62cff8dd61415a93c241353f1a112c : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a01fec36675482c81ff5b2892b670b8 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38205b8712f84e2283d511490efd9b58 : SUCCESS in 8m 51s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6a0cea4f90dd4263a50abe014510a71e : SUCCESS in 39m 20s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d4444b525bc14627992e48d1416d5360 : SUCCESS in 4m 13s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b9612ae4d52c4a28a8f6a8dbcfffb021 : SUCCESS in 6m 58s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/d6d795fcc8164df891e4233e6e89f0cb : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9227d4af567341238593ad164f322ccf : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aa4ecff9b9e442fbfbe8d4ffa9df004 : SUCCESS in 3m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/511836d5b28242dd84be8b9297e95fbd : SUCCESS in 1h 07m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4d27467c80bd4ae79aa25e6973a4d42b : SUCCESS in 1h 26m 46s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/480dc9a2e10840518461f5627e64902a : FAILURE in 1h 30m 25s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a0d5ee5ece20444ab248b9350e756832 : FAILURE in 36m 33s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/9acf007305cc42c7a2b8c60368cd3e5b : SUCCESS in 34m 33s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/b635aeee9796417f86b62b85d13cc431 : SUCCESS in 14m 41s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ffdd2bfe53d94166b6de90a5cef39de8 : SUCCESS in 28m 39s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e85d0e2436ce4247ba48516fa6ea7414 : SUCCESS in 24m 13s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d292d44b88a54b109c7019089a6ba9d9 : SUCCESS in 48m 08s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/f5924767642d45ef945f664ed4063537 : FAILURE in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5ca504b5ed2a1fd1b18991c2f97f6d1fc149dc99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-10 21:03:18.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"80538e962c44cc226bc0180fc44399edcf325098","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-10 21:04:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"07e367062610a5e7df60bbf9dd9615cb874953a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 22:39:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93cb3c57c6f840349e0dedf85e6ec385\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bd09de46a3040c5a044d1b38d03f956 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37f81234cf694f6fad17dfdb45a813fc : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee817664f4cb4935975ea027803b3a77 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/723ded4abe3c467497a44df07f979a10 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b2cc7a5565e4797a6bdb6c84d0f407d : SUCCESS in 4m 21s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6cd59288af1549999963fce039208a39 : SUCCESS in 1h 11m 44s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2783a2b067b3477faaeceab39470ce44 : SUCCESS in 3m 18s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2510ead7d4814e7281636260db059885 : SUCCESS in 5m 26s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/4dbe84a7dacf44a1ac3436f0137ceefd : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1975144c370645bea7cee2e79ee46348 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a083321eb6c436fbf819af691f999a7 : SUCCESS in 6m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b37e0e67ea154001bfee33288e14dbfd : SUCCESS in 39m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cf190e68022349cd869d5688849d9a85 : SUCCESS in 1h 11m 00s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f3f90cb089c44badb570fcb67f984aec : FAILURE in 1h 05m 09s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b8209d7c1b7a4b55bda2acf1c40d8d15 : FAILURE in 43m 29s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/accb377e161d4726897cb9ff715589bb : SUCCESS in 24m 43s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/a9bc22e2a1ab450d804e4672400ad81d : SUCCESS in 32m 32s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6656e3eb2eac49ebb9afd6af987b4b2e : SUCCESS in 19m 15s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5a5d17fca5fd45be96137471298c13b3 : SUCCESS in 19m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f9dc43f65f5e48d89add3e8aed78ff5f : SUCCESS in 1h 04m 56s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c8de384baf4946e1ba55be866a03ad92 : SUCCESS in 1h 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a4bebcca4beddacdfb8bd1ab65a677ccee195352","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-18 10:49:38.000000000","message":"Abandoned\n\nI think we\u0027re done with this now?","accounts_in_message":[],"_revision_number":7},{"id":"9889bd9c4a9abde88f2f02302ab6b756dfe8d5f9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-18 11:02:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c40c60c49152668f4623c426a816934bf4776ba9","revisions":{"697afaf7481303d66ba9fef29a8e423ac2e38616":{"kind":"REWORK","_number":1,"created":"2026-02-18 14:43:06.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/977226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/977226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/1"}}},"commit":{"parents":[{"commit":"e0845449d3570e0a6bc90660440fb6f46721fd2b","subject":"Run mypy from tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e0845449d3570e0a6bc90660440fb6f46721fd2b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"subject":"DNM: Test with dependent neutron change","message":"DNM: Test with dependent neutron change\n\nChange is removing tenant_id support from subnet API.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/977172\n\nChange-Id: I6d4d1d12853251d0713113568047648189916c31\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/697afaf7481303d66ba9fef29a8e423ac2e38616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/697afaf7481303d66ba9fef29a8e423ac2e38616"}]},"branch":"refs/heads/master"},"806223bc9ea82143f58d1183b3208715c7f82560":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-01 21:25:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/977226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/977226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/2"}}},"commit":{"parents":[{"commit":"e0845449d3570e0a6bc90660440fb6f46721fd2b","subject":"Run mypy from tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e0845449d3570e0a6bc90660440fb6f46721fd2b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-01 21:25:08.000000000","tz":0},"subject":"DNM: Test with dependent neutron change","message":"DNM: Test with dependent neutron change\n\nChange is removing tenant_id support from network API.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/978363\n\nChange-Id: I6d4d1d12853251d0713113568047648189916c31\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/806223bc9ea82143f58d1183b3208715c7f82560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/806223bc9ea82143f58d1183b3208715c7f82560"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b3ef7447b989ec658e9ada32e62eaa0692129850":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-04 20:46:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/977226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/977226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/3"}}},"commit":{"parents":[{"commit":"e0845449d3570e0a6bc90660440fb6f46721fd2b","subject":"Run mypy from tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e0845449d3570e0a6bc90660440fb6f46721fd2b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-04 20:46:45.000000000","tz":0},"subject":"DNM: Test with dependent neutron change","message":"DNM: Test with dependent neutron change\n\nChange is latest removing tenant_id support from neutron API.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/978963\n\nChange-Id: I6d4d1d12853251d0713113568047648189916c31\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b3ef7447b989ec658e9ada32e62eaa0692129850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b3ef7447b989ec658e9ada32e62eaa0692129850"}]},"branch":"refs/heads/master","description":"Edit commit message"},"20f9e6d3c167dba4275a98583123a400ccc4d74a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-03-29 23:35:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/977226/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/977226/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/4"}}},"commit":{"parents":[{"commit":"e0845449d3570e0a6bc90660440fb6f46721fd2b","subject":"Run mypy from tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e0845449d3570e0a6bc90660440fb6f46721fd2b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-03-29 23:35:03.000000000","tz":0},"subject":"DNM: Test with dependent neutron change","message":"DNM: Test with dependent neutron change\n\nChange is latest removing tenant_id support from neutron API.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/982593\n\nChange-Id: I6d4d1d12853251d0713113568047648189916c31\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/20f9e6d3c167dba4275a98583123a400ccc4d74a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/20f9e6d3c167dba4275a98583123a400ccc4d74a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"234122e8fb36c5a13590c0b816714d2caae13298":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-04-12 17:06:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/977226/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/977226/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/5"}}},"commit":{"parents":[{"commit":"e0845449d3570e0a6bc90660440fb6f46721fd2b","subject":"Run mypy from tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e0845449d3570e0a6bc90660440fb6f46721fd2b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-04-12 17:06:50.000000000","tz":0},"subject":"DNM: Test with dependent neutron change","message":"DNM: Test with dependent neutron change\n\nChange is latest removing tenant_id support from neutron API.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/984202\n\nChange-Id: I6d4d1d12853251d0713113568047648189916c31\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/234122e8fb36c5a13590c0b816714d2caae13298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/234122e8fb36c5a13590c0b816714d2caae13298"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b8a31e94d8d3e8a9c8ce00f536d398dfc01f18f4":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-06-09 20:43:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/977226/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/977226/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/6"}}},"commit":{"parents":[{"commit":"e0845449d3570e0a6bc90660440fb6f46721fd2b","subject":"Run mypy from tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e0845449d3570e0a6bc90660440fb6f46721fd2b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-09 20:43:39.000000000","tz":0},"subject":"DNM: Test with dependent neutron change","message":"DNM: Test with dependent neutron change\n\nChange is latest removing tenant_id support from neutron API.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986962\n\nChange-Id: I6d4d1d12853251d0713113568047648189916c31\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b8a31e94d8d3e8a9c8ce00f536d398dfc01f18f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b8a31e94d8d3e8a9c8ce00f536d398dfc01f18f4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c40c60c49152668f4623c426a816934bf4776ba9":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-06-10 21:03:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/977226/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/26/977226/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/26/977226/7"}}},"commit":{"parents":[{"commit":"e0845449d3570e0a6bc90660440fb6f46721fd2b","subject":"Run mypy from tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e0845449d3570e0a6bc90660440fb6f46721fd2b"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-02-18 14:42:11.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-10 21:03:14.000000000","tz":0},"subject":"DNM: Test with dependent neutron change","message":"DNM: Test with dependent neutron change\n\nChange is latest removing tenant_id support from neutron API.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/986962\nDepends-on: https://review.opendev.org/c/openstack/openstacksdk/+/992536\n\nChange-Id: I6d4d1d12853251d0713113568047648189916c31\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c40c60c49152668f4623c426a816934bf4776ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c40c60c49152668f4623c426a816934bf4776ba9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
