)]}'
{"id":"openstack%2Fopenstacksdk~983908","triplet_id":"openstack%2Fopenstacksdk~master~I42dbdf0fb9c4f22f7deffed0fe8efa7da6969f65","project":"openstack/openstacksdk","branch":"master","topic":"bp/manila_shares_attachments","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-08 12:14:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I42dbdf0fb9c4f22f7deffed0fe8efa7da6969f65","subject":"zuul: Have magnum job derive from base job","status":"ABANDONED","created":"2026-04-09 16:23:03.000000000","updated":"2026-09-08 12:14:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3c05cc2796a8f8315fcc8b79b688fddacd725bd6","_number":983908,"virtual_id_number":983908,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-07 23:58:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-09 17:46:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c42a97c6d7549ae09553d8d82340b98445aef48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-09 16:23:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ca542f5ebd4208bdeeb9d080ae32007d971c622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 17:46:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8136db931543d8a7730a1d1b34ccc0\n\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/6e49d0c864fe4ef4856e343ed6897492 : SUCCESS in 27m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6494f7a47e412aa45720ac18ff7c3d : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81773748f1b949cfb03e608194f95c59 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0634b9dee6d54942ac41d103e8b5a419 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/808890b77888440eb324c356f7384865 : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06f438ebce024852950acee4318f78e4 : SUCCESS in 8m 53s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c4157e70eba64877b086007aeeac6e61 : SUCCESS in 1h 09m 02s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c98191e45a584e2b8fb5ffe2bd05acbb : SUCCESS in 2m 41s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/d59bee430c734b6ea9066be856d3dacb : SUCCESS in 4m 38s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/ed8ac7b709da4e55a67006faecb72c63 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aefdc357a5b4735a8b8cbee27c3b27c : SUCCESS in 7m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e804fb8523084d1f91271b19485c8734 : SUCCESS in 44m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b80dc75f84ac48de9cb01a057783c7bb : SUCCESS in 1h 19m 49s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/0070844063bb40168cf7ec05c7a65aca : SUCCESS in 1h 18m 48s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b354c68682d14110bf200e3ea5d357e6 : FAILURE in 56m 21s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/53c7379335984c5aab620994176bd2fb : SUCCESS in 28m 17s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/0a82daab50cf41a795fa0c782337e924 : SUCCESS in 30m 15s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/7b435283f5954272bbe929bca9b1a8cb : SUCCESS in 25m 12s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0fac28a11a0d4b5889af7ab428d5fb4c : SUCCESS in 34m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3b8fae0de732465cbdcfb685589a7453 : SUCCESS in 1h 04m 03s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/45faf67260054446a42fd93388c2af8b : SUCCESS in 1h 09m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d63e7e96df34d36a5a0da32e24061f73704f7467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 22:20:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7d768a656a75012818ca1ce8ae92701910bdeea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:58:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41b89268a1bf4f2b8bcd14cf90068296\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c754f25eeda4461bad98e355be9219b9 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd9638167ea943abb6e86c444f3ff16c : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ba86d0de7f34cc7896036d7923486ae : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1b62c561c1f46a5812ad8b228b8a5c2 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc6f9fc4d98747ee84bd695e054454d7 : SUCCESS in 6m 31s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b909ecb1f35c4c81a219134b3b470061 : SUCCESS in 6m 59s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b66db8c417e346a88d3054fc3a560460 : SUCCESS in 32m 10s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d2e0fd5b41a342749c84d2b3f4823194 : SUCCESS in 4m 12s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b20aba126a7049619d716e94cbe558d4 : SUCCESS in 4m 45s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/178af0fcb3104007b9dfc40e10213703 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9536b9309dd4440b1c7bbec44ee58d8 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f7c310acf8e424ba5c5da9283e2068f : SUCCESS in 2m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfe4c56c4ebe4f218e81ab97e58e684e : SUCCESS in 31m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/72fadfec81fb4091ac0b38147014f380 : SUCCESS in 1h 17m 24s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4ccbeccf6dc546bc8f529293593115ef : FAILURE in 1h 28m 48s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/cc0b26ac0e9545889bd3c802c3a0e07d : FAILURE in 53m 37s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/2d9b002831e14a8388c932f5a1354a02 : SUCCESS in 24m 14s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/6922f6041ad541199569087b148ab0ff : SUCCESS in 17m 58s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/6ad0425d38764ae6b4b37a13662e9a1a : SUCCESS in 29m 27s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/34a56ae936e7422e913a80a3515308b9 : SUCCESS in 23m 09s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/000d686e762a45228a862c00ff3334b8 : SUCCESS in 36m 58s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/222788606be54419b0e458aed4408c7f : SUCCESS in 44m 25s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/a8b37044088c499483b2d43e98c6e836 : SUCCESS in 1h 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c05cc2796a8f8315fcc8b79b688fddacd725bd6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-08 12:14:52.000000000","message":"Abandoned\n\nSquashing this into Ice3c0cdbd77aa958635115afbddb176d170b9e53","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22851cff312050f7800d71439de8b2356e2e053d","revisions":{"40940746cc92a28c202ee2dc326bbf5f059c9746":{"kind":"REWORK","_number":1,"created":"2026-04-09 16:23:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/983908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/08/983908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/1"}}},"commit":{"parents":[{"commit":"f27f1002d0f5ea2814140b8639c95dc80d3bf251","subject":"Add shares API to compute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f27f1002d0f5ea2814140b8639c95dc80d3bf251"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-09 16:21:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-09 16:21:48.000000000","tz":60},"subject":"zuul: Have magnum job derive from base job","message":"zuul: Have magnum job derive from base job\n\nThis job does not need heat.\n\nChange-Id: I42dbdf0fb9c4f22f7deffed0fe8efa7da6969f65\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/40940746cc92a28c202ee2dc326bbf5f059c9746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/40940746cc92a28c202ee2dc326bbf5f059c9746"}]},"branch":"refs/heads/master"},"22851cff312050f7800d71439de8b2356e2e053d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-07 22:20:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/983908/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/08/983908/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/08/983908/2"}}},"commit":{"parents":[{"commit":"26640114dbf2e08c63f3ddf0b6935b3ff38354d3","subject":"zuul: Make openstacksdk-functional-devstack-magnum voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/26640114dbf2e08c63f3ddf0b6935b3ff38354d3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-09 16:21:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 22:18:51.000000000","tz":60},"subject":"zuul: Have magnum job derive from base job","message":"zuul: Have magnum job derive from base job\n\nThis job does not need heat.\n\nChange-Id: I42dbdf0fb9c4f22f7deffed0fe8efa7da6969f65\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/22851cff312050f7800d71439de8b2356e2e053d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/22851cff312050f7800d71439de8b2356e2e053d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
