)]}'
{"id":"openstack%2Fopenstacksdk~991204","triplet_id":"openstack%2Fopenstacksdk~master~I593bb60a6f0a3bb137c5b7ce1f681689d9cf8f18","project":"openstack/openstacksdk","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-03 13:17:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I593bb60a6f0a3bb137c5b7ce1f681689d9cf8f18","subject":"typing: Add annotations to remaining methods (orchestration)","status":"MERGED","created":"2026-06-02 19:36:11.000000000","updated":"2026-06-03 19:25:42.000000000","submitted":"2026-06-03 19:25:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991204-typing","meta_rev_id":"608cb4328785214ef5da33af82d8bb66966c879a","_number":991204,"virtual_id_number":991204,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-03 19:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-03 13:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-03 13:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-02 20:13:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"050d912a3e1f4dc2dee16fac573f9150d5b746ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-02 19:36:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ae0c1b27c37dc64e0c0fbc6e954164e9e0cc1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 20:13:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8af5ecee34c4e6a8a1ac767b1d45fb6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e7ce3eb3a75d4af3a85eed3398a4e7f4 : SUCCESS in 3m 34s\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/bb0208d9bc5240ebae3ae93c18a315c9 : FAILURE in 8m 32s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae7d98e35dff45fab7a4329eeebae309 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00db9b90d40a4cf9ae4c2faa8c96d3a5 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/541f0a68c6b242b1ad4540f39dbd3e99 : FAILURE in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db0331b6c1ee445fb7aeb1d650200d2b : FAILURE in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/465f576872e7492db90899902bd0f4ef : FAILURE in 4m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fcc52a31887e483f8cf35b7d9d6f731f : SUCCESS in 6m 58s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/df797a4544b64d8e8c218131253dddce : POST_FAILURE in 18m 20s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f60027ad62584054ba53c68be40dcfe6 : FAILURE in 4m 47s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/48f141a290544917be4e0f13113c27cf : FAILURE in 4m 50s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/044e4c4e840e4d40a6c5660ce132d50e : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3fda7b35e647dc95511e8fdbd25357 : FAILURE in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aba879f2a67540c8ab8a94bcb4e80a46 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b8dae9bd8684649a4a756444a20536a : POST_FAILURE in 23m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/be1a1511b77f4615936cbcab6ed085ab : POST_FAILURE in 23m 23s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/fc2f1e7fad1345c58594d4863539f7ef : POST_FAILURE in 26m 46s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/0b8e397cba0047c08bc208fa35d31c7c : POST_FAILURE in 22m 48s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/6a679c2c3a934dd999e40b2ff7d8976b : POST_FAILURE in 9m 09s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/d786607595cb4285a83159ac1e91b1c3 : POST_FAILURE in 28m 02s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/8aebebe2c8c34fffb21cbfc298ab0875 : POST_FAILURE in 12m 19s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d6a3d52265f841e3a1a3c580117aec6d : POST_FAILURE in 19m 10s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/299052d76f204a52920ecbc988dfcc40 : POST_FAILURE in 12m 03s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3dbbee5f12af42f3b5ab1107f989ec36 : POST_FAILURE in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c60c966d61639f1229bc1e665f468dcbb70c2ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-03 10:12:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"26cd39290fa1764f46a33263065c4fc6d5b77297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 11:56:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa2288f423d2462a9024c787fc5cc828\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51fd57ec9eae495583c8f28c6bb6f892 : SUCCESS in 2m 17s\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/839fbaa2dcce4657b908c9445e274eaa : FAILURE in 6m 08s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/a67f5bebd25a4c00a7d4bf24f1d35760 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ba878e4b2542468aa3dcd812f20345 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d38edece30e14069a9fc2f2b4d689650 : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcea9610f83e4e0294c18e620fbdc8f4 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfc85fd202cb4358b4b6082f9efdbf02 : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eac0ac4495cc4beaaa85923aec586a64 : SUCCESS in 7m 35s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d7a8c29f35da46e285c3c5c63ac52f1f : FAILURE in 1h 02m 45s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/98d95374f98946ffbef013ed2d274f84 : SUCCESS in 3m 10s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d4a9d9d8b21e4c3ba7568861a968a2c0 : SUCCESS in 6m 06s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/4d7cae59d95f4d4ea77c87f161331d24 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4347c576a440cea628ebc7abab494b : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4036aa765294104aae1115601456483 : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/295c72c86f5a4936a104c04edf29aa20 : SUCCESS in 1h 05m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/303943cb6fe04b6a86ff6c51c10427b3 : SUCCESS in 1h 24m 38s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/965c357b1b9c4d3da388e63977a6d731 : FAILURE in 1h 37m 42s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2ed0bb4abec64edda86193c87f84cef1 : FAILURE in 1h 03m 59s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/db764486265646c1bf00efb216218ebb : SUCCESS in 36m 20s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/3bb32596d9be4c618e706b06fb02e1c0 : SUCCESS in 31m 27s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/4604fed0edf2468b96f4b92f3d01a4cc : SUCCESS in 29m 08s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c51761f69f88463083e083470d2c4d3e : SUCCESS in 42m 37s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a4dba579195f403797fd9b2c4a5f7f14 : FAILURE in 1h 18m 06s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/9d0bc9165b9940e7897e65f63f076198 : FAILURE in 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b6ac3d40267ab78ce0d1bcb42f2ef34505a42bf7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-03 13:17:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"586cf205591c9133148c407226229306815de6fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 15:05:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cff78d63cc243f7b779d68c311b22b1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b974d88ee9e74b92aa45c719d3e5f740 : SUCCESS in 3m 14s\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/04cd1885e6b2445fbefbc306a5e34909 : FAILURE in 9m 23s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/9204d3d890264ab0a6ea330d8cdf785d : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b118b38a15c48338435ae403ff75b2e : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc65bc3d4d9a4984a928c4994dbea186 : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55fcde877c76418c870f0d999b2446d3 : SUCCESS in 8m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f32d3ed269747ea98d0ac041babd6bb : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/373c936f2ab847dc82d450875d825a49 : SUCCESS in 8m 03s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/89e76ff931944ccc8d10d9ec9815bd87 : SUCCESS in 1h 04m 43s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6a0b71a88cc34595a9625f47f9c6d052 : SUCCESS in 4m 13s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6ee43b689c10463b9b6c42efe228ff51 : SUCCESS in 6m 53s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f5d97c0f440741d3ad02d1a2cc22fd26 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4267ceeb67fb40858a748b34d1a3493e : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e319bf9479954960bb63535f4bffac6c : SUCCESS in 4m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10f3fccf2fb643e795c85595362f64e0 : SUCCESS in 1h 05m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0fff1dc7886e45bda705e422a24fd278 : SUCCESS in 1h 31m 04s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4b3af3d2fe7543e7add62a23f06ec9ab : FAILURE in 1h 38m 30s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/89ddd32807ee4fbebe9d068ea1af9a12 : FAILURE in 1h 02m 36s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7c62b5e298c5438d9925ac9ec3d842fb : SUCCESS in 35m 23s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/07bd1a3151a04cf0b9d81a12dfae89c6 : SUCCESS in 32m 11s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/642ebb902cf24d3f9a374e1a236df212 : SUCCESS in 25m 04s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1632ba5661534608894d394049a63402 : SUCCESS in 45m 22s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/44edb1f3c4c642b09b233fbaf99cd1de : FAILURE in 1h 01m 08s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/080e154783db498883fa929028a268bd : FAILURE in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"174875962215827d53b34fe88a41703a83f26fa3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-03 15:38:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"284dbe9134a2feedfbd9f4984778d2c8863130b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 17:49:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3307aaa66a2bbeeefda5f0d5f428f2e4a312244c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 17:56:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95b1223dd10f4956a702e77b9ca5238d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56c17d4084a84f4ebfaac9eefc096cb2 : SUCCESS in 4m 58s\n- bifrost-integration-openstacksdk-src https://zuul.opendev.org/t/openstack/build/20ce28becac7475f9f6fd799a52284c7 : FAILURE in 9m 02s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/89df023c818b44e38804e72f0d898f43 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da0d4e0c58dd45448870ec49e689a12a : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3ad5afdb9124d53ba3b9e2e8330fd4a : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/170faf94ee7a40e3878cbd4a96334fb2 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6267b40881ed4e2f9856794ffa5120d1 : SUCCESS in 6m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1ee62b2f0714207bc62cde9cbde337f : SUCCESS in 7m 34s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d8fb54f1229046e5ae3c78898b9c64d9 : SUCCESS in 37m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/26123eb0eac141d9b886f8347f948ff9 : SUCCESS in 5m 19s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5451b71df57448ce9c3d6d6c111f68be : SUCCESS in 7m 47s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b5de21e9377c46c7a445885f237ede04 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2886a85f1494fa292272a1670b4b1e6 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/757ffcd6977442d3a1335bae4d9eb8dd : SUCCESS in 6m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88793470709d49ccb7b29edd8a6bf30a : SUCCESS in 56m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1fef999377a14744a08b07cc6c90cec0 : SUCCESS in 1h 24m 27s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/5f1f5c55891142e1a411b579e51dd068 : FAILURE in 1h 43m 34s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ba94a570c9c04ef08fb7b49966506657 : FAILURE in 1h 04m 56s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e6077ae0e13a4d83af8d62018ce40621 : SUCCESS in 34m 28s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/c72416285c304c25ab481f513d9163e8 : SUCCESS in 35m 52s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ea627d326476403dbc5d7b31134d6365 : SUCCESS in 28m 22s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/02936074941b457e8ff0112c8e5f3cb2 : SUCCESS in 48m 31s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/de40df7f10904eb6a524217fc22450a5 : FAILURE in 1h 22m 52s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0c7c50ec66074a7ea23b7136bd57677b : FAILURE in 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5f38cca71d21c805ef2916b67b98274920c7adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 19:25:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918b9b3c13f545d59928fb4f65b43331\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce72cbe86aaa425eb37d56ab00d6f5cc : SUCCESS in 3m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7cb3d4ab2390475db448ca28d26bd48f : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a72151bb5826491782a11999779c2cd3 : SUCCESS in 7m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67cc659398dc484582d464c8c5b1b10f : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5439ddd9e3e74bb1a4335471b97bdb71 : SUCCESS in 4m 16s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0fdcec21e2e5423fa519f75aa12c055d : SUCCESS in 54m 54s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ddff5ec6740048ba9ebe28bad19d8900 : SUCCESS in 4m 31s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4b50abadd9e641eb932af6765f77621e : SUCCESS in 5m 02s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/242057d5c8624cf1ad770fe51e53348c : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7616c4ac214628a900a5a434af2fa4 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/590f7e90befa488b9ffdce2a99bdc887 : SUCCESS in 4m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/148ffbaf00f74cc2b6d0efd1fa960c40 : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/45b44b2213694a12873a2d843d5a272e : SUCCESS in 1h 14m 18s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f23b2bb5b6804e13a9f91967cc5f45bd : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":2},{"id":"608cb4328785214ef5da33af82d8bb66966c879a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 19:25:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"119f57314adbf7a5ba486a99bafcf8824d8ec812","revisions":{"6769de19c9b893771e6378b692aee8d42d41b1bd":{"kind":"REWORK","_number":1,"created":"2026-06-02 19:36:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/991204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/04/991204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/1"}}},"commit":{"parents":[{"commit":"4ae2d50154732b36fd0e37f7bbf15aebf560047b","subject":"typing: Add annotations to remaining methods (object_store)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4ae2d50154732b36fd0e37f7bbf15aebf560047b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-25 08:42:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-02 19:36:05.000000000","tz":60},"subject":"typing: Add annotations to remaining methods (orchestration)","message":"typing: Add annotations to remaining methods (orchestration)\n\nChange-Id: I593bb60a6f0a3bb137c5b7ce1f681689d9cf8f18\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nAssisted-By: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6769de19c9b893771e6378b692aee8d42d41b1bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6769de19c9b893771e6378b692aee8d42d41b1bd"}]},"branch":"refs/heads/master"},"119f57314adbf7a5ba486a99bafcf8824d8ec812":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-03 10:12:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/991204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/04/991204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/04/991204/2"}}},"commit":{"parents":[{"commit":"f88641745e8dc33e56ad5c1f939aa3ff5e268a0c","subject":"typing: Add annotations to remaining methods (object_store)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f88641745e8dc33e56ad5c1f939aa3ff5e268a0c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-25 08:42:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-03 10:12:43.000000000","tz":60},"subject":"typing: Add annotations to remaining methods (orchestration)","message":"typing: Add annotations to remaining methods (orchestration)\n\nChange-Id: I593bb60a6f0a3bb137c5b7ce1f681689d9cf8f18\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nAssisted-By: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/119f57314adbf7a5ba486a99bafcf8824d8ec812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/119f57314adbf7a5ba486a99bafcf8824d8ec812"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
