)]}'
{"id":"openstack%2Fopenstacksdk~991685","triplet_id":"openstack%2Fopenstacksdk~master~I8a9e118b4aa504ef570b81c38b9d46b6f41c812a","project":"openstack/openstacksdk","branch":"master","topic":"bug/2144617","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-07 10:12:23.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-05 10:05:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-08-07 13:00:28.000000000","reason":"Change was submitted"},"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-08-07 13:00:28.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2026-08-06 15:43:31.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-08-07 13:00:28.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2026-08-07 13:00:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a9e118b4aa504ef570b81c38b9d46b6f41c812a","subject":"network: Add evpn_vni resource attribute to the Router resource","status":"MERGED","created":"2026-06-04 15:52:20.000000000","updated":"2026-08-07 13:01:33.000000000","submitted":"2026-08-07 13:00:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991685-bug/2144617","meta_rev_id":"a1013a2fec3cd8ddc4cde40cd890f88e3828eca4","_number":991685,"virtual_id_number":991685,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-07 13:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-05 10:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-06 15:43:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-06 16:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 16:00:58.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"state":"REVIEWER"},{"updated":"2026-06-04 16:00:58.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2026-06-04 16:00:58.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2026-06-04 16:00:58.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-04 16:00:58.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-04 17:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 14:45:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"3173be5a0098505eab900fba13d786fb63239d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-04 15:52:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f51511340e57c6e986e202fa9cac7b7efa423555","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-04 16:00:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"193f9bb4f8176127da596b1db424bd2c65573680","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-06-04 17:22:04.000000000","message":"Topic set to bug/2144617","accounts_in_message":[],"_revision_number":2},{"id":"934a15b7c7129abaa04ed4ec87b8b818ce8ecb3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:36:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e3b58991cde46d5b017eb0cf9798ba1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69f7e9e7b0047df9902edb17d0b1ddc : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b473f7d293d74f12bb700f05ca1a3ebd : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fda995473ca484483d5b19763464dc6 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd469af399c64558bbd0267934a31219 : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a0b499693254aa39a7f12718eabd0f7 : SUCCESS in 3m 42s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/aa72b93c794c45a2b49752376764c520 : SUCCESS in 1h 05m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2989e07de755443d893efa5e78dc7824 : SUCCESS in 2m 42s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e80318b0e7c04d089b435297a325c94a : FAILURE in 3m 59s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/725e047bd7af419685c8969b752f3b08 : FAILURE in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5113df0a52b645c5bc6e1bcb0d20f169 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d98d96351eb14bc4bd7fa419e4bd770b : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cb223b43d4647a5aa523afb33b4abff : SUCCESS in 1h 09m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9777be1be8214e469110f72ad6408c23 : SUCCESS in 1h 26m 02s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/2cf09fc00f7940c0a6de016a1094748d : FAILURE in 1h 34m 39s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/387eab0960364ae09c16105fa7a45178 : FAILURE in 1h 05m 05s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/10a3698ebf4644609f29ea81ec2d0849 : SUCCESS in 26m 28s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/65698bed040741e9ace059559078d052 : SUCCESS in 18m 10s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/07deb4f83439470e832141163caed499 : SUCCESS in 17m 35s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/51c22b19a4514750ad448b2acd2d77c7 : SUCCESS in 25m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5febb2c0f70f457e9e0bb2baa50cf3a8 : FAILURE in 1h 18m 37s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4492b7cdcdc8403c9a0acd01dfc0f2cb : FAILURE in 20m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55aca9d57be64d96c740415c9bbd4bd654a0a93d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-06-08 14:45:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4d771bd4f0273e92e9748fec6badf37823be0b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-30 17:33:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f77e6bcb58ebbd134fcb93be4c1b89c70f11b6e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 18:54:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc3085f5edf64d3392fd9d1af7f3f98a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e875db4c2b463aae3045e8e2098a36 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f84e4eef13540168336edf4d7dd7b44 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ecc97c99145413585ddb4e98a26039a : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34abb020a3c643c9baa8aa703e5e8231 : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d0f58ca03b84757a14bc3866588d4b1 : SUCCESS in 8m 24s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1bb1a036398649dab87d684a78f18153 : SUCCESS in 4m 57s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/31070c0c6b7c41aabaaa59d004f9bd9c : SUCCESS in 54m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/562e0dfacb274a4eaece96c0f103058a : SUCCESS in 2m 16s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/556895b2a4ef4c5fa3d405413d9fe389 : FAILURE in 4m 36s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/878f2e8425e74347b82b6f2bffae3298 : FAILURE in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aefb68e70e944fa9986b147eee5de9f : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d73dc108076c437e93c0e12baeba947b : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ff378dfd2934c98afa85b6a007e38da : SUCCESS in 59m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/404a3eff006c4110af3cd882d83025bf : SUCCESS in 1h 06m 58s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/2b8cc2c404dc4cdc9a2eb7c1ddc70378 : FAILURE in 1h 15m 09s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2d596df166114c1aa12701b6045d2545 : FAILURE in 51m 32s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/3631cff0aed84677954f4935255c7036 : SUCCESS in 33m 26s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/691da760613844a9a0f3683010b6792b : SUCCESS in 28m 08s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9797839e62ed4fb2b08d7197a091fb01 : SUCCESS in 15m 46s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/736decdc349747eea119c1b9255fe26e : SUCCESS in 40m 59s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1438040e9cb54ee4959bb6d541a2e341 : SUCCESS in 1h 15m 43s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4f0cf79ae6694bb7be8bc9af2d3bcdc0 : SUCCESS in 1h 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"711b8e235447e0241432ec53907781faa90fa6bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-30 19:24:03.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"85ab20f9944ac8a2a546b87e87616e6849b1fee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 20:56:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26f21f3a342433eb6abfeda51a0bacb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab6e283dd7e443e9cd0b8822ec20372 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/422268c1ab584c868004b144faf861e0 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec3e9614fad64d77988375ed7f2371ad : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8880fdeb6d984d46a28dffc997ec2913 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/919aa6476d6d4fe79fc3c23596fee9d3 : SUCCESS in 6m 21s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/70db4f732dcf4313b6aa8708338ee6c9 : SUCCESS in 7m 10s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7841f989ed9a4d57b8c396b4a1011266 : SUCCESS in 31m 42s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/81fbef038696404ba6cde972af8b7eef : SUCCESS in 5m 31s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/02bcd372c59d48da88eddb6041c22643 : SUCCESS in 4m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/28ece9d5179d44fdbe74329f5332e34d : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888c28f278b944479983cd3a0933958e : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fcaf4a5dfe74ae3bfdad12c1d45f1fb : SUCCESS in 4m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddd95641a4b7466fa3172bb03dc8cd48 : SUCCESS in 1h 03m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/58242b7676cf49bab3c6aacd40c6e327 : SUCCESS in 1h 25m 22s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/799e624d1af345849d84e645b285db48 : FAILURE in 54m 11s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/1aa720d0960f4cd4ba216139f8fe3ede : FAILURE in 29m 51s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/bed77ba9c50540d3920181e5b17c5c25 : SUCCESS in 30m 50s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/edb2747478dd4bbcb782af60aa492767 : SUCCESS in 25m 30s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0e05f7a30c794caca780ee7dc163eb16 : SUCCESS in 31m 35s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5288c4e58ae94c1f965a19ae961756e1 : SUCCESS in 38m 31s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/652a7d6ec2914be49ec07ef573fa6fd3 : SUCCESS in 1h 07m 54s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/86e698c40f8c4335baa5cbe9c3b932dd : SUCCESS in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c517611e8e4c02b906db1c6b9d55048028f970a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 10:05:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"00ed47b7f536805d518a7cc16e48037d1ea1b084","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2026-08-06 15:43:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e28d3a881b5a705fd5955b95a5708ca64b43e1e8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-06 16:36:06.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1ceedcc11e8246db3fe38f47a55b0b49b31cefbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 16:36:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fccb5caeaf69f1c5548ebde5d675516c60418181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 19:23:53.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed1278d0adc246eb950c65637b8a406b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676a8ffcdf8c483e957c79487bfba4f3 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31ac13bed43641d99e9e16496c2b7788 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6848bef269e42b3aca3951a59d7b05a : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4de3b2e26fbd4f74a4aecc7f386fe48e : SUCCESS in 7m 14s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e4cd2a421a834412842ebf72bf56dfb2 : FAILURE in 1h 03m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a91473bc35c14c32b78e0d84071ee64a : SUCCESS in 3m 25s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0ebefbbd303e41f686969433c5224c42 : SUCCESS in 4m 01s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/3c6bedb2590a4378b4ff51bfdbbbc176 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ce36b759c94e27aab61ff094bb279b : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b628e1d30859478e8f389f098ffe3563 : SUCCESS in 4m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaf3ecbf9f624f459f29e2472c1f05ba : SUCCESS in 52m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/08cdeeca9bc14b94a35ce32f27572cb1 : SUCCESS in 1h 28m 45s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/cb8131a62bd24eaeaa73899b47e3f9b9 : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":4},{"id":"0016f1b39ef15843ec627153e1857fc8cb25a415","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-08-06 19:27:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b80adfa248a9cadaa3d9a9dfbe4e5432a319f0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 21:01:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431937ab2cef40069e5ca63c429e2572\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf8ce28671e4768afa70e83b106c414 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63fc15d14cc445b49f7fa11f4bbfa3d5 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bbfba07018bd4fbab54088e20bce91fc : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c89410c0d7b2405b8b6080116e4e8119 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/223fcd5845a241eb94b4c0ebffbedc83 : SUCCESS in 6m 23s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/62aaed1930064af2a7fdd7ab7b768ee3 : SUCCESS in 5m 28s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/470db3e79e504146b869992e6e2dd3aa : SUCCESS in 35m 26s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/64a89b9c98a04e10bf9d205318a8b863 : SUCCESS in 3m 43s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ffb166dca76444f3954c1ead1e75d24c : SUCCESS in 6m 55s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/4bdcc1ea2c6b4ebc8474097d8e4beee0 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8d178d0dee4c40a24b9f746b95a512 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ec674f25dbe4277b20062a4d55bba2d : SUCCESS in 4m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/833ef5b81d5d40118dc809e75dd2d669 : SUCCESS in 1h 01m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/93a3b247fb964ddea22292f40018fe7f : SUCCESS in 1h 22m 28s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/827fd64c32004028a81e9f7b37ab3415 : FAILURE in 1h 25m 38s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/db0327bcea484a05b18b5b2425a6703f : FAILURE in 34m 20s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/539c394b6b9348e6a2f7ff345a6fc2c9 : SUCCESS in 30m 02s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/36687e3e6754401ea5c08b0fed843a5e : SUCCESS in 28m 42s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/49edd21c8b304051b906a7438133e5cc : SUCCESS in 24m 31s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/9695d2b5949d4a16b922ae0e74df3722 : SUCCESS in 40m 42s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e03947f3c5644d709d2c987e8cfd675c : SUCCESS in 1h 12m 54s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/825c3f63dfec446aa1175542f312d24a : SUCCESS in 1h 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e23265cc214ce68e640dd71eae5ecb5bc03b3816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 21:02:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"563ab7f77cbfc8dbee8607bed4cf0cddd19be363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 22:14:03.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9096c3cd8e514c669b610ce0db46b727\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d37a56316944b2a9aaeed4bf3da5c5 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6114f2d2ad244f5dbc5837b8b68bad87 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d1ad4f3d72b439a81d55d06b8f2587d : SUCCESS in 5m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3dfc593d435c42a7ac305d1dfdc8c854 : SUCCESS in 7m 13s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a28b239a1e1b4371829dd804d14bacfd : FAILURE in 1h 06m 39s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/94e9d6a3d10b4bccb8d99e966f90b4ee : SUCCESS in 3m 50s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4356722d53574e6d92fab581cb41c40f : SUCCESS in 4m 13s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/10992145c61c45a0a5cdda17e8feefef : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd50bf2e800548d0af1614e4a9aedbbd : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fe0c6e85e71455ab8044a772a238d30 : SUCCESS in 4m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78f8b759c4f54e8bb73a256a8a093976 : FAILURE in 56m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d8d0496c257145c48430cdf981be678f : FAILURE in 52m 47s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/21a3a36e46e946a8872e56f780f6d4cf : SUCCESS in 42m 55s","accounts_in_message":[],"_revision_number":4},{"id":"f420a6501e4d8dfab28d65ef889415d85239aa01","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-08-07 10:12:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2cb451f7fbd950bab3b0f8d2c08b19c607031d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 11:44:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4fc9dc9469141d69412736815f7adec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abcb3865fc30431a9098a7a7f7aa25a3 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f91913a46df04a65af63aef011128919 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65919e8cce4646debcd2561483cad84d : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bff96f4375f94dc3bbe9e115b56bf76f : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd50d1c6f1ca4cf89babb0457cd0ec95 : SUCCESS in 6m 38s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ed8b90b7abe6477390167e865959fe65 : SUCCESS in 6m 46s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a01ddd22c1874420983101c356bf3c79 : SUCCESS in 59m 46s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8cabba7170c44d499d415a36af6970c4 : SUCCESS in 6m 28s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7b0ebe86ff194ec49657bd4610997465 : SUCCESS in 4m 20s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/912c92faccc44263bbc34e522e06d50c : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ab406cdbb349a4835afb62dbb5ea80 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a02d8ab804f48358fcc26d2a0ea882e : SUCCESS in 6m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7ce32d1d1fb42ab82e33afaae826056 : SUCCESS in 58m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f51873ff1e8745ad822d6bc96b7d236d : SUCCESS in 1h 23m 41s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/531ac41e69044487afb9dbcd89d31dbb : FAILURE in 1h 24m 37s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/4d25691db4ca414ba664c109df213bd6 : FAILURE in 1h 00m 54s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/bb24fcb160874b3a81b93d082e984ed7 : SUCCESS in 35m 19s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/2c63675c4a9d4e5fb097c5b21c9c95e7 : SUCCESS in 27m 46s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/dbe45faf934d425b9ccc5e36f7dc16db : SUCCESS in 29m 11s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/2d5340633fd7456297f78b5f292a6036 : SUCCESS in 40m 48s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0330761fdbe940a9817b360078a7259b : SUCCESS in 1h 05m 09s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/1638c365a9e742abb0a6e0151c1fffb6 : SUCCESS in 1h 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"25dd6a5ef25154743863374cda419b0395e325b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 11:45:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6028f56b5d41fe6d8915e4a518987a147899f71b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 13:00:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a772c7afc14a52ae07c456822f055c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e53299f5a6a4143b28a3c67861db8aa : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/348bdf0d40c6466ca5e27f345b80b58f : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8f8b24631af46c3813c49eb53dd1e48 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5aeed033db43429f96a0c6086647a68a : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cb7f6dd9b2eb44358a02611d5190cd94 : SUCCESS in 27m 31s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4fe8bb416a104f21a6f78d0c3085770a : SUCCESS in 2m 46s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/528cea5fa7514c238a40656143776e17 : SUCCESS in 3m 23s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/324cc7cd6b5d42b5b668c17001e88897 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6089158d6cc8486280191d86541ffed2 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d67bcc1afb14b8782f795e6af4481d4 : SUCCESS in 4m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dddb6a4d8cf44d58ed3d1c1bde1506e : SUCCESS in 42m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/77f4ec4876734ec1a544a3921a30842c : SUCCESS in 1h 13m 55s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/a7cfa4cf0409438d87532e9f3e767464 : SUCCESS in 26m 37s","accounts_in_message":[],"_revision_number":4},{"id":"472234e0c8fc8bf3195d6c5b28b11ad89db1a89b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 13:00:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a1013a2fec3cd8ddc4cde40cd890f88e3828eca4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 13:01:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7200c7cbf1d6431a8069e98c8a7ec831\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c739081bb74f5ca43ade4568bf8391 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bf22bb239e25de406d28ed6f791d2d4de68aab2b","revisions":{"6f522c4d76478f59b9106fcdae47a07c956a24ce":{"kind":"REWORK","_number":1,"created":"2026-06-04 15:52:20.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/85/991685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/85/991685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/1"}}},"commit":{"parents":[{"commit":"c39888aa88d48bc301baa1dffa28dfb2d1411435","subject":"Merge \"typing: Invert checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c39888aa88d48bc301baa1dffa28dfb2d1411435"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-04 15:45:29.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-04 15:47:23.000000000","tz":-240},"subject":"network: Add evpn_vni resource attribute to the Router resource","message":"network: Add evpn_vni resource attribute to the Router resource\n\nThe patch adds the evpn_vni Int attribute to the Router resource\naccording the spec [1]. It will be set with the openstackclient\nwhen --evpn-vni is passed when creating the router.\n\nRelated-Bug: 2144617\n\n[1] https://opendev.org/openstack/neutron-specs/src/branch/master/specs/2026.2/bgp_evpn_type_5_route_support.rst#changes-to-cli-and-rest-api\n\nChange-Id: I8a9e118b4aa504ef570b81c38b9d46b6f41c812a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f522c4d76478f59b9106fcdae47a07c956a24ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f522c4d76478f59b9106fcdae47a07c956a24ce"}]},"branch":"refs/heads/master"},"c2e5e9cc3ea854a0fe1c8145b4816c978dffbf4e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-04 16:00:25.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/85/991685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/85/991685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/2"}}},"commit":{"parents":[{"commit":"c39888aa88d48bc301baa1dffa28dfb2d1411435","subject":"Merge \"typing: Invert checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c39888aa88d48bc301baa1dffa28dfb2d1411435"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-04 15:45:29.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-04 15:59:56.000000000","tz":-240},"subject":"network: Add evpn_vni resource attribute to the Router resource","message":"network: Add evpn_vni resource attribute to the Router resource\n\nThe patch adds the evpn_vni Int attribute to the Router resource\naccording the spec [1]. It will be set with the openstackclient\nwhen --evpn-vni is passed when creating the router.\n\nRelated-Bug: 2144617\nAssisted-By: Claude Opus 4.6\n\n[1] https://opendev.org/openstack/neutron-specs/src/branch/master/specs/2026.2/bgp_evpn_type_5_route_support.rst#changes-to-cli-and-rest-api\n\nChange-Id: I8a9e118b4aa504ef570b81c38b9d46b6f41c812a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c2e5e9cc3ea854a0fe1c8145b4816c978dffbf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c2e5e9cc3ea854a0fe1c8145b4816c978dffbf4e"}]},"branch":"refs/heads/master"},"7ab69449c49a58c36245604efd7de913595f7db7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-30 17:33:09.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/85/991685/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/85/991685/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/3"}}},"commit":{"parents":[{"commit":"d842781c87ee2c4e775c64116efca10f81ff7a15","subject":"Merge \"Network: Add fields filtering option to Networking resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d842781c87ee2c4e775c64116efca10f81ff7a15"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-04 15:45:29.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-07-30 17:32:50.000000000","tz":-240},"subject":"network: Add evpn_vni resource attribute to the Router resource","message":"network: Add evpn_vni resource attribute to the Router resource\n\nThe patch adds the evpn_vni Int attribute to the Router resource\naccording the spec [1]. It will be set with the openstackclient\nwhen --evpn-vni is passed when creating the router.\n\nRelated-Bug: 2144617\nAssisted-By: Claude Opus 4.6\n\n[1] https://opendev.org/openstack/neutron-specs/src/branch/master/specs/2026.2/bgp_evpn_type_5_route_support.rst#changes-to-cli-and-rest-api\n\nChange-Id: I8a9e118b4aa504ef570b81c38b9d46b6f41c812a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7ab69449c49a58c36245604efd7de913595f7db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7ab69449c49a58c36245604efd7de913595f7db7"}]},"branch":"refs/heads/master"},"bf22bb239e25de406d28ed6f791d2d4de68aab2b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-30 19:24:03.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/85/991685/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/85/991685/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/85/991685/4"}}},"commit":{"parents":[{"commit":"d842781c87ee2c4e775c64116efca10f81ff7a15","subject":"Merge \"Network: Add fields filtering option to Networking resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d842781c87ee2c4e775c64116efca10f81ff7a15"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-06-04 15:45:29.000000000","tz":-240},"committer":{"name":"Jakub Libosvar","email":"jlibosva@redhat.com","date":"2026-07-30 19:24:03.000000000","tz":0},"subject":"network: Add evpn_vni resource attribute to the Router resource","message":"network: Add evpn_vni resource attribute to the Router resource\n\nThe patch adds the evpn_vni Int attribute to the Router resource\naccording the spec [1]. It will be set with the openstackclient\nwhen --evpn-vni is passed when creating the router.\n\nRelated-Bug: 2144617\nAssisted-By: Claude Opus 4.6\n\n[1] https://opendev.org/openstack/neutron-specs/src/branch/master/specs/2026.2/bgp_evpn_type_5_route_support.rst#changes-to-cli-and-rest-api\n\nDepends-On: https://review.opendev.org/c/openstack/python-openstackclient/+/999330\n\nChange-Id: I8a9e118b4aa504ef570b81c38b9d46b6f41c812a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bf22bb239e25de406d28ed6f791d2d4de68aab2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/bf22bb239e25de406d28ed6f791d2d4de68aab2b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
