)]}'
{"id":"openstack%2Fopenstacksdk~995724","triplet_id":"openstack%2Fopenstacksdk~master~I6a3f5cb851dcfd850982f1a9922c498b866fa02a","project":"openstack/openstacksdk","branch":"master","topic":"fields_for_network","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-08-19 11:45:51.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-10 15:14:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-08-19 11:45:51.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-08-19 11:45:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a3f5cb851dcfd850982f1a9922c498b866fa02a","subject":"Add extra param fields to get_xy networking methods","status":"MERGED","created":"2026-07-01 17:15:10.000000000","updated":"2026-08-19 11:49:28.000000000","submitted":"2026-08-19 11:45:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"995724-fields_for_network","meta_rev_id":"9aee3b95c1bac382d7dc2c5816c10bf2b53ae27b","_number":995724,"virtual_id_number":995724,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-19 11:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-08-10 15:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-08-11 21:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-08-19 10:11:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 18:42:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 12:58:34.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-07-30 09:52:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-08-05 10:32:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-11 21:20:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"5e42af953260721fad4315e2e825666f3fc396b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-01 17:15:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30ea5d451ce94190ad6debe2749a77201a5a61bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 18:42:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64146f1825f24ead9c24f18db7017589\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d4e2569c76e4851af0288453029ee54 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25475d5fbdc444a79c4d9b7758837bd6 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9099168a386b418589f307c2e0b1af45 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/307a7819dd0148779861270d22d54309 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3580c326dfb46b19188f3a94647eef5 : SUCCESS in 6m 31s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/41a0155d13c94b88877024c25d65f656 : SUCCESS in 7m 02s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b004b176b16d46f08f3ca02cb9a55136 : SUCCESS in 1h 10m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/19c4bd03005345d5a29830f996d7beb7 : SUCCESS in 3m 18s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cbb81fa914434c99bd8245d285ed735d : SUCCESS in 6m 13s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/4021d297a5b34039a5e25ee865f0646d : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6970cd2097064804a275a848e5f277b2 : SUCCESS in 14m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3829ef7787504d63bd57df6d49f5722b : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba056bef6d1d4270894599817f6ba7f5 : SUCCESS in 59m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/23e093f9850d40abaef7ea41249ae6ae : SUCCESS in 57m 04s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/16b23038b15944b698b3a7cef766020a : FAILURE in 1h 02m 18s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/512d92d3e5e844d8a74d3a75636602a7 : FAILURE in 46m 36s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f19bc2f087ab488bb603ad53a0d8358e : SUCCESS in 34m 56s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/f8fcfba9a5bc4ca186aca38c45c98234 : SUCCESS in 15m 34s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/4da1971f29f7460599c334c383b254ea : SUCCESS in 29m 55s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/4f1d9b200b2e4e4e8662505b4f5deec3 : SUCCESS in 38m 03s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3bd1b0a4db0249a5a5baca20d98fd85a : SUCCESS in 44m 40s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6e3689addf4a44c59a5598645948af18 : SUCCESS in 1h 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f0e4f247be13f6b0da79aa6e22b59374ec56ab1","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-07-21 12:58:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"825799779ccc11406c6519dfaf42739608199d5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:00:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b58aa37d0cb81c2b31bddc5526f33bd6a66c510","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:52:59.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstacksdk/+/995652 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"2dd316fa5906785c359792e03080d9807104abc0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-23 09:14:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b05d8bc7df110e9a0b30ef8ec8a44dc6802babb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 10:55:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/513a275426cb448c8d44432cf7be4562\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad89151ea16e43f19407d51c0a274dc9 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48388daca0484edc9673be5752cdcf87 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7a9e5a5e68746a58a66eb131392aeb4 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d64de6dabec4347a342e8aa443bd205 : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e63926c711034d8cad4ba810c09f9b22 : SUCCESS in 7m 05s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d4e867dfd20c441a87c2b708676b2550 : SUCCESS in 5m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a5f0d80fef244c999547f44283a48329 : SUCCESS in 58m 38s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/eca49d6fb43e40fcb76cbf7ea167e228 : SUCCESS in 3m 31s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7e2bf7c610c44f20be344bec2dd970ee : SUCCESS in 3m 52s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/53188f59f57d41779046394c6762054e : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808028fcbffb4ad2ad1ea92e35df06ce : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b6bc7c58d0c41fbbb800e73994eec5f : SUCCESS in 4m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bd3b1cc69cb40d5949e5edd33417c6d : SUCCESS in 1h 04m 59s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/202fdcaa6b7a4980959a3b6877d13da4 : SUCCESS in 1h 29m 12s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e33cdc09715d41eb8f434572f18580ab : FAILURE in 1h 26m 16s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e4cfde7506ee4e93ac7ff9199e2efad0 : FAILURE in 25m 38s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/fbf873fb87714c64b4fb3099cfc9a6ec : SUCCESS in 27m 55s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/49de2c18d2a7485fa3429e941501ca2e : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/21a5d43d722d4b74bf8058804e7468eb : SUCCESS in 28m 52s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1bd51f7223884353a8d8959c534f3e2f : FAILURE in 31m 05s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3d5f8f9e6e8742258f292d53317fe714 : SUCCESS in 1h 17m 26s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/7ef51467f01e4823a70089cd3f2466b3 : SUCCESS in 1h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfd519dc48c2eaead07e507930265e126cb5c993","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-07-27 13:56:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2df75c33a463adf2efae7cf670f711a88aa5f769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 15:41:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50774654b2d1458cbc0cfd373d8dbdbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2064a0185ac8481484e444c7315dda23 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01fd696da90d4bbf9b792f47bff10ddc : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/887b3d2d46f441c992e8848dcc10b867 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57c77b228b2b422ab1931e837310f5d7 : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54d668ab3f8c433aba545efe24646660 : SUCCESS in 7m 14s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/679a43a762744c86a49afb94549c4cd5 : SUCCESS in 7m 55s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f946d57e958440638f51817a87b30161 : SUCCESS in 1h 08m 06s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0078029526cf4b7cb548fa671e766930 : SUCCESS in 4m 19s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d97b125d6d91432384538b0e820d5c7a : SUCCESS in 4m 56s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f13bc0bfaf9d409986aecaf5a344e210 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad03f56ca63b40a6ab13296a4f1e2311 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96486464d16b4d759e61be7fc48e3245 : SUCCESS in 4m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5da7f985e0e24771b3391457d38f0644 : SUCCESS in 1h 04m 29s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/477d6b4355354842a54b85deb327bce7 : SUCCESS in 1h 18m 43s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/41351b4da8b3400484e3262adb24fc48 : FAILURE in 1h 31m 58s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c3f5517220754418a91db16d376645bf : FAILURE in 1h 12m 17s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/cfe160f4dc4643ab85dbe9ec74b90493 : SUCCESS in 37m 05s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/50dd8513b1834fc2a087d8186dada87e : SUCCESS in 31m 39s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ec58f915d3634aa1b5677ec909a22dc7 : SUCCESS in 36m 03s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1a157da9b0984053a24d1410491bb529 : SUCCESS in 43m 20s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3b45f6a5cf8948449c7290e39eb2e2cf : SUCCESS in 1h 11m 04s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/98dd8b69f1bc439fae0270d068dc2a27 : SUCCESS in 1h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"905a590319ac9902ea23d67db8fcaa0fce90ee00","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-07-30 09:52:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1af17a4725b88ba866ec17bbf02dcb0f46d4d854","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 10:32:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"21c9cc89df747be6a4e7655e5b52fc83bb18c6ab","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 10:33:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2462c13538e1bdbc3894fd436e5641295c4c00c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-07 16:00:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5b0010909cb27bcf31a64ed1bf34d797563f9b82","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-07 16:01:43.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fa3d070b6f28f0eff82e9bcd7ae7b05438157e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 17:44:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/313f3430f2fa499e8720392ad1d8dcf2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1636387acaa4d5bb7b3f6cee338f420 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/819400b3d5d54721930f2be3d0baa7d4 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d70eb6928f34a9191e83c2ce7119564 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb42350f684f4737a74a9e95b5f59896 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56ae84f40ff94b498ce257579df44034 : SUCCESS in 7m 55s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/eefd1cf18de14aa581b3bd8e334db93d : SUCCESS in 4m 44s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b422da3b08174c93aaf1fd5a1a73e710 : SUCCESS in 28m 53s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d47ef3823c17428cb8e8a2fc97f18d79 : SUCCESS in 4m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7215ff87e35a463689c66de68eed42cb : SUCCESS in 6m 45s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/89c29db85d2947c5ac91d6dfa823665f : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d636ab5122e141f292c7daf0e8d6d140 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5486de277e14418e8698cd643a60e655 : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/423e21f0bbb547f58975d09a9444ea9f : FAILURE in 1h 01m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/192a843edf3b4b92a9484abaa06f0aba : SUCCESS in 1h 25m 11s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/bf8538195c134c368a5ffebbf934bb51 : FAILURE in 1h 16m 12s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/33963777036e4f1bb8d9dcdba0c4ed1d : FAILURE in 29m 04s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/99cb3eb190494e74a7f886f9f260be45 : SUCCESS in 21m 19s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/6bd780a652fc4884ae0b954bef3c925f : SUCCESS in 29m 20s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/a042e612bd1a45ad8bc96d5ac0c1101a : SUCCESS in 31m 31s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/cd2ce371b74944f38a39820a4f99582d : FAILURE in 30m 31s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/226b0971c4b24f50be50af1316df55dd : SUCCESS in 45m 02s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/aefbbb51b06b4fa2b10c3a0121cc633d : SUCCESS in 1h 06m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"534abd2841b249e3b9a54b9964ddb158683eae30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-08 11:26:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6aef01e27ace210366f44700e58518042bbdd69c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-08 11:56:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b915867116ca577e11ff466a0db55b6f87c1a072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 12:51:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9995974af4d4fb9a12f0e90f6474179\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666fc828cc9845c2a4ddd511cac21eb7 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60f9e895db1545318ffbb741d890e78f : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ab666610261445aa26c148a2d06a0a6 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ffb30035f2a4742997c57b439f56b40 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3c8885f3d154e6a883bcb47af341f3b : SUCCESS in 5m 51s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9168eebcb5474da386dff5b653ed51e4 : SUCCESS in 7m 39s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f224743df912495d9930c47f1b719e14 : SUCCESS in 40m 22s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ab4ead8eb41e45b4b070e43e5469e1fd : SUCCESS in 3m 33s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1dda963579ff4176a63698375e530f94 : SUCCESS in 4m 28s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/117654c7686b46b191a534a277086eb7 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93710df91bd940c58ad4e25f875a93f6 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dff2222e131e43799f46852593c6349d : SUCCESS in 3m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fbbb639192445499a9fdf0645e9d010 : FAILURE in 45m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bb929b1e948c46399bf99801207398ae : SUCCESS in 1h 18m 28s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/546a7ed2efc14ed090913dd3428cb882 : FAILURE in 1h 10m 50s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e6dfd35675114731aae42ee43a289cfd : FAILURE in 39m 03s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/ba05868035394232a7b2531b83becbad : SUCCESS in 28m 10s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/b2b9abd5748b4435b9a0da43612bce4f : SUCCESS in 17m 58s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/4d15d64e24604cce8a50bff4aa47e358 : SUCCESS in 28m 48s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/200eccb3a50641288564a07f2b28abf8 : SUCCESS in 38m 50s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/55c54e5981d94d39af2f7b74451d25b8 : SUCCESS in 49m 46s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/dc79b59fddb34c4391a80fae71717e4a : SUCCESS in 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e70d6bae218b2d92e1387e8c65018b0a4dccaed8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-10 09:40:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67c9d756370d24358eeac36071edd6a18bb91be8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-10 10:40:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2b9be0d8ee54c32d1891ed135a9d8f71f9088f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 11:13:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ff6d84480ff4ca887855046585adee6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4a39b18c4140a09c44cef8962350db : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69cd11aa06784cd094dddfcb361e8d43 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07c1ffbf9e0f4aa79fa254b05b2c0e6f : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e8d316525c24b43b26f78747779d1f2 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2ba548253f244aa96f1acbf86dc7cd8 : SUCCESS in 6m 23s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ddb785735d964d7c8b90c43b1154cf22 : SUCCESS in 6m 41s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6aaadc91a1344f6baeb7c10e0531ec69 : FAILURE in 46m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8007d7f3492940ae8fd8de929204a17c : SUCCESS in 4m 58s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/896cb9f6552c42a5af92bb45d42e8154 : SUCCESS in 6m 34s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0d86f672c1084f7abe2bf01f08492983 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61af95dd48a47598bb8ccab069384de : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3a7d1e77b8e492583313514815e01bc : SUCCESS in 4m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4765ddafd03e4f46a5eb45815d337206 : FAILURE in 57m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0739967c0de5437dad0573a7b8bef94d : SUCCESS in 1h 16m 23s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/9edbc227343143f9b9233b2a3d03a802 : FAILURE in 1h 30m 59s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d1399546566548ad8833edd51902c3b0 : FAILURE in 54m 44s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/3c2955283d00409fb74491c1e09aa12a : SUCCESS in 33m 03s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/ff5f71be09a94908997efed5a1299d59 : SUCCESS in 18m 07s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/03d48ec05b99481f92d784390df2ac72 : SUCCESS in 28m 50s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/02c98497a1524deaba2b2710cc3f7733 : FAILURE in 44m 03s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2ae4ef858ce4484d86f244863a1415fb : SUCCESS in 1h 10m 05s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/1ea749705228472bb6d71b00e1a04ce0 : SUCCESS in 1h 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"22c78d68a58adc60816195b6117d15675265c165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-10 13:23:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"1a18101dee5d0eee9929828f2c541086e5e67eb6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-10 14:10:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bcc70399d4e399b1ab19e7e21cd53dd566ee6f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 14:43:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0114636f0b54879a4dbea9328970ffb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462731337182427dae6bf82671404272 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6916608c402471b83d21dda9328545a : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f38556bfc26e43db8573cd0d3cac03a8 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/018b2abe9e334dd9acffe1aa913ab6f6 : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3261e6b803454ad1816466f3fda9d162 : SUCCESS in 6m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bbd05ba91e834cd3bf2cb8b8e0c8d150 : SUCCESS in 6m 50s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a0101440cbce469889f8c5a959bb4b56 : FAILURE in 1h 03m 23s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/77e10f6c88304c349d4d9249b5edd4a8 : SUCCESS in 3m 32s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2a67470fda114c05869e5e2f1eba122b : SUCCESS in 4m 28s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1f6c2c6716374bf9a2042cc1e689655b : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae879e7ce734e40a9303d05ee44afb3 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e05c0a74da79491a990a0cd2dd46b623 : SUCCESS in 3m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/789aecb3aa62479f90c88f06719de8c1 : SUCCESS in 32m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/760773e7171c460b812c06c99c3dba6a : FAILURE in 56m 07s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/6ce2079f36ab42e0b3ed35a245d72ea9 : FAILURE in 1h 16m 25s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c240919272ab43e4997f7c2f5225afb9 : FAILURE in 55m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/9e0e65ec352d4d92b4c8b01f271c53f8 : SUCCESS in 21m 54s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/dcf955e887cf466784518b6f298ae581 : SUCCESS in 16m 54s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1a90df1cf64a4f4292d93c1590d8ad99 : SUCCESS in 17m 18s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/28e72ec798dc46949b246020f46e93eb : SUCCESS in 45m 21s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ef7cfb4269bf4e69820a958efa18fbfe : SUCCESS in 45m 44s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/965a668e7160451f9e2e2902f44cbf92 : SUCCESS in 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d48d808389418ff65c7d62e58b8d47cc7c2244b7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-10 15:14:13.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f5e4e69dc1a0e3c87735ca849e564a40f17f0900","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-11 09:49:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"897e9c2cfa3bbd5d7ef8cb7037070c3b4444e938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 11:23:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78a231ba611848f9a20fec1b054ff483\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc955586d504556b1216882630b96cf : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96a1f7dc3ec84d3794e605ac4a1de3fa : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c65f9b5666784b3e99fc297bc8937398 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e1e281cfb2340f496159fb619e1097c : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24fc60ea58f3442e9149684fcc90c0c4 : SUCCESS in 7m 04s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b1e75f1773cc406ba18120c937400760 : SUCCESS in 6m 59s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/51f0ef046bb04e0f91cbbc4276715757 : SUCCESS in 1h 04m 16s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ab02c3d4d10646f1aa8d9967a2a54046 : SUCCESS in 3m 21s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/becd90d78c81445fb8f57e17a714066a : SUCCESS in 4m 13s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/2f2565feff1f41dc95a5ea814d5fcab0 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b4e18bcec5477ea177a6b6b10a9753 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69fe0ebdb55c44268203db5beaf328a0 : SUCCESS in 3m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c6fd985544f440c9917f821a98ae704 : SUCCESS in 44m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/64a13817952f4ea09f856c965c061fda : SUCCESS in 1h 04m 16s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/8cdf19bf0bc647dcaf632bbe51fe9a8c : FAILURE in 1h 28m 45s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/8e10dbbe6e154861aaf574ea66afad96 : FAILURE in 35m 48s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/71c2b3dc26d54ea7ae748c6ef4692421 : SUCCESS in 23m 55s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/0280a0dd8da3414081c0ac2ac89d0b99 : SUCCESS in 29m 11s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/4980c108d8924cf3a7fd45dc20ff5728 : SUCCESS in 29m 08s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/907710b65aeb40899521538ebc575cf6 : SUCCESS in 41m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fbc890c574d247f5a8667d53ea4e7c42 : SUCCESS in 1h 00m 53s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4efd0b55223d4469b93175a0d258e692 : SUCCESS in 1h 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2415d6c6c271effed6ec041a19531b8c4b982baf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-08-11 21:20:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4d45009dd10ab53cd6da7e334bfb7ed8dccd6e3e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-19 10:11:37.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a17930bc35bdd798e8a1deff8d40c55f91ec696e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 10:13:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0393a3689d409dba144f3cc9aa53c872a7eee4a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 11:45:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd07f0428f244b6195fa8301823abee0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e135a61595458ba58ff83728dda0e1 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/189f4f0c761e4ceeb4314e549cfbe257 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa48c085de3e41f284f61de9babe5b97 : SUCCESS in 7m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e752b0eee50e4da78cd835b2abeb1b36 : SUCCESS in 6m 55s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/804685379a844ab2b9ab5d0b5faeab08 : SUCCESS in 1h 05m 50s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ecf2868a347e43cb8bcf4fd843506fd6 : SUCCESS in 4m 02s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/71c9b3199b9e47f59786c95762883dd6 : SUCCESS in 5m 06s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/88a00de031514428afcb8ced2a9731d2 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3701854cd494b6e96522c7678f60a56 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8018591d299f4104964b2000a83d6d79 : SUCCESS in 3m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ef4897cd3d04e9da9f2cbb4f0641c52 : SUCCESS in 49m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/60aa0d7ea2a743f6afae6e4cabc9b2d2 : SUCCESS in 1h 10m 56s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/8e3109bff0634ea1a74cd0225c214e48 : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":5},{"id":"e00824eaece099f3cd50188db3f5a5b0620b9810","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 11:45:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9aee3b95c1bac382d7dc2c5816c10bf2b53ae27b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 11:49:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf0c7350dcd41a18f944c895c627e1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a553150dc4a4425a9795c1738bbc3d6 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"841a36f4e5f12502aa50ffb9a41c11e07eb91259","revisions":{"7969d31619bf48f774b70cf5f879f1af4d5cd64b":{"kind":"REWORK","_number":1,"created":"2026-07-01 17:15:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/24/995724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/24/995724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/1"}}},"commit":{"parents":[{"commit":"8b5201a10c73fb87663ae0b8fb753c832c57ecd6","subject":"Network: Add fields filtering option to Networking resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8b5201a10c73fb87663ae0b8fb753c832c57ecd6"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-01 10:39:09.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-01 17:14:51.000000000","tz":120},"subject":"Add extra param to get_xy networking methods","message":"Add extra param to get_xy networking methods\n\nOpenstack APIs support fetching only selected fields.\nTo make it possible from SDK also this patch adds the\noptional **pparam to these methods to pass extra params\nlike fields\u003d[\u0027name\u0027, \u0027status\u0027] to retrieve only these fields\nfrom the API.\n\nChange-Id: I6a3f5cb851dcfd850982f1a9922c498b866fa02a\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7969d31619bf48f774b70cf5f879f1af4d5cd64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7969d31619bf48f774b70cf5f879f1af4d5cd64b"}]},"branch":"refs/heads/master"},"974f2787bd0c22c1e9dbfa5fae174517fc2a590e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-27 13:56:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/24/995724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/24/995724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/2"}}},"commit":{"parents":[{"commit":"d842781c87ee2c4e775c64116efca10f81ff7a15","subject":"Merge \"Network: Add fields filtering option to Networking resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d842781c87ee2c4e775c64116efca10f81ff7a15"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-01 10:39:09.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-07-27 13:56:22.000000000","tz":0},"subject":"Add extra param to get_xy networking methods","message":"Add extra param to get_xy networking methods\n\nOpenstack APIs support fetching only selected fields.\nTo make it possible from SDK also this patch adds the\noptional **pparam to these methods to pass extra params\nlike fields\u003d[\u0027name\u0027, \u0027status\u0027] to retrieve only these fields\nfrom the API.\n\nChange-Id: I6a3f5cb851dcfd850982f1a9922c498b866fa02a\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/974f2787bd0c22c1e9dbfa5fae174517fc2a590e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/974f2787bd0c22c1e9dbfa5fae174517fc2a590e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8b5201a10c73fb87663ae0b8fb753c832c57ecd6","ours":"7969d31619bf48f774b70cf5f879f1af4d5cd64b","theirs":"d842781c87ee2c4e775c64116efca10f81ff7a15","merge_strategy":"recursive","contains_conflicts":false}},"c9f4db63d7751a3721c3ef087c88117716ef506f":{"kind":"REWORK","_number":3,"created":"2026-08-07 16:00:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/24/995724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/24/995724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/3"}}},"commit":{"parents":[{"commit":"d842781c87ee2c4e775c64116efca10f81ff7a15","subject":"Merge \"Network: Add fields filtering option to Networking resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d842781c87ee2c4e775c64116efca10f81ff7a15"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-01 10:39:09.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-08-07 15:59:08.000000000","tz":120},"subject":"Add extra param fields to get_xy networking methods","message":"Add extra param fields to get_xy networking methods\n\nOpenstack APIs support fetching only selected fields.\nTo make it possible from SDK also this patch adds the\noptional fields to these methods to pass extra params\nlike fields\u003d[\u0027name\u0027, \u0027status\u0027] to retrieve only these fields\nfrom the API.\n\nChange-Id: I6a3f5cb851dcfd850982f1a9922c498b866fa02a\nAssisted-by: claude-sonnet-5\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c9f4db63d7751a3721c3ef087c88117716ef506f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c9f4db63d7751a3721c3ef087c88117716ef506f"}]},"branch":"refs/heads/master"},"fec5339c137f2de2ea51b5b386649b12f004fb2b":{"kind":"REWORK","_number":4,"created":"2026-08-08 11:26:00.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/24/995724/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/24/995724/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/4"}}},"commit":{"parents":[{"commit":"d842781c87ee2c4e775c64116efca10f81ff7a15","subject":"Merge \"Network: Add fields filtering option to Networking resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d842781c87ee2c4e775c64116efca10f81ff7a15"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-01 10:39:09.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-08-08 11:21:57.000000000","tz":120},"subject":"Add extra param fields to get_xy networking methods","message":"Add extra param fields to get_xy networking methods\n\nOpenstack APIs support fetching only selected fields.\nTo make it possible from SDK also this patch adds the\noptional fields to these methods to pass extra params\nlike fields\u003d[\u0027name\u0027, \u0027status\u0027] to retrieve only these fields\nfrom the API.\n\nChange-Id: I6a3f5cb851dcfd850982f1a9922c498b866fa02a\nAssisted-by: claude-sonnet-5\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fec5339c137f2de2ea51b5b386649b12f004fb2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fec5339c137f2de2ea51b5b386649b12f004fb2b"}]},"branch":"refs/heads/master"},"841a36f4e5f12502aa50ffb9a41c11e07eb91259":{"kind":"REWORK","_number":5,"created":"2026-08-10 13:23:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/24/995724/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/24/995724/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/24/995724/5"}}},"commit":{"parents":[{"commit":"d842781c87ee2c4e775c64116efca10f81ff7a15","subject":"Merge \"Network: Add fields filtering option to Networking resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d842781c87ee2c4e775c64116efca10f81ff7a15"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-07-01 10:39:09.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-08-10 13:13:23.000000000","tz":120},"subject":"Add extra param fields to get_xy networking methods","message":"Add extra param fields to get_xy networking methods\n\nOpenstack APIs support fetching only selected fields.\nTo make it possible from SDK also this patch adds the\noptional fields to these methods to pass extra params\nlike fields\u003d[\u0027name\u0027, \u0027status\u0027] to retrieve only these fields\nfrom the API.\n\nChange-Id: I6a3f5cb851dcfd850982f1a9922c498b866fa02a\nAssisted-by: claude-sonnet-5\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/841a36f4e5f12502aa50ffb9a41c11e07eb91259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/841a36f4e5f12502aa50ffb9a41c11e07eb91259"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
