)]}'
{"id":"openstack%2Fopenstacksdk~997374","triplet_id":"openstack%2Fopenstacksdk~master~I22617ab2ae9eb2285522b6d25bfdddf7de51bff9","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"7654":{"account":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"last_update":"2026-08-05 14:43:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I22617ab2ae9eb2285522b6d25bfdddf7de51bff9","subject":"New cloud vendor config for new Switch Cloud service.","status":"MERGED","created":"2026-07-15 11:35:45.000000000","updated":"2026-08-05 14:45:29.000000000","submitted":"2026-08-05 14:43:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997374","meta_rev_id":"561379ef5f6ba144732f79989788c0319446a16e","_number":997374,"virtual_id_number":997374,"owner":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-05 14:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-05 10:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-05 10:17:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-15 13:09:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 10:17:25.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f96528b3a99d98dcd93f42c704c65b508bc9a7b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"date":"2026-07-15 11:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"268e24cfa4d721b03082a806613633dceddaa5e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 13:09:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a794443cec4823bfad6d2ba2d96961\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fed3479ebe7470abf5ea9f7cd108445 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac0798ec6d4649dabf5fabb1ade361f3 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba8691fddfa54c019cd54a94d2a28451 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e3d65b328254e77a413d4873f8cdbea : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7889ed35a24a4fc7bf0dc7b571b010c1 : SUCCESS in 4m 03s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/5840267eaed4411db018ab4eb0d5170e : SUCCESS in 4m 46s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/02392d0d75084d058b03a09d27c02d0a : SUCCESS in 55m 44s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/31473da5900e4ceebd746db9648d0809 : SUCCESS in 3m 55s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d644b79cd7d347b98fe158a209a29aa3 : SUCCESS in 4m 03s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/75e3f24026ff402d8250b1d251d5cab6 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/948068da4a00409cb7a140d293d32f90 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00ccfb4d215642fbbdfe072bffbc47b8 : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32938fb8fc8444d1aff55abec7b4bee5 : SUCCESS in 1h 04m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5d4e4c4a2810411085f4cd74920a44a1 : SUCCESS in 1h 22m 57s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/77e861ead1cd438b8540fa75a53d903f : FAILURE in 1h 28m 02s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/83c9dc78ad324e01b7fdfd2fb7a8d1f0 : FAILURE in 56m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/efd396db0577455e9eab290d5e8cfd0f : SUCCESS in 33m 51s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/011ad88ffbfd4f8c8e0f80b7b868f32d : SUCCESS in 15m 36s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0991005172af475093c16931ec7b40c7 : SUCCESS in 20m 26s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d5dbde83fecd4c829d41d2794350b912 : SUCCESS in 44m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8f795d011cba4f85ad08e7e0added57f : SUCCESS in 1h 09m 57s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/cb79d0aa6f524232b195d6ab8d263c38 : SUCCESS in 1h 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"722e495ff265b7a37cfae35053d988f20cac3a8f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 10:17:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22230ae4fded4d1ae41454fa225ed5fac6f5495e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 10:18:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e9ca856a2d4d20e4f40d28baf3bec0ddc1f4ca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:43:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53775c3e14814b8eb85f736167c52424\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75cc35ba54044c97a1b0456075f4025c : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93bbc34ea4424e329354807b4ad315a4 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c261563ab12460f8ea87e12891aacbf : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16120b353666456c867e17adadc8ab0f : SUCCESS in 7m 18s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/13784bf8b0464ca997953e686f912694 : SUCCESS in 59m 24s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b0ebd16bd9cd4d6fae34ef6eeb73cfdf : SUCCESS in 3m 53s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ef1ef0541c354aaa867efc69979362c0 : SUCCESS in 6m 24s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/3b0ff6b5c3714620951e11e5e479c620 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c615b739c842429db1d0b5e793d1bf34 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/097aff8cdf5d45618a4e76c0f6612b5e : SUCCESS in 5m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0833326900864fb598214c9e9a8a945d : SUCCESS in 48m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c7bd5c59dbef4147a7a56fd7e5c751b8 : SUCCESS in 1h 21m 47s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/9653fd35e2bd4710aa63218cd2c3ce05 : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"afec9b5219fc857481cf32e51f70af9ab2475c91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:43:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"561379ef5f6ba144732f79989788c0319446a16e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:45:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77cffeef3460470a827d8a41eed1ae5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac47a5a6fc924e6eb908f2af33aa7706 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4445217b3862c3906e685b217753ecd5040295b5","revisions":{"4445217b3862c3906e685b217753ecd5040295b5":{"kind":"REWORK","_number":1,"created":"2026-07-15 11:35:45.000000000","uploader":{"_account_id":7654,"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","username":"valery.tschopp","status":"SWITCH"},"ref":"refs/changes/74/997374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/74/997374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/74/997374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/74/997374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/74/997374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/74/997374/1"}}},"commit":{"parents":[{"commit":"8d9426395c172b35ef4ed1a8cbaa525c17944c6b","subject":"Merge \"share: Add missing query-string parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8d9426395c172b35ef4ed1a8cbaa525c17944c6b"}]}],"author":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-07-15 11:24:04.000000000","tz":120},"committer":{"name":"Valery Tschopp","email":"valery.tschopp@switch.ch","date":"2026-07-15 11:34:57.000000000","tz":120},"subject":"New cloud vendor config for new Switch Cloud service.","message":"New cloud vendor config for new Switch Cloud service.\n\nDocImpact\nImplements: New Switch Cloud configuration\nSigned-off-by: Valery Tschopp \u003cvalery.tschopp@switch.ch\u003e\nStory: 2011850\nTask: 55199\nChange-Id: I22617ab2ae9eb2285522b6d25bfdddf7de51bff9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4445217b3862c3906e685b217753ecd5040295b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4445217b3862c3906e685b217753ecd5040295b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
