)]}'
{"id":"openstack%2Fopenstacksdk~999040","triplet_id":"openstack%2Fopenstacksdk~master~Ifce0542755d092a5065d204880e01fbe3b5d66cf","project":"openstack/openstacksdk","branch":"master","topic":"identity-gaps/endpoint-group-endpoint-project-associations","attention_set":{},"removed_from_attention_set":{"36482":{"account":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"last_update":"2026-08-20 15:39:14.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-20 09:37:22.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ifce0542755d092a5065d204880e01fbe3b5d66cf","subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","status":"MERGED","created":"2026-07-29 01:31:57.000000000","updated":"2026-08-20 15:41:45.000000000","submitted":"2026-08-20 15:39:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999040-identity-gaps/endpoint-group-endpoint-project-associations","meta_rev_id":"4e85e2ff092ab87f1f421de930658e48c48cfa10","_number":999040,"virtual_id_number":999040,"owner":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-20 15:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-20 09:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-20 09:37:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-29 03:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-04 11:24:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d3a1c911fd4befd9f1febc02d539522afb29a7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-07-29 01:31:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8b1ad18fe8968d9f68ebbf5000f8a96f82a0287","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-07-29 01:32:48.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9dc417ab4d93e65706f3abb61536cc860317931c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 03:20:11.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b24935f15f544deea445e4260e32b39c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f5846f7af941d1a8f833de06a81d7e : FAILURE in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1978ed0a6e74a1ab66e2dac5fb9861a : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08de4019a49b4da6bc81992715ce7910 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b1bff7638e04e539898c095b9f8dd98 : SUCCESS in 6m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38f633be485a42daab96712630fd8f54 : SUCCESS in 8m 51s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/901834f806a741789cf91e241372f304 : SUCCESS in 6m 43s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/53ba5dda83a84523bd55c123ba20f459 : FAILURE in 34m 05s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1091566d25814ebfb4710ca53a25b44f : SUCCESS in 3m 09s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ff34d2ee4785422b98262a20163037da : SUCCESS in 4m 03s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/e84a938e5dd14dd4b3f4abee5044f7af : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b090cb2d60054d2fbca221e95b2e64de : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e04b15097f4244bcb90b1d71da3a8953 : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6470069c5b97437ca501bf7a65db620a : FAILURE in 59m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/73b5495ff55649d58b7ddb7c99389e91 : FAILURE in 1h 20m 36s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/3e53d735e7a047dbbea2a24fbe704585 : FAILURE in 1h 42m 05s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d9f1367f3a0c491fa69b152f6f7f4499 : FAILURE in 30m 24s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7a4eb8cb638c4d1db6a83b7c1e8e565d : SUCCESS in 33m 55s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/45a734836e5e4e50ba79c3164b80d222 : SUCCESS in 29m 00s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/8e1fa1fcccc445f1b9f3a6be97d0d15b : SUCCESS in 31m 21s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/27e2c4a0f1fa4267ac8cfbd0ee0fe27d : SUCCESS in 38m 17s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/30a76891321d42dcb6918714e86d75aa : SUCCESS in 1h 10m 27s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/9acc426e0d50408db12d8d9041dd1cfb : SUCCESS in 1h 04m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b98bca1c8d003a38de1c1eed30cf127d78fd5f81","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-08-04 01:18:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"37ed484a6d8869788c9a75897dafb7351b4901be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 02:58:13.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46a1b50ea0484bb8bbe5251b93bd39fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389b12870ca046979b1cba3ed38b4323 : FAILURE in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fde9b3e7b39544c085179c65510d32b0 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0212387c837145ad92e61526fb02e3f6 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdc1652a9fbf42139fa6a84a82532012 : SUCCESS in 2m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/425fa7c722074ac59912621297fafde7 : SUCCESS in 6m 19s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9f58e7b7d35f4acd90cf788a9e91a8b9 : SUCCESS in 8m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f18025052870468f886e7e13691671d4 : FAILURE in 51m 13s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/08ba7575a8984cc288f209dc992244b7 : SUCCESS in 3m 03s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a38c7c77963b4307867b84034a029e17 : SUCCESS in 4m 12s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/16315069f68e4c6ebcff4d222fb36212 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b92d752e054705864391e6dca87ec3 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9f0a7ff107741e589581ee7e5a77918 : SUCCESS in 4m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e13c3a9f78a485682456ffb4a450195 : FAILURE in 1h 05m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/72fab112d29646f9af6821a176ff31cc : FAILURE in 1h 11m 52s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/43d985ebadbb4144accef7dce5c689e1 : FAILURE in 1h 33m 49s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/eafa569cc78b466b8599823c92551e28 : FAILURE in 1h 06m 08s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/95d2dc4569524d57b70fcbd5ecdd379d : SUCCESS in 34m 45s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/6a10d62a4c354d3eb62a58d0ce348421 : SUCCESS in 15m 24s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/3b4c7e15384c43ad8b00fa9c9df9de0c : SUCCESS in 32m 27s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1a401a022b594aecaa1c0a02812f29b2 : SUCCESS in 40m 08s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6978889533e7433aa825afc0d9c4a115 : SUCCESS in 1h 17m 14s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/47a191b9932a4dcbbc2fa8910224c7fc : SUCCESS in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19f5467adc23403611340dae4e43e98221085747","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-08-04 20:20:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"959edec76490304e7f25014d6f196128313de6e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-08-04 20:34:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"687196342a20d50be2430bf791f17df179dc4e55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 22:11:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17837c3c90f944acbbbbca3f88aab530\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7d762e3372495c99ba27b967ce6c7e : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b437ebe0e814e6bbd7456e360b9e3f9 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11db719c345646ee9143b53e0f3a6b5e : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66e28c453b1e44b4872cc24d6d705424 : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f836d17a56224695bd34b7b735d406ba : SUCCESS in 9m 24s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4790917fda1c4d93867fc44b3ef22a17 : SUCCESS in 9m 57s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d421252c3b69498db946babee86a1a33 : FAILURE in 55m 41s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1ce05d8eb87d4ad8a73e1959f4e13e8a : SUCCESS in 3m 29s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cfc9e96b3d03421f8c5b5242c25c4858 : SUCCESS in 7m 22s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f8d4fcc98a034f6d8b61dc4fb6521a13 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd7600e16244462a198913aa02a18a6 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d93b9cdbdf3453382aea3b718526b3e : SUCCESS in 4m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89ba8ed2f7994c428051b4a60eadde1e : FAILURE in 59m 29s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ca12d840bcbb41cc9558206add4f480d : FAILURE in 1h 22m 45s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d50a93e11726432f8df5b9addae90159 : FAILURE in 1h 28m 53s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/77af290f36a142f08206ad12a26c1c8b : FAILURE in 28m 35s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/187c8ab2725148118ca5860069f02803 : SUCCESS in 35m 08s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/58417138a7cc4440a19e2b5836bdd387 : SUCCESS in 22m 28s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/5e6c5fb098f540f9b69e42917c3d7eee : SUCCESS in 28m 54s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/d57b94412d914fe58ddf37a2db8ce4e5 : SUCCESS in 20m 01s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7cdec02a693e43e4ac2fdb16a395c604 : FAILURE in 1h 13m 57s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/d5cae32e14ca4310ad2a2dc19aeef28d : SUCCESS in 1h 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"78b30285e6fc85fbb4e70ccfef01fcc628f3b98e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-08-04 22:27:56.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"4ed44a12cf384f54805a33ef680fa4944b06271b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 00:05:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a8ab677f85e4d56b08f29b485679179\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b942a4b5a7f74f32a19e73df922d1ac4 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e9e4a0bea6246eca8eb999097866b1b : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63708057390647a79cfeee081ebec7b6 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7673e54f90124afbbf0df5554aa32fc0 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f45f30064f746f0ac6b088e39a58080 : SUCCESS in 7m 45s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9c0278ed8b184f7e8ce2b3dfd3ff4e37 : SUCCESS in 7m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6a8a4a399018467b94a10571397d0e64 : SUCCESS in 58m 09s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bb81d720997a4f099d8d04850ea7728b : SUCCESS in 3m 34s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c1bfdebddd9641fc9b05b3792e6fcc0d : SUCCESS in 4m 22s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/e050bc8d806145b2ae7f5396b26c0a6f : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79804c8780ca48fc9eb3b7dceb0d23f2 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78bda40e407c4bbba68b2924da8734c2 : SUCCESS in 4m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00026b66e024468fa6bcc3b5e618116a : SUCCESS in 27m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ed737cc66fc849a0a02846aa3fd8108a : SUCCESS in 1h 29m 20s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a6077ea60a6f4e53a4797941389d2a7d : FAILURE in 1h 36m 06s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/4a8856407c284d9db04ae4c985c4aae4 : FAILURE in 28m 56s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/bc0c2e6937e247c2bb5f5d870595fb02 : SUCCESS in 40m 33s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/41400029e8964140a63f12c43bc879c1 : SUCCESS in 16m 36s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6c49366ca399447db00836ddeef74fcf : SUCCESS in 23m 12s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0c7e10d6099b42f3a92470af76654512 : SUCCESS in 40m 01s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c3d04bedd79448a8baa6047eab64335c : SUCCESS in 41m 07s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/d177ac63c5f54542bf9fc6124f746899 : SUCCESS in 1h 13m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"28d3a9f851f99532f61a6e9b99e4b32750bcf8ce","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-08-05 00:56:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"187785385fdfca63a71ccf014177c8f9c8a50556","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 10:04:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2b420346e066b3fef07576f8c10a600e0a47eae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 10:05:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"67394592887971e4b6713a540bde7bffac334c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 11:34:22.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstacksdk/+/995762 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"561c08430f0d98441b34de67a8e5cf5b42c55b47","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-11 09:08:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4af3af1bf0b71e3a05ec71d986042b737707c5f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 10:41:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc7b565398eb4229bccfe7dadd769e6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbcbf1c57d974e159c5c10ea757aa525 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2243c499fcd34e4fb63d56a5d3033d8e : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bce269039e0a4c848ff5659a624ed73c : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa680eb5f7264831821dbf0ba5f6bed5 : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbb912b6db1144668fc6a5f16d8acf72 : SUCCESS in 6m 30s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/884c23a859974b149d1a17e8ef9b328a : SUCCESS in 7m 23s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2600d479785c4633b3db9a5d1b8da1b6 : SUCCESS in 1h 02m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/85af2f61bb2a420abd5ec4551159dcf1 : SUCCESS in 2m 43s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/294e4877eea54070adb0e71530150268 : SUCCESS in 4m 12s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/a70322d1b643410d9db3c9dc8c7483ae : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3213edd7a7eb4711b168e3ed506bd475 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2808e9ae6bbc4727ae49f76cb5dff641 : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/502209068bc24aa1871b223fef572e0a : SUCCESS in 47m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/534d36c819df4b2296c850d0d3d7758f : SUCCESS in 1h 26m 40s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/423262c8fb2747f5a2b8190efe43fe00 : FAILURE in 55m 44s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e6ed43b95c404403a1091ae9d4dc24d5 : FAILURE in 55m 09s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/5f0f6afc741e49a6aa1f5930cebc5fa2 : SUCCESS in 34m 04s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/b86ef95e03c34bef8bc383ec2a06425f : SUCCESS in 15m 27s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/c57dab16ff5149cdae41ce0b59888319 : SUCCESS in 30m 55s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/eba1647c48c0431f9ae86f7e826656af : SUCCESS in 28m 40s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/76cf789ca5504a2c8afb60371ffafacd : SUCCESS in 55m 51s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/420ecd94e6a44944814615586efbecf4 : SUCCESS in 1h 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"391a96fb9add458827a29e9f67751d392669ad6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-08-18 21:17:36.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"23513c2e2a662a3981b1f5ed333716440f4089c4","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2026-08-18 21:19:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"72cc8fd22c7acda08773321c955717477b01ef7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 23:13:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb948c4bdd32477cb7181c6398cf0b01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8324c11d53624780bd8f44fb0851b4ea : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7677d243195e4f68bb87a2669e7f1ffe : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad0873b7780d4e7482f1e2b72c4389a4 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/702efd43153849d5828d60f1493d8c46 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da0d8c21420e4c919eb17bcd30f4080c : SUCCESS in 6m 19s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0429acb248a141568512bb9c041ff968 : SUCCESS in 7m 02s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/57b53c8dc55b42fb98d0e5757b1ffd20 : SUCCESS in 1h 02m 15s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/33c2f8f3b9a6453da91fa66621e5a424 : SUCCESS in 4m 34s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8d1913ea8c1f453ca7775c3bfe2cf161 : SUCCESS in 4m 28s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/c8c13bbbdd4e4ec68fd83b36348a010c : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef6d80c0f25d456eb7200e5d49fca267 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc320288cf8e423ba64bf303dcb36e0c : SUCCESS in 5m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc2d6aa82cb04f599d347e4c4b076a72 : SUCCESS in 32m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/58fcc96c590f40718744edc2f84318e8 : SUCCESS in 1h 24m 32s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/9892e944b36f4a7b90e761bf2b4746e0 : FAILURE in 1h 48m 20s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9476449978424f3db8241e637b1a4d5d : FAILURE in 54m 22s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/46769d46da8047109cda545c745594ed : SUCCESS in 33m 53s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/4d982db0776d4c76aaf74f614a9c1d9f : SUCCESS in 27m 48s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/ed5754ca097a4c48a7c399f71db28006 : SUCCESS in 27m 58s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9f771fe176094d0ca58596e159d7862a : SUCCESS in 29m 48s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/4a5cd2fd218d43bea51d9e2c4ccc3127 : SUCCESS in 44m 47s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ff591e46841b4f08b1c6cf35be871d30 : SUCCESS in 1h 19m 36s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2dfd0680d44c43109d12e45f922924ef : SUCCESS in 1h 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8db7f77d5533f2caea10c5c102e4d54db73e75dd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-19 10:04:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7f388b624d6d5302db5e6fe3cb72ceb83afdcfc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 10:05:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4cac58ca0b588cfbbebf3d7b9e36c66df61e2495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 11:28:57.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstacksdk/+/995762 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"e1b0583a4b5905415941918ff965bc7dcdd2414b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-19 15:33:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"29d8c4cf096dff4ebdfd92f92ee81d49fa00817d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 17:08:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c699a0644584ea4acba57560500529b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee3cd63f7e14f4e9ab5b96d09cc8153 : FAILURE in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75dd4af4b593437dbfa81b35032c280d : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4086a6eab1074f6f9a700a6f34967ecd : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92b7cdab8ecd49e8b2ca13e25ac9dea5 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f183298db1c1415ba9f7c6de5f60ea04 : SUCCESS in 6m 30s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0b8e9a4e4f9d4618b836352af77c0042 : SUCCESS in 7m 17s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d399129e893a4977afcbe85fcda00c68 : SUCCESS in 59m 01s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f178394068604e6d934864e7268ce2b2 : SUCCESS in 2m 08s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/acecdc3551b646a39d2bcae4d902763c : SUCCESS in 4m 38s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/8226ec92f9f3431b991dd92b392ffbce : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f64d775c464fe1a98ed7098153c976 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55c8faad0be64e76b9b35ae892ece0fe : SUCCESS in 4m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8a2728c334d48ceb996a708b149a604 : SUCCESS in 58m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4446772f658a44c39e79c4ab101fe631 : SUCCESS in 1h 28m 43s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/48b09b912b30452e91cce0de92ef8047 : FAILURE in 1h 04m 19s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/614738685967429c90095dfa4d70c63a : FAILURE in 55m 41s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/4bf9726815864891bddabb69f8d6fc8d : SUCCESS in 35m 38s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/fa8aea5a85f74949a663ee09eb14539b : SUCCESS in 29m 13s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/8fec7a17e02b44428460b4f96471f3f0 : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/94b09a9ab5344b668f6fbe88716518c1 : SUCCESS in 29m 28s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/50d0916a0c5441729a37e2271ca24663 : SUCCESS in 26m 13s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/60e409ecee5f42dda217e331d4b6094a : SUCCESS in 55m 41s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/35c2dff0feb24931b49751884901b786 : SUCCESS in 1h 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"20461321cbf805d645a412e195d70c6a02d8f67a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-20 09:37:19.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_36482\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"}],"_revision_number":7},{"id":"1a47cae8469af8a3fde0b955bb66c313e7923a60","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-20 09:37:22.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1538213c1f5f0ef69db6338d401b23aa414fcd71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 11:16:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de6dad459f749c9b1e35a7246be1175\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d010d8d566ae49399ce78c198fc991af : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c53557ea9d80497384c2608363ed2019 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf4cb33799894951bdb4c064094f153d : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cb4a8cf178342a1aa6f55833e1a161a : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d41901175a84ac3825e5d594dd907f1 : SUCCESS in 6m 32s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3e0fcbfa12494dec98b6b28e320ac23c : SUCCESS in 3m 57s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6e129fe56407445ca13ee8cbb2b2193e : SUCCESS in 1h 04m 40s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/28bd2f48fcff477a8c0cb509d72c4b5d : SUCCESS in 3m 21s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b63896dd2785434f8e46b82c6bf373df : SUCCESS in 3m 01s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/a9be2d9181454a4d9502c2fba90f9abc : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6e37c2db084c78a6ac02623318388b : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f7f49852f694ef8a0a7ccd11853301e : SUCCESS in 4m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f6866471b5d4bfd98e6276b8ccae6bf : SUCCESS in 59m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d43bfe5264e5473d81fb8d2e9c45cf21 : SUCCESS in 1h 11m 44s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/23b2fe81486644b1b28e18a8557ecaa6 : FAILURE in 1h 24m 55s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/4efef545a7e84eea86a3c4b1aceb7c68 : FAILURE in 40m 53s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7b69be996b2b42f4abd360edf91bf7f7 : SUCCESS in 33m 15s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/913dee5052fe4b32bc7cc8a6066e8ea9 : SUCCESS in 28m 32s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/783d705baf594e57a0fc0ae17875ecdf : SUCCESS in 21m 30s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/f01b6a79646d449b81a1120e931c54e9 : SUCCESS in 31m 45s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f4839532b0ae432eb26c4bc19c79ec0b : SUCCESS in 19m 50s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/288960a740d5460d9879f0c1b34d5713 : SUCCESS in 1h 12m 56s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/f972e781a84549ba9c50f55d0ac65d61 : SUCCESS in 1h 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6f985f864244cdf662b88527a133569f6de1fe3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 13:48:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fb592d58c661619400526af782f1fb0e357981ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:39:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221dccdc856b4589bc5bcb19ce899315\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac5387865ae49d1a72b55c813ba7c40 : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f7dc32ff45e4862aada2faef58fbef7 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87ba890a2c214e3f835f58ad04b1d82f : SUCCESS in 6m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/76babd00984843978c7e70b166dfdcfa : SUCCESS in 7m 41s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/efea6574506344a2bad8a4c67f968900 : SUCCESS in 1h 02m 38s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fada1dae43644019a134fb2909168987 : SUCCESS in 4m 23s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/747efc48e58c4440b0dbcc678e5170f3 : SUCCESS in 5m 02s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/81a7f1b1c8f5402ca0e6f507d26fbb5c : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/870a8d79c074409a97d5ccebc4ef56cb : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f97b3599e2294c3f933d625f51d11c0a : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/776e190a3d54449c8e4e41e54f4f3859 : SUCCESS in 58m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5095bde919b8478c9600cca3895a587c : SUCCESS in 1h 43m 17s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/55a32d9cfff44c2aaa067fd02a33c8a0 : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":7},{"id":"e19973fc9a4cb295f8cff8999ac1561b5b5d8bcd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:39:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"4e85e2ff092ab87f1f421de930658e48c48cfa10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:41:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e0d1fe14794fe893ec19f92d20a4c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c6630be8154003b76e5427e4f32b67 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"180d635dd7c3f54426c4ee122c3c366782199fb3","revisions":{"73a740b4e8493f37e31b5654ce1def654b65c8f3":{"kind":"REWORK","_number":1,"created":"2026-07-29 01:31:57.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/40/999040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/40/999040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/1"}}},"commit":{"parents":[{"commit":"58e78e9127a91ce938ebd93b139790b2d8706b1f","subject":"Identity: Add support for endpoint groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58e78e9127a91ce938ebd93b139790b2d8706b1f"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:30:14.000000000","tz":-420},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:31:44.000000000","tz":-420},"subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","message":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations\n\nChange-Id: Ifce0542755d092a5065d204880e01fbe3b5d66cf\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/73a740b4e8493f37e31b5654ce1def654b65c8f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/73a740b4e8493f37e31b5654ce1def654b65c8f3"}]},"branch":"refs/heads/master"},"cca54aaea1f150f6b53bd2439da6f8a737692348":{"kind":"REWORK","_number":2,"created":"2026-08-04 01:18:46.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/40/999040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/40/999040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/2"}}},"commit":{"parents":[{"commit":"58e78e9127a91ce938ebd93b139790b2d8706b1f","subject":"Identity: Add support for endpoint groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58e78e9127a91ce938ebd93b139790b2d8706b1f"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:30:14.000000000","tz":-420},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-08-04 01:18:24.000000000","tz":-420},"subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","message":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations\n\nChange-Id: Ifce0542755d092a5065d204880e01fbe3b5d66cf\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/cca54aaea1f150f6b53bd2439da6f8a737692348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/cca54aaea1f150f6b53bd2439da6f8a737692348"}]},"branch":"refs/heads/master"},"3487c963954e51bc542db1ff7f7c80836df3b759":{"kind":"REWORK","_number":3,"created":"2026-08-04 20:20:29.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/40/999040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/40/999040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/3"}}},"commit":{"parents":[{"commit":"58e78e9127a91ce938ebd93b139790b2d8706b1f","subject":"Identity: Add support for endpoint groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58e78e9127a91ce938ebd93b139790b2d8706b1f"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:30:14.000000000","tz":-420},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-08-04 20:19:29.000000000","tz":-420},"subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","message":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations\n\nChange-Id: Ifce0542755d092a5065d204880e01fbe3b5d66cf\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3487c963954e51bc542db1ff7f7c80836df3b759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3487c963954e51bc542db1ff7f7c80836df3b759"}]},"branch":"refs/heads/master"},"64db9c79362f1b39a96150ad16f54f2948186d1c":{"kind":"REWORK","_number":4,"created":"2026-08-04 20:34:43.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/40/999040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/40/999040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/4"}}},"commit":{"parents":[{"commit":"58e78e9127a91ce938ebd93b139790b2d8706b1f","subject":"Identity: Add support for endpoint groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58e78e9127a91ce938ebd93b139790b2d8706b1f"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:30:14.000000000","tz":-420},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-08-04 20:34:12.000000000","tz":-420},"subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","message":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations\n\nChange-Id: Ifce0542755d092a5065d204880e01fbe3b5d66cf\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/64db9c79362f1b39a96150ad16f54f2948186d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/64db9c79362f1b39a96150ad16f54f2948186d1c"}]},"branch":"refs/heads/master"},"09ef668ec7a30509fb05e03a1cee7dadf3fed05c":{"kind":"REWORK","_number":5,"created":"2026-08-04 22:27:56.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/40/999040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/40/999040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/5"}}},"commit":{"parents":[{"commit":"58e78e9127a91ce938ebd93b139790b2d8706b1f","subject":"Identity: Add support for endpoint groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/58e78e9127a91ce938ebd93b139790b2d8706b1f"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:30:14.000000000","tz":-420},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-08-04 22:26:56.000000000","tz":-420},"subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","message":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations\n\nChange-Id: Ifce0542755d092a5065d204880e01fbe3b5d66cf\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/09ef668ec7a30509fb05e03a1cee7dadf3fed05c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/09ef668ec7a30509fb05e03a1cee7dadf3fed05c"}]},"branch":"refs/heads/master"},"3fbb09ccfa92be9194dce9e786a74eecf78358d8":{"kind":"REWORK","_number":6,"created":"2026-08-18 21:17:36.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/40/999040/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/40/999040/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/6"}}},"commit":{"parents":[{"commit":"394b9c45b4843035df73584f9fe70e956e7f9327","subject":"Identity: Add support for endpoint groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/394b9c45b4843035df73584f9fe70e956e7f9327"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:30:14.000000000","tz":-420},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-08-18 21:16:56.000000000","tz":-420},"subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","message":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations\n\nChange-Id: Ifce0542755d092a5065d204880e01fbe3b5d66cf\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3fbb09ccfa92be9194dce9e786a74eecf78358d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3fbb09ccfa92be9194dce9e786a74eecf78358d8"}]},"branch":"refs/heads/master"},"180d635dd7c3f54426c4ee122c3c366782199fb3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-08-20 09:37:19.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/999040/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/40/999040/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/40/999040/7"}}},"commit":{"parents":[{"commit":"29a49e3b9064f77606024ccb068d0ee521f00baa","subject":"Identity: Add support for endpoint groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/29a49e3b9064f77606024ccb068d0ee521f00baa"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2026-07-29 01:30:14.000000000","tz":-420},"committer":{"name":"Oria Weng","email":"oweng@osuosl.org","date":"2026-08-20 09:37:19.000000000","tz":0},"subject":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations","message":"Identity: Add support for endpoint group/endpoint and endpoint group/project associations\n\nChange-Id: Ifce0542755d092a5065d204880e01fbe3b5d66cf\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/180d635dd7c3f54426c4ee122c3c366782199fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/180d635dd7c3f54426c4ee122c3c366782199fb3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"394b9c45b4843035df73584f9fe70e956e7f9327","ours":"3fbb09ccfa92be9194dce9e786a74eecf78358d8","theirs":"29a49e3b9064f77606024ccb068d0ee521f00baa","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
