)]}'
{"id":"openstack%2Foperations-guide~267354","triplet_id":"openstack%2Foperations-guide~master~I35276e0635893c3b5e73bb8a4c939544c1302443","project":"openstack/operations-guide","branch":"master","topic":"openstack/openstack-manuals","hashtags":[],"change_id":"I35276e0635893c3b5e73bb8a4c939544c1302443","subject":"Updated from openstack-manuals","status":"MERGED","created":"2016-01-14 07:22:14.000000000","updated":"2016-01-14 08:32:41.000000000","submitted":"2016-01-14 08:32:41.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"61a44986588d389dc852aa629efce68a22da1251","_number":267354,"virtual_id_number":267354,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":2,"date":"2016-01-14 08:32:41.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"all":[{"value":2,"date":"2016-01-14 08:27:31.000000000","_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-01-14 07:22:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"all":[{"value":1,"date":"2016-01-14 08:27:31.000000000","_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-01-14 07:22:58.000000000","updated_by":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"reviewer":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"state":"REVIEWER"},{"updated":"2016-01-14 08:27:31.000000000","updated_by":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"reviewer":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"state":"REVIEWER"},{"updated":"2016-01-14 08:32:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1e3dc7105f4be6d5afa2d97ac94311cfe0dff29d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2016-01-14 07:22:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73b22a7eff58002a19a6ebd59da2da22b01498cd","author":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"date":"2016-01-14 07:22:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ec555ce17161193bcb4fbcdfe05a7ca8bca646ff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-14 07:27:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-operations-guide-tox-checkniceness http://logs.openstack.org/54/267354/1/check/gate-operations-guide-tox-checkniceness/f1d45fa/ : SUCCESS in 2m 29s\n- gate-operations-guide-tox-checksyntax http://logs.openstack.org/54/267354/1/check/gate-operations-guide-tox-checksyntax/fff0c7b/ : SUCCESS in 1m 49s\n- gate-operations-guide-tox-checkdeletions http://logs.openstack.org/54/267354/1/check/gate-operations-guide-tox-checkdeletions/2bbf5b8/ : SUCCESS in 2m 36s\n- gate-operations-guide-tox-doc-publish-checkbuild http://docs-draft.openstack.org/54/267354/1/check/gate-operations-guide-tox-doc-publish-checkbuild/003da27//publish-docs/ : SUCCESS in 2m 19s\n- gate-operations-guide-tox-checklang http://logs.openstack.org/54/267354/1/check/gate-operations-guide-tox-checklang/0414996/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"92495b60809efc9e3b5dac51f8a4da7cd7b153c2","author":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"date":"2016-01-14 08:27:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fb7cf175c6d5b9a9b42bbb90ad33dbc171a2394","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-14 08:27:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"6aac9e26b11bf2d8bae4c70ebd44ca4e83a150a2","date":"2016-01-14 08:32:41.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"6b795e84673644de6bd6b9579aa53e4f9b1a8819","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-14 08:32:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-operations-guide-tox-checkniceness http://logs.openstack.org/54/267354/1/gate/gate-operations-guide-tox-checkniceness/7aacfc7/ : SUCCESS in 2m 22s\n- gate-operations-guide-tox-checksyntax http://logs.openstack.org/54/267354/1/gate/gate-operations-guide-tox-checksyntax/cba1100/ : SUCCESS in 2m 00s\n- gate-operations-guide-tox-checkdeletions http://logs.openstack.org/54/267354/1/gate/gate-operations-guide-tox-checkdeletions/05737eb/ : SUCCESS in 2m 02s\n- gate-operations-guide-tox-doc-publish-checkbuild http://docs-draft.openstack.org/54/267354/1/gate/gate-operations-guide-tox-doc-publish-checkbuild/1b3a1ef//publish-docs/ : SUCCESS in 1m 50s\n- gate-operations-guide-tox-checklang http://logs.openstack.org/54/267354/1/gate/gate-operations-guide-tox-checklang/4112fe2/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e7cb23918de4e7c58b7734862be981c9f35a9c9","revisions":{"3e7cb23918de4e7c58b7734862be981c9f35a9c9":{"kind":"REWORK","_number":1,"created":"2016-01-14 07:22:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/267354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/operations-guide","ref":"refs/changes/54/267354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/54/267354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/54/267354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/54/267354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/operations-guide refs/changes/54/267354/1"}}},"commit":{"parents":[{"commit":"453bb94b3906cb149183227662657113d255b998","subject":"Updated from openstack-manuals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/453bb94b3906cb149183227662657113d255b998"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-01-14 07:22:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-01-14 07:22:14.000000000","tz":0},"subject":"Updated from openstack-manuals","message":"Updated from openstack-manuals\n\nChange-Id: I35276e0635893c3b5e73bb8a4c939544c1302443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/3e7cb23918de4e7c58b7734862be981c9f35a9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/3e7cb23918de4e7c58b7734862be981c9f35a9c9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"453bb94b3906cb149183227662657113d255b998","is_merged_in_target_branch":true,"change_id":"I60eed731608087d0a73fc0d88691f6452e53a83c","change_number":266735,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
