)]}'
{"id":"openstack%2Foperations-guide~619499","triplet_id":"openstack%2Foperations-guide~master~Ia19f18bdb209d43d00cb0684337da6634dd255a1","project":"openstack/operations-guide","branch":"master","hashtags":[],"change_id":"Ia19f18bdb209d43d00cb0684337da6634dd255a1","subject":"Sync Glossary from openstack-manuals","status":"ABANDONED","created":"2018-11-22 10:26:27.000000000","updated":"2018-11-26 16:51:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0781cd071b4cd09413edbfaa54be3fbfa0929fb","_number":619499,"virtual_id_number":619499,"owner":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-11-24 16:25:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"date":"2018-11-24 16:20:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2018-11-24 16:20:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-22 16:03:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-11-24 16:20:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-11-24 16:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca9f79bba578ad6216241f19d0f7146f7d40454b","author":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"date":"2018-11-22 10:26:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fa34f131ef929741bd6cbea461b1653fdffa8ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-22 10:31:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/619499/1/check/openstack-tox-docs/b810037/html/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3663a037876f9ba19b9d1c1809d2e8362c4b3be8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-11-22 16:03:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"021d06db1c0f9683b3a9ff1d56798b7a974565db","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-24 16:20:23.000000000","message":"Patch Set 1:\n\nwe sync this now automatically, this is obsolete - let\u0027s rebase to show it ;)","accounts_in_message":[],"_revision_number":1},{"id":"1bb85d8dcf982b1c5bb133f54c0f71d763f94bb5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-24 16:20:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6e599d1d8b40f23b8aa34128500d50b565d09f1e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-24 16:20:46.000000000","message":"Patch Set 2: Code-Review-1\n\nnothing to do here anymore after rebase - so, let\u0027s abandon.","accounts_in_message":[],"_revision_number":2},{"id":"4b8fee463cd5fdd8ddc8b294482f602866a84993","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-11-24 16:21:07.000000000","message":"Patch Set 2:\n\nhttps://review.openstack.org/#/c/619886/1 did the change that was proposed here","accounts_in_message":[],"_revision_number":2},{"id":"489ad2ed0dd2e07082c26dfd00811033abe1d0dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-24 16:25:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/619499/2/check/openstack-tox-docs/2203a34/html/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1b7f5531eba64ca7b22e0688f0318a2bbefbada5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-11-25 19:39:31.000000000","message":"Removed Code-Review+2 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"a0781cd071b4cd09413edbfaa54be3fbfa0929fb","author":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"date":"2018-11-26 16:51:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f40c197c94d2544d136a76dbe8be9201d1c09dc0","revisions":{"9ec7c2597566fcea3eebb85ea461e3b40a3279f6":{"kind":"REWORK","_number":1,"created":"2018-11-22 10:26:27.000000000","uploader":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"ref":"refs/changes/99/619499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/operations-guide","ref":"refs/changes/99/619499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/1"}}},"commit":{"parents":[{"commit":"d647a8d0d925422dc1d759f599bf2dbd6415ef15","subject":"Fix broken links","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/d647a8d0d925422dc1d759f599bf2dbd6415ef15"}]}],"author":{"name":"Frank Kloeker","email":"f.kloeker@telekom.de","date":"2018-11-22 10:26:07.000000000","tz":60},"committer":{"name":"Frank Kloeker","email":"f.kloeker@telekom.de","date":"2018-11-22 10:26:07.000000000","tz":60},"subject":"Sync Glossary from openstack-manuals","message":"Sync Glossary from openstack-manuals\n\nChange-Id: Ia19f18bdb209d43d00cb0684337da6634dd255a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/9ec7c2597566fcea3eebb85ea461e3b40a3279f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/9ec7c2597566fcea3eebb85ea461e3b40a3279f6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d647a8d0d925422dc1d759f599bf2dbd6415ef15","is_merged_in_target_branch":true,"change_id":"I5787531c3034c070bbdff6f1b904f391b40f2031","change_number":613758,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f40c197c94d2544d136a76dbe8be9201d1c09dc0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-11-24 16:20:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/99/619499/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/operations-guide","ref":"refs/changes/99/619499/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/operations-guide refs/changes/99/619499/2"}}},"commit":{"parents":[{"commit":"a3e1dd0672918db07f9c346f42ce87b72b79e55c","subject":"Updated from openstack-manuals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/a3e1dd0672918db07f9c346f42ce87b72b79e55c"}]}],"author":{"name":"Frank Kloeker","email":"f.kloeker@telekom.de","date":"2018-11-22 10:26:07.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2018-11-24 16:20:27.000000000","tz":0},"subject":"Sync Glossary from openstack-manuals","message":"Sync Glossary from openstack-manuals\n\nChange-Id: Ia19f18bdb209d43d00cb0684337da6634dd255a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/f40c197c94d2544d136a76dbe8be9201d1c09dc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/f40c197c94d2544d136a76dbe8be9201d1c09dc0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a3e1dd0672918db07f9c346f42ce87b72b79e55c","is_merged_in_target_branch":true,"change_id":"I6698d2787dedc14e3e3cb430431642a4778662dc","change_number":619886,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
