)]}'
{"id":"openstack%2Foperations-guide~633195","triplet_id":"openstack%2Foperations-guide~master~I30c39d1eea901b41efece96c95cfb4f9bdc0b49c","project":"openstack/operations-guide","branch":"master","hashtags":[],"change_id":"I30c39d1eea901b41efece96c95cfb4f9bdc0b49c","subject":"skip translation warnings on tox build","status":"MERGED","created":"2019-01-25 12:57:37.000000000","updated":"2019-01-25 14:02:02.000000000","submitted":"2019-01-25 14:02:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"633195-1548424922339-55140b31","meta_rev_id":"7c1c39b6a84d1a1d9ec47b35ccc06e5e08e9c78e","_number":633195,"virtual_id_number":633195,"owner":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-01-25 14:02:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-01-25 13:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-01-25 13:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-01-25 13:50:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-25 13:50:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-01-25 13:57:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-01-25 14:02:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89b4a050d5bcc63e017ac0991941be0357afbcd7","author":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"date":"2019-01-25 12:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea6427a69ab5f38b1f7f2275f5a3f0099843bdbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 13:10:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633195/1/check/openstack-tox-docs/467ebcc/html/ : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"bbad76427ac0b02a40edafbdb189f5ac2d2dd1f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-01-25 13:50:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c143d4d22871ed6f306fd355d251a61d18e6279c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 13:50:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cda9cd291ff7128aeac74249ec44e737787cfd42","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-01-25 13:57:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d53dff4920ae3660e74b3c7b100d4dfa1d363e7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 14:02:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c1c39b6a84d1a1d9ec47b35ccc06e5e08e9c78e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 14:02:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/633195/1/gate/openstack-tox-docs/972b71e/html/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef75f4e066e0a8b010b0c78240553bb5dd86563e","revisions":{"ef75f4e066e0a8b010b0c78240553bb5dd86563e":{"kind":"REWORK","_number":1,"created":"2019-01-25 12:57:37.000000000","uploader":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"ref":"refs/changes/95/633195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/operations-guide","ref":"refs/changes/95/633195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/95/633195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/95/633195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/95/633195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/operations-guide refs/changes/95/633195/1"}}},"commit":{"parents":[{"commit":"d67cc82953dc92902edaaf42148f9e803a5d0f32","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/d67cc82953dc92902edaaf42148f9e803a5d0f32"}]}],"author":{"name":"Frank Kloeker","email":"f.kloeker@telekom.de","date":"2019-01-25 12:57:21.000000000","tz":60},"committer":{"name":"Frank Kloeker","email":"f.kloeker@telekom.de","date":"2019-01-25 12:57:21.000000000","tz":60},"subject":"skip translation warnings on tox build","message":"skip translation warnings on tox build\n\nChange-Id: I30c39d1eea901b41efece96c95cfb4f9bdc0b49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/ef75f4e066e0a8b010b0c78240553bb5dd86563e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/ef75f4e066e0a8b010b0c78240553bb5dd86563e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d67cc82953dc92902edaaf42148f9e803a5d0f32","is_merged_in_target_branch":true,"change_id":"I95411e02adb7d461ed15f86b84744cab01494ae3","change_number":631709,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
