)]}'
{"id":"openstack%2Foperations-guide~989928","triplet_id":"openstack%2Foperations-guide~master~Ibb0fb7e2159123463ecdbeaff41b9e573afc7e14","project":"openstack/operations-guide","branch":"master","topic":"openstack/openstack-manuals","hashtags":[],"change_id":"Ibb0fb7e2159123463ecdbeaff41b9e573afc7e14","subject":"Updated from openstack-manuals","status":"MERGED","created":"2026-05-25 06:12:04.000000000","updated":"2026-09-25 11:53:59.000000000","submitted":"2026-09-25 11:52:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989928-openstack/openstack-manuals","meta_rev_id":"d6eb3eeec93dc50e49f88d6d1f7b99b82eddfda3","_number":989928,"virtual_id_number":989928,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 11:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-25 11:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-25 11:37:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 06:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 11:37:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8be1dc9c21b727246bcf058ea58d0f4b92859b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-25 06:12:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7ec3ee145e0fdba928dbf9b03c2250b8adb9259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 06:23:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e37265966d241719c93ef7f6e634e85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0eb78400cca4c6381d0ef6d501afc33 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6c3cc61fa74a38ad9549c98cc37d8c176976cedd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-25 06:30:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8df32c715a5fbbca347436b44acdcaf44468f33e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 06:43:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baaf853e72f949d2b2b3ef664f531e06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/346898260e0d41d585d5e8b9c97dcd43 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"a63f86d43e44c11743a87b7c18833af040c1c84e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-25 11:34:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"109b9bbc0cc2cea5008f33d2ed72937761265f80","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-25 11:37:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"84b7ae69471d4dbfb5d9654f5ccd00c7ecc4567c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:42:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861e6194fbf3476593f92ca4ebd425e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5510cc2e32499ea646b4107711b5a9 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":3},{"id":"950604292b265369aa067ea29f42a6d3bd267666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:42:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8bfcb30a5907dd7781d87a646921452e5f4b5fc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:52:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f1a9f9477949e394f59c84ee0c15d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84facaec2c1b476eab4fe097a26680c8 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":3},{"id":"30f0a49918425964312c41ed75f826a979609902","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:52:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d6eb3eeec93dc50e49f88d6d1f7b99b82eddfda3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:53:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43884f0abc92490a93f44b808e768109\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/91630ce6553d48a2956719dfd2f3d9a9 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3eae2dee237fe2255ed1e0e2f182983e3b2840a3","revisions":{"c3e4b3933b3288ff89a403551ac67cb703bc3837":{"kind":"REWORK","_number":1,"created":"2026-05-25 06:12:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/989928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/operations-guide","ref":"refs/changes/28/989928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/1"}}},"commit":{"parents":[{"commit":"e88f8a41c444345757805c584fb0f4dfb4bbf66c","subject":"Updated from openstack-manuals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/e88f8a41c444345757805c584fb0f4dfb4bbf66c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-25 06:12:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-25 06:12:02.000000000","tz":0},"subject":"Updated from openstack-manuals","message":"Updated from openstack-manuals\n\nChange-Id: Ibb0fb7e2159123463ecdbeaff41b9e573afc7e14\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/c3e4b3933b3288ff89a403551ac67cb703bc3837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/c3e4b3933b3288ff89a403551ac67cb703bc3837"}]},"branch":"refs/heads/master"},"04da7a1a21930b18eb84e00b88760951e7006b80":{"kind":"REWORK","_number":2,"created":"2026-05-25 06:30:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/989928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/operations-guide","ref":"refs/changes/28/989928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/2"}}},"commit":{"parents":[{"commit":"e88f8a41c444345757805c584fb0f4dfb4bbf66c","subject":"Updated from openstack-manuals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/e88f8a41c444345757805c584fb0f4dfb4bbf66c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-25 06:30:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-25 06:30:33.000000000","tz":0},"subject":"Updated from openstack-manuals","message":"Updated from openstack-manuals\n\nChange-Id: Ibb0fb7e2159123463ecdbeaff41b9e573afc7e14\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/04da7a1a21930b18eb84e00b88760951e7006b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/04da7a1a21930b18eb84e00b88760951e7006b80"}]},"branch":"refs/heads/master"},"3eae2dee237fe2255ed1e0e2f182983e3b2840a3":{"kind":"REWORK","_number":3,"created":"2026-09-25 11:34:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/989928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/operations-guide","ref":"refs/changes/28/989928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/operations-guide refs/changes/28/989928/3"}}},"commit":{"parents":[{"commit":"e88f8a41c444345757805c584fb0f4dfb4bbf66c","subject":"Updated from openstack-manuals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/e88f8a41c444345757805c584fb0f4dfb4bbf66c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-25 11:34:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-25 11:34:43.000000000","tz":0},"subject":"Updated from openstack-manuals","message":"Updated from openstack-manuals\n\nChange-Id: Ibb0fb7e2159123463ecdbeaff41b9e573afc7e14\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/3eae2dee237fe2255ed1e0e2f182983e3b2840a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/operations-guide/commit/3eae2dee237fe2255ed1e0e2f182983e3b2840a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
