)]}'
{"id":"openstack%2Fos-acc~667367","triplet_id":"openstack%2Fos-acc~master~I081ef03d418b21bb7c7d95e27d378f55e39c34c3","project":"openstack/os-acc","branch":"master","hashtags":[],"change_id":"I081ef03d418b21bb7c7d95e27d378f55e39c34c3","subject":"Update sphinx version","status":"ABANDONED","created":"2019-06-25 13:47:38.000000000","updated":"2019-09-17 06:03:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a573242b9985cd7c8f5ba723e985a5e7d8ac3bb8","_number":667367,"virtual_id_number":667367,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":-1,"date":"2019-06-25 19:02:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":-1,"date":"2019-09-16 22:03:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 19:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-16 22:03:42.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"}],"messages":[{"id":"abb1ffacab2af7a57d918752e79c5745ac520447","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-25 13:47:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b16137875fa8980eb44a62f744ae82501b3d6da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:02:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/67/667367/1/check/requirements-check/94a40ab/ : FAILURE in 2m 09s\n- openstack-tox-docs http://logs.openstack.org/67/667367/1/check/openstack-tox-docs/46f751d/html/ : SUCCESS in 2m 35s\n- openstack-tox-pep8 http://logs.openstack.org/67/667367/1/check/openstack-tox-pep8/700c729/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/67/667367/1/check/openstack-tox-py27/b626b4d/ : SUCCESS in 2m 40s\n- openstack-tox-py35 http://logs.openstack.org/67/667367/1/check/openstack-tox-py35/a5a6dab/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/67/667367/1/check/openstack-tox-py36/3ac440a/ : SUCCESS in 2m 36s\n- tempest-full http://logs.openstack.org/67/667367/1/check/tempest-full/63f09c4/ : SUCCESS in 2h 00m 10s\n- tempest-full-py3 http://logs.openstack.org/67/667367/1/check/tempest-full-py3/51b4d29/ : SUCCESS in 1h 37m 02s\n- build-openstack-releasenotes http://logs.openstack.org/67/667367/1/check/build-openstack-releasenotes/fc6dccf/html/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3479d447100ee77ff5e9e39e535acbfe1bd657d2","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-16 22:03:42.000000000","message":"Patch Set 1: Code-Review-1\n\nos-acc is going away. We can drop this.\n\nSee https://review.opendev.org/#/c/682498/","accounts_in_message":[],"_revision_number":1},{"id":"a573242b9985cd7c8f5ba723e985a5e7d8ac3bb8","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-17 06:03:30.000000000","message":"Abandoned\n\nAbandoned because os-acc project is being retired. See https://review.opendev.org/#/c/682498/.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2a28c2d8813ceb3ad136b9079b84d78392265b5","revisions":{"e2a28c2d8813ceb3ad136b9079b84d78392265b5":{"kind":"REWORK","_number":1,"created":"2019-06-25 13:47:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/667367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-acc","ref":"refs/changes/67/667367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-acc refs/changes/67/667367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-acc refs/changes/67/667367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-acc refs/changes/67/667367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-acc refs/changes/67/667367/1"}}},"commit":{"parents":[{"commit":"a78513033bb185aa4817817fb83b8126bf41b990","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-acc/commit/a78513033bb185aa4817817fb83b8126bf41b990"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-06-25 13:47:04.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-06-25 13:47:04.000000000","tz":0},"subject":"Update sphinx version","message":"Update sphinx version\n\nThis commit updates sphinx version which aligns to g-r. Without this\npatch, py27 tests fail because of an error like this:\n\n ERROR: Sphinx requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.15\n\nChange-Id: I081ef03d418b21bb7c7d95e27d378f55e39c34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-acc/commit/e2a28c2d8813ceb3ad136b9079b84d78392265b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-acc/commit/e2a28c2d8813ceb3ad136b9079b84d78392265b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
