)]}'
{"id":"openstack%2Fos-brick~1007282","triplet_id":"openstack%2Fos-brick~master~I0e387b5d42c549ad3f8c5746977cfa4420ab7336","project":"openstack/os-brick","branch":"master","topic":"privsep","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-25 13:09:25.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0e387b5d42c549ad3f8c5746977cfa4420ab7336","subject":"privsep: Harden opencas utils","status":"NEW","created":"2026-09-24 17:37:01.000000000","updated":"2026-10-05 18:26:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4a4fa80cf5b126ee4891f15a53ae0fe23c47fe4","_number":1007282,"virtual_id_number":1007282,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-02 19:00:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-24 19:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-24 20:38:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-25 12:11:09.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-25 13:09:25.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-28 14:17:08.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-28 15:22:55.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"ea367484a0660faf2c3ca52d6ab4e9cdbc630466","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-24 17:37:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55b15599b777ae0a582272271f2602b6e2545743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 19:38:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc9b860fe0954c378bbe52b3abdef62c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7619febbda734f5db7c75bbf2ae1396c : SUCCESS in 1h 36m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7761a98d2efe47949fe8fd524faf4dec : FAILURE in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c09733e254a5492e91e7f53b1de61c2c : SUCCESS in 4m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1dbc6aa98aa4d67b828aa63cf833936 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54c10c4d77a43ff8c79ebccefffbb40 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcc922a4f0304f588814ab8d00bfb814 : SUCCESS in 7m 22s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/088f4810058b42f0990877928060e7e7 : SUCCESS in 4m 29s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/90fdf02a6eb041a9916dbd81d48cf305 : SUCCESS in 1h 35m 46s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2b3577e107b40f7a3fd0eddc38b01ad : SUCCESS in 1h 45m 01s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/13aa496d1ea5468d90cc06f3dfc78e8c : SUCCESS in 1h 44m 58s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/1139099ff49e4cd985c885c20a620689 : SUCCESS in 3m 49s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/caches/__init__.py","accounts_in_message":[],"_revision_number":1},{"id":"407f30389ae41924a5d73c25a51bdf766b3c67b4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-24 20:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/91f56bceb74e4cb9a4e7a6514e2e6064 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ecbe8f23e05837df5782800c4f3eafb89176e98b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 12:11:09.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0e387b5d42c549ad3f8c5746977cfa4420ab7336 .","accounts_in_message":[],"_revision_number":1},{"id":"4b69ba119f81a403e96be361761d2d9b5e5d84a1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-25 13:09:25.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5696/.","accounts_in_message":[],"_revision_number":1},{"id":"0e8d0b0655b4fd7c6575a0587400817ca540a83c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-25 15:03:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5519/.","accounts_in_message":[],"_revision_number":1},{"id":"0e1f83c1675834a9184a183fc573eeef1c2c70f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-28 12:23:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f613fb04fa07d32846d9fd96d47a2e7edf4a28ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-28 13:08:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4b083d03ba499fb0512648d05aa6db8f2d7cd3d6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-28 13:30:55.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0e387b5d42c549ad3f8c5746977cfa4420ab7336 .","accounts_in_message":[],"_revision_number":2},{"id":"9dd3ac9ef831e338b500c29b86465d1cdbab8aba","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-28 14:17:08.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/os-brick-CI/tree/gh-pages/60e95d0cf0cd4b73b1aa6ed34d0b3a2f/","accounts_in_message":[],"_revision_number":3},{"id":"8e4ab1904f35a67c7a1c82fa5324304902a12842","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-28 14:30:57.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0e387b5d42c549ad3f8c5746977cfa4420ab7336 .","accounts_in_message":[],"_revision_number":3},{"id":"ac3e31bd5c34a4383eeb53a26af52a86e91aa9aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 15:19:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1e0dc9ad78843d5b61b2e9edd630215\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8179899b843542ea8b4f5f4e0299cde1 : SUCCESS in 42m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/890795650a744e5d9f85021c7b2f0005 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f624f602fce44ee7b897b25e90528a00 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fabaed14d1b740c79aacc2bde6209b38 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896aa947651747eea5446a101eff0506 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/602dd1e578c94b42a2faf0683b026a92 : SUCCESS in 7m 00s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/501243f4bdf148b2ae1221c568d26136 : SUCCESS in 4m 33s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/8e75a834dfdd46edace53e0f9f8ecb59 : SUCCESS in 45m 07s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1e2f8066861b48489b50a803c489cacb : SUCCESS in 2h 05m 20s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/46634f400c684b26b40dd5133c8602bf : SUCCESS in 1h 30m 08s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/1368ac46a3ce4b1ca921ccf8bfb1f400 : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e42c0e8939544f776c96afe1a5b263d6cb42691","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-28 15:28:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/60e95d0cf0cd4b73b1aa6ed34d0b3a2f\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fdf950e745214a2a947861a15d033ace : SUCCESS in 38m 50s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/762ef0009e004a438dc9530b18f12340 : SUCCESS in 40m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b0ff0b37b9249d6ac3b91ea9f79c526 : SUCCESS in 44m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/96a86b7123db479da0a333b90d0742b3 : SUCCESS in 29m 43s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57bbf674106948218fcae73441eeb6d2 : FAILURE in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ce7ad3b2a3598779904698caec70d8f6b2fb219","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-28 15:32:35.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_09_2026_15_26_45\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"970f9c2d5184114663463178716d129c6ae68ccf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-28 19:05:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/96030b3a4ed6473d8174175a94b9b02b : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":3},{"id":"acd90359e1fe12d74c79c893e109ea79920764e6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-28 22:00:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5558/.","accounts_in_message":[],"_revision_number":2},{"id":"b97afda37c72a536371e507cf1635f70371baa2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-28 22:15:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5735/.","accounts_in_message":[],"_revision_number":2},{"id":"00b5b8ec1dacacf7eb83f0e0dc6c0e2de269a736","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-29 05:01:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5565/.","accounts_in_message":[],"_revision_number":3},{"id":"fc9b50f3a8c103d54e54c1817a4b75f9197292d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-29 06:06:23.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5742/.","accounts_in_message":[],"_revision_number":3},{"id":"0892620a2edf52cef95d925c0470ee4c64a112a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 17:06:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"40df52c4f7d250951890f1ad2cf610576686f383","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 17:56:51.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/os-brick-CI/tree/gh-pages/1c52d954be6244aba6fadfc4a62b7d0f/","accounts_in_message":[],"_revision_number":4},{"id":"005a3c56aa365ed65e2b0e13a66012059e5f7616","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 18:05:14.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0e387b5d42c549ad3f8c5746977cfa4420ab7336 .","accounts_in_message":[],"_revision_number":4},{"id":"cb2d1717c6d434a1e4384a0b14c359b4552635e1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 18:33:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1c52d954be6244aba6fadfc4a62b7d0f\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba794b91f6af44928a2066d04addb509 : SUCCESS in 37m 31s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c0ed555144f48699b2c1b3f6b7bf011 : SUCCESS in 35m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6dce0404f294bd19076b0f4cf721a69 : SUCCESS in 40m 22s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c04dfe4be63845a8a13110b61c4aeddd : SUCCESS in 28m 31s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c7c0e0fe78943b5bba8034699e4f55e : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"779c3d95648f9d455b836d364961d5e03ce9de7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 19:00:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5592559e11aa4d209f794b40eadb5357\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64b09070c3664f5795394af5fc7a18ae : SUCCESS in 46m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c622749a2a8942228d81620130a03dc7 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8b168388fed4423858e2711ff2bc4c8 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b3226883ff01459b86bb44db4fa4cdb7 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a939d0f9a9ec4b44809a257d74a44478 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8d15052086a46c2911acaf64d791316 : SUCCESS in 6m 35s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/4b74740f996644739b9f62a993f15ba0 : SUCCESS in 4m 26s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/145b130193de4a08ae9363dd726abac2 : SUCCESS in 45m 47s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d3ae8da6c93847c0bbca6a3e0c292de5 : SUCCESS in 1h 47m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/2da11918260046b6b8664cf725203da1 : SUCCESS in 1h 47m 24s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4a700cacfb0e439dbc17ef830ab6a3ed : SUCCESS in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8a9fa8e075b3864a7f24e510a90a263a7b52905","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-10-02 19:18:58.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_10_2026_19_13_01\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"d8f1936a9f81ba8872f455d298aae97dd5726650","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-02 21:41:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/2524a03dda4a4c3889d940bd5094f8b0 : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":4},{"id":"41bc242c55c47e679cb362301102669e81486924","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 16:43:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5659/.","accounts_in_message":[],"_revision_number":4},{"id":"a4a4fa80cf5b126ee4891f15a53ae0fe23c47fe4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-05 18:26:40.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5838/.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"57ac59ba05b336c2787ece48d86385e87c141dba","revisions":{"4314a2c08e009f99235ffa429e980ec795643948":{"kind":"REWORK","_number":1,"created":"2026-09-24 17:37:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/1007282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/1007282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/1"}}},"commit":{"parents":[{"commit":"e8279339461ed498fb77e523290f0725ad80955b","subject":"privsep: Convert \u0027os_brick.caches.opencas\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e8279339461ed498fb77e523290f0725ad80955b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-24 17:32:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-24 17:35:31.000000000","tz":60},"subject":"privsep: Harden opencas utils","message":"privsep: Harden opencas utils\n\nReplace use of find, which can return partial matches, with exact\nmatches.\n\nChange-Id: I0e387b5d42c549ad3f8c5746977cfa4420ab7336\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4314a2c08e009f99235ffa429e980ec795643948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4314a2c08e009f99235ffa429e980ec795643948"}]},"branch":"refs/heads/master"},"05b4168587a972e85b9c2f4a4e02d9f687cc4009":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-28 12:23:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/1007282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/1007282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/2"}}},"commit":{"parents":[{"commit":"0ad2c55ce5aa1620805b4a5cc9737b57422b88e0","subject":"privsep: Convert \u0027os_brick.caches.opencas\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ad2c55ce5aa1620805b4a5cc9737b57422b88e0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-24 17:32:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-28 11:20:48.000000000","tz":60},"subject":"privsep: Harden opencas utils","message":"privsep: Harden opencas utils\n\nReplace use of find, which can return partial matches, with exact\nmatches.\n\nChange-Id: I0e387b5d42c549ad3f8c5746977cfa4420ab7336\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/05b4168587a972e85b9c2f4a4e02d9f687cc4009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/05b4168587a972e85b9c2f4a4e02d9f687cc4009"}]},"branch":"refs/heads/master"},"c3d3d8550d81c4d4289d99b534605c1a0f344359":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-28 13:08:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/1007282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/1007282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/3"}}},"commit":{"parents":[{"commit":"9f0f2f65a7bf9ec1404c707e9cbc3d81629e66f8","subject":"privsep: Convert \u0027os_brick.caches.opencas\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9f0f2f65a7bf9ec1404c707e9cbc3d81629e66f8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-24 17:32:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-28 13:02:30.000000000","tz":60},"subject":"privsep: Harden opencas utils","message":"privsep: Harden opencas utils\n\nReplace use of find, which can return partial matches, with exact\nmatches.\n\nChange-Id: I0e387b5d42c549ad3f8c5746977cfa4420ab7336\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c3d3d8550d81c4d4289d99b534605c1a0f344359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c3d3d8550d81c4d4289d99b534605c1a0f344359"}]},"branch":"refs/heads/master"},"57ac59ba05b336c2787ece48d86385e87c141dba":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-10-02 17:06:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/1007282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/1007282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/1007282/4"}}},"commit":{"parents":[{"commit":"da4a3a6fcb1896753c3ac5057d0db619ca84570c","subject":"privsep: Convert \u0027os_brick.caches.opencas\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/da4a3a6fcb1896753c3ac5057d0db619ca84570c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-24 17:32:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-02 17:06:21.000000000","tz":60},"subject":"privsep: Harden opencas utils","message":"privsep: Harden opencas utils\n\nReplace use of find, which can return partial matches, with exact\nmatches.\n\nChange-Id: I0e387b5d42c549ad3f8c5746977cfa4420ab7336\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/57ac59ba05b336c2787ece48d86385e87c141dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/57ac59ba05b336c2787ece48d86385e87c141dba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
