)]}'
{"id":"openstack%2Fos-brick~658481","triplet_id":"openstack%2Fos-brick~master~Icc2ff9a6be9e6464b897238839b58aae38f52ba5","project":"openstack/os-brick","branch":"master","topic":"cve-2018-18074","hashtags":[],"change_id":"Icc2ff9a6be9e6464b897238839b58aae38f52ba5","subject":"DNM: assess impact of requests update","status":"ABANDONED","created":"2019-05-10 20:48:49.000000000","updated":"2019-05-10 20:51:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e435fefab649877b04fb2c6f28fc0ce7594e72b","_number":658481,"virtual_id_number":658481,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"all":[{"date":"2019-05-10 20:50:01.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 20:50:01.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f3e153b6b4a1f16e47d08c33b212e95a83a6687","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-10 20:48:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d52bd8b39abd729b73bf65b7751ae7146cb4f1a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-05-10 20:50:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/81/658481/1/check/dsvm-tempest-kvm/6649c3b : FAILURE in 3s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/81/658481/1/check/dsvm-tempest-exe/c0f91d6 : FAILURE in 3s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"6d9b7c5809377b9c1046bc02bd66d2994e026992","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-10 20:51:32.000000000","message":"Abandoned\n\nIgnore this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2a9a730aaee5e256424dd9f2f824d984a9257fd","revisions":{"f2a9a730aaee5e256424dd9f2f824d984a9257fd":{"kind":"REWORK","_number":1,"created":"2019-05-10 20:48:49.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/81/658481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/81/658481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/81/658481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/81/658481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/81/658481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/81/658481/1"}}},"commit":{"parents":[{"commit":"8cb5fe3623f2145dc49742211f69f1f773aa7d3a","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8cb5fe3623f2145dc49742211f69f1f773aa7d3a"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-10 20:47:12.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-10 20:48:48.000000000","tz":-240},"subject":"DNM: assess impact of requests update","message":"DNM: assess impact of requests update\n\nSee http://lists.openstack.org/pipermail/openstack-discuss/2019-May/005956.html\n\nDepends-on: https://review.opendev.org/640710\nChange-Id: Icc2ff9a6be9e6464b897238839b58aae38f52ba5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f2a9a730aaee5e256424dd9f2f824d984a9257fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f2a9a730aaee5e256424dd9f2f824d984a9257fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
