)]}'
{"id":"openstack%2Fos-brick~665641","triplet_id":"openstack%2Fos-brick~stable%2Fstein~I02c208142f5b342f894666831449243863eccbfe","project":"openstack/os-brick","branch":"stable/stein","hashtags":[],"change_id":"I02c208142f5b342f894666831449243863eccbfe","subject":"FC: Ignore some HBAs from map for single WWNN","status":"MERGED","created":"2019-06-17 10:33:12.000000000","updated":"2019-06-26 04:55:14.000000000","submitted":"2019-06-21 17:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665641-1561136840795-f790302f","meta_rev_id":"3b7a9b0bfaf10bf07a9a92be7884f8e8d5f34fcf","_number":665641,"virtual_id_number":665641,"owner":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-06-17 13:05:27.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-17 12:09:42.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-17 15:27:13.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":2,"date":"2019-06-21 17:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-17 13:43:03.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-21 15:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":1,"date":"2019-06-18 06:05:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":2,"date":"2019-06-19 08:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-21 15:19:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"date":"2019-06-18 13:56:02.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 12:09:42.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-17 13:05:27.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-17 13:43:03.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-06-17 15:27:13.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-06-18 06:05:18.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-06-18 13:56:02.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-06-19 08:54:34.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2019-06-21 15:19:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-21 17:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e174d6dc4afbe2d5251553f955c9fbd2c12ed88","author":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"date":"2019-06-17 10:33:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dbc00cb980b9ee029e34c7c60989d3902ac17c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-17 12:09:42.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/os-brick/41/665641/1/check/tempest-dsvm-full-xenial/27ca7c8/ : FAILURE in 1h 36m 15s\n- devstack-plugin-ceph-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/os-brick/41/665641/1/check/devstack-plugin-ceph-dsvm-xenial/5a7a181/ : FAILURE in 1h 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f375bb6eb19c28b47c55de0e324d97edc6fd2d77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 12:36:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/41/665641/1/check/tempest-full/74af382/ : SUCCESS in 1h 59m 26s\n- tempest-full-py3 http://logs.openstack.org/41/665641/1/check/tempest-full-py3/cd8e464/ : SUCCESS in 1h 55m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/665641/1/check/openstack-tox-lower-constraints/14b560c/ : SUCCESS in 7m 26s\n- openstack-tox-pep8 http://logs.openstack.org/41/665641/1/check/openstack-tox-pep8/8ff5360/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/41/665641/1/check/openstack-tox-py27/be48187/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/41/665641/1/check/openstack-tox-py36/be83d8e/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/41/665641/1/check/openstack-tox-py37/36176ba/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/41/665641/1/check/openstack-tox-docs/042c5d7/html/ : SUCCESS in 3m 31s\n- build-openstack-releasenotes http://logs.openstack.org/41/665641/1/check/build-openstack-releasenotes/129170b/html/ : SUCCESS in 3m 53s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick http://logs.openstack.org/41/665641/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-os-brick/b557967/ : SUCCESS in 1h 24m 55s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick http://logs.openstack.org/41/665641/1/check/legacy-tempest-dsvm-full-lio-src-os-brick/4366542/ : SUCCESS in 1h 22m 47s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/41/665641/1/check/os-brick-cinder-tempest-dsvm-lvm-lio-barbican/2871a54/ : SUCCESS in 1h 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"763c85a7891b0f00b8536a4f898954c55ecc99d4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-06-17 13:05:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/41/665641/1/check-cinder/Cinder-ISER-ISCSI/0ddf89d : SUCCESS in 1h 04m 05s\n- Cinder-ISER-LIO http://13.74.249.42/41/665641/1/check-cinder/Cinder-ISER-LIO/b5086a4 : SUCCESS in 1h 13m 34s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"ce752e343ebeaf4c4e55cdc9accef8deceb9d464","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-06-17 13:43:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://13.56.179.158/logs/41/665641/1/upstream-solidfire-check/cinder-solidfire-iSCSI/2242ef1/ : FAILURE in 1h 06m 10s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"82ebd0f4a0a8d9525e5003a2be28cd0f480275b3","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-06-17 15:27:13.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/41/665641/1/check/kaminario-dsvm-tempest-full-iscsi/7ee0c2f : FAILURE in 1m 26s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/41/665641/1/check/kaminario-dsvm-tempest-full-FC/3e51718 : SUCCESS in 1h 47m 19s","accounts_in_message":[],"_revision_number":1},{"id":"09c91341f174d9af7aa91245ebddbc42cfb7151c","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-06-18 06:05:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5fc2a8b2edf2b4053ddcd3de825cc9d5ef96a0bb","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-06-19 08:54:34.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for the backport","accounts_in_message":[],"_revision_number":1},{"id":"0ef67352fc2d15d621118b58ee4834ed4907df0e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-21 15:19:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27b9adec5ed85947f65b72c88cca02dc18f86026","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 15:20:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39dc2cf126b9a220832bc8bd3a449a4f1bd4d6d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 17:07:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/41/665641/1/gate/tempest-full/a8569b8/ : SUCCESS in 1h 45m 49s\n- tempest-full-py3 http://logs.openstack.org/41/665641/1/gate/tempest-full-py3/b62b3fc/ : SUCCESS in 1h 38m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/665641/1/gate/openstack-tox-lower-constraints/836d2a9/ : SUCCESS in 5m 59s\n- openstack-tox-pep8 http://logs.openstack.org/41/665641/1/gate/openstack-tox-pep8/6a6237f/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/41/665641/1/gate/openstack-tox-py27/a566fec/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/41/665641/1/gate/openstack-tox-py36/d410af3/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/41/665641/1/gate/openstack-tox-py37/e5a7307/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/41/665641/1/gate/openstack-tox-docs/74d4635/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/41/665641/1/gate/build-openstack-releasenotes/9016107/html/ : SUCCESS in 3m 33s\n- legacy-tempest-dsvm-full-lio-src-os-brick http://logs.openstack.org/41/665641/1/gate/legacy-tempest-dsvm-full-lio-src-os-brick/cf778a5/ : SUCCESS in 1h 28m 21s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican http://logs.openstack.org/41/665641/1/gate/os-brick-cinder-tempest-dsvm-lvm-lio-barbican/9b09d0d/ : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"28559627ab7252844f1eafac06ac8246826f9dac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 17:07:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b7a9b0bfaf10bf07a9a92be7884f8e8d5f34fcf","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2019-06-26 04:55:14.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 64cbaaea47822dff3ae5fbd104f9bff8fe79d08d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ee6acb509910d7de1213fb71cd0d6166991609a","revisions":{"1ee6acb509910d7de1213fb71cd0d6166991609a":{"kind":"REWORK","_number":1,"created":"2019-06-17 10:33:12.000000000","uploader":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"ref":"refs/changes/41/665641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/41/665641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/665641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/665641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/665641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/41/665641/1"}}},"commit":{"parents":[{"commit":"d7f9ff61fb5854fd17c2d713ecd60654d1c9a452","subject":"luks: Default to LUKS v1 when formatting volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d7f9ff61fb5854fd17c2d713ecd60654d1c9a452"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-05-09 15:25:03.000000000","tz":120},"committer":{"name":"Pablo Caruana","email":"pcaruana@redhat.com","date":"2019-06-17 10:33:10.000000000","tz":120},"subject":"FC: Ignore some HBAs from map for single WWNN","message":"FC: Ignore some HBAs from map for single WWNN\n\nCurrent FC OS-Brick code only checks for single WWNN to exclude some\nHBAs from scanning when we don\u0027t receive an initiator_target_map from\nthe backend.\n\nThere are storage arrays,like XtremIO, that due to their architecture\nand design have all target ports automatically mapped to LUNs and are\nreturning the initiator_target_map, but some of those ports may not be\nconnected to our HBAs, so we end up with another case of bug #1765000.\n\nThis patch makes sure that we always check if the target implements\nsingle WWNN, not only when we don\u0027t receive the initiator_target_map.\n\nWith this we decrease the likelihood of ending with unexpected devices\nin our system, because now we will ignore unconnected HBAs (even if\nreported in the initiator_target_map) if we are working with single WWNN\ntarget.\n\nRelated-Bug: #1765000\nCloses-Bug: #1828440\nChange-Id: I02c208142f5b342f894666831449243863eccbfe\n(cherry picked from commit 70899a9aa3e4bdd3f1b1a7024807390cd55c4958)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1ee6acb509910d7de1213fb71cd0d6166991609a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1ee6acb509910d7de1213fb71cd0d6166991609a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
