)]}'
{"id":"openstack%2Fos-brick~674205","triplet_id":"openstack%2Fos-brick~master~Ie1952c7b65542ac493e506c8916da0e1a856e838","project":"openstack/os-brick","branch":"master","topic":"sphinx","hashtags":[],"change_id":"Ie1952c7b65542ac493e506c8916da0e1a856e838","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-08-02 06:57:42.000000000","updated":"2019-08-30 19:39:21.000000000","submitted":"2019-08-28 18:19:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674205-1567016380308-6c2ca45c","meta_rev_id":"66b8fea5de7ea9c01368f0b1df48ea6432baa264","_number":674205,"virtual_id_number":674205,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-22 10:24:37.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 06:23:45.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-30 19:39:21.000000000","post_submit":true,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-08-22 07:03:48.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 17:57:54.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-23 12:03:36.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"date":"2019-08-22 08:00:27.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":2,"date":"2019-08-28 18:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 09:06:59.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2019-08-22 09:08:47.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 09:21:38.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 07:30:41.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 09:07:12.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 09:57:46.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-08-22 06:29:28.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-28 16:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-08-22 22:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-28 16:12:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 10:03:37.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 10:07:42.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-08-14 15:52:46.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-08-21 09:11:26.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-08-21 20:51:46.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-08-22 06:23:45.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 06:29:28.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-08-22 07:03:48.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 07:30:41.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 08:00:27.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 09:06:59.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-08-22 09:07:12.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-08-22 09:08:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 09:21:38.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 09:57:46.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-08-22 10:24:37.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 17:57:54.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 22:14:07.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-08-23 12:03:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 16:12:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-28 18:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 19:39:21.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9904114302b8a002e30632498a6ccaa4e30c3ae","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-02 06:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60d53e784c27d8cff2e6c9efbcdd098c4831d068","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 07:10:26.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/05/674205/1/check/check-dsvm-tempest-vz7-exe-minimal/7974635 : FAILURE in 12m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"d8f98c1d67dad57cf1fc55fd8edfc3eafb8759be","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 07:32:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/674205/1/check/dsvm-tempest-kvm/2842dc1 : FAILURE in 34m 44s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/674205/1/check/dsvm-tempest-exe/3c963c1 : FAILURE in 1m 29s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"b0b63518d9fb2b753d613b77ee1efa1949996d7e","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-08-02 07:47:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/674205/1/iscsi/ : SUCCESS in 49m 33s\n- os-brick-smb http://cloudbase-ci.com/os-brick/674205/1/smb/ : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"87f7e1cbb761dd16de0cbd8449a5c414d71a1aa1","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-02 07:59:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/05/674205/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/83f6010 : SUCCESS in 51m 00s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/05/674205/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/fa3c9d6 : SUCCESS in 1h 01m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"e8a58357e50e50e0236d13cdaf85b5d8cf60302a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-02 08:28:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/05/674205/1/check/tempest-dsvm-full-xenial/2ff54e0/ : SUCCESS in 1h 19m 13s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/05/674205/1/check/devstack-plugin-ceph-dsvm-xenial/4fb4975/ : SUCCESS in 1h 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"130e613729b9c59f72a5bac5b7e1dc8326533722","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 08:56:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/05/674205/1/check/requirements-check/ede6e4e/ : SUCCESS in 6m 10s\n- tempest-full https://logs.opendev.org/05/674205/1/check/tempest-full/728b9bb/ : SUCCESS in 1h 39m 56s\n- tempest-full-py3 https://logs.opendev.org/05/674205/1/check/tempest-full-py3/0f80c28/ : SUCCESS in 1h 52m 35s\n- openstack-tox-lower-constraints https://logs.opendev.org/05/674205/1/check/openstack-tox-lower-constraints/41e4537/ : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://logs.opendev.org/05/674205/1/check/openstack-tox-pep8/4321cf9/ : SUCCESS in 3m 40s\n- openstack-tox-py27 https://logs.opendev.org/05/674205/1/check/openstack-tox-py27/fcdcc3f/ : SUCCESS in 5m 51s\n- openstack-tox-py36 https://logs.opendev.org/05/674205/1/check/openstack-tox-py36/bd813ae/ : SUCCESS in 4m 53s\n- openstack-tox-py37 https://logs.opendev.org/05/674205/1/check/openstack-tox-py37/55e4f0a/ : SUCCESS in 4m 30s\n- openstack-tox-docs https://logs.opendev.org/05/674205/1/check/openstack-tox-docs/4d6d5fd/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://logs.opendev.org/05/674205/1/check/build-openstack-releasenotes/db4bfdf/html/ : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cb0845f5a3e4d8cefd6cc4d1ad37c0958b749678","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-02 10:03:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.161.150/logs/05/674205/1/upstream-check/cinder-cDOT-iSCSI/49a29f3/ : SUCCESS in 54m 40s\n- cinder-cDOT-NFS http://13.56.161.150/logs/05/674205/1/upstream-check/cinder-cDOT-NFS/f19891e/ : SUCCESS in 50m 26s\n- cinder-cDOT-FCP http://13.56.161.150/logs/05/674205/1/upstream-check/cinder-cDOT-FCP/e4e2493/ : SUCCESS in 1h 06m 33s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1b91c2eceded18c0f72b8621e86051d61531e87","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-02 10:07:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.161.150/logs/05/674205/1/upstream-solidfire-check/cinder-solidfire-iSCSI/b0c83ce/ : SUCCESS in 1h 10m 32s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"a7fc021ea778911a8ae9fa09ab0945622721cc4c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-02 10:24:33.000000000","message":"Patch Set 1:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-674205-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"91a604494b8931c937a2c41ebf5d9dff009db11c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-02 11:07:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/05/674205/1/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/23536a3 : SUCCESS in 2h 03m 57s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/05/674205/1/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/14d8a31 : SUCCESS in 2h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fb75d8523ba68b82190a10996ca91175ffac721d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-02 11:20:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_VNX_FC_os-brick/46fbc06/DellEMC_VNX_FC_os-brick/25 : SUCCESS in 1h 14m 09s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_VNX_iSCSI_os-brick/24578e5/DellEMC_VNX_iSCSI_os-brick/25 : FAILURE in 1h 08m 46s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"72155d4f2e7712b53f68d943a5acd1a50adcfbdb","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-02 11:30:22.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/05/674205/1/check/kaminario-dsvm-tempest-full-iscsi/6c15b04 : SUCCESS in 1h 45m 22s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/05/674205/1/check/kaminario-dsvm-tempest-full-FC/c9ea871 : FAILURE in 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4088e1eb0bf036019570a31d45458f6850027a1a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-02 14:19:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_Unity_FC_os-brick/ddcc8e9/DellEMC_Unity_FC_os-brick/27 : SUCCESS in 1h 29m 27s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_Unity_iSCSI_os-brick/33de0f8/DellEMC_Unity_iSCSI_os-brick/26 : SUCCESS in 1h 33m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a674202419cf634e3a193894d1c562042aefd6b1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-02 17:16:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_SC_FC-os-brick/7435052/DellEMC_SC_FC-os-brick/19 : FAILURE in 1h 21m 37s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_SC_iSCSI-os-brick/2e54c44/DellEMC_SC_iSCSI-os-brick/19 : FAILURE in 2h 52m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"7442a30d0f18e579bb654fd69ba99e461e267bd1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-03 06:35:07.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_XtremIO_FC_os-brick/071af61/DellEMC_XtremIO_FC_os-brick/50 : SUCCESS in 54m 37s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/05/674205/1/check/DellEMC_XtremIO_iSCSI_os-brick/30da976/DellEMC_XtremIO_iSCSI_os-brick/48 : FAILURE in 2h 55m 31s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"557fd054d64495df94e900535c6a0c631cdacb6a","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-08-14 15:52:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"79d2e5bb8503356056f056712d986df4a93ac427","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-08-14 16:52:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1899e624c97fdba5da52f5b7993f2fd7eca91d31","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-08-21 09:11:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e260587ae12467002fd191668de851d92b21f03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 09:11:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c5218784c87fd6b7ea6089d8cc6ef37a372bb1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 10:44:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f5c0ca3ed0d4d70b3f573ccd812665e : SUCCESS in 3m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/08e6e7c2a6454e638c355a39d7161d11 : FAILURE in 1h 23m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13a3dddc2ebb4825a6c084b63d8c1e61 : SUCCESS in 1h 26m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e04336e421024369ad381dfa2ac657dc : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd8780d509e44a1e86b50e201f1d4a08 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb2d8cb7edad404da9685dc51ffeabd8 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a515a5345be94760ac49865f27ef8e52 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3962f12b516487a8f061bd3d27eb042 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/778be68405844099b7e08730657eeba2 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd161bc5a9ee44ae89893ad8a9147355 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c3efdc2b8c4c65e9f565bc42e4922aebe21c4552","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-21 20:51:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de95d3a8e7c5d69e7ad4187d261c4944f30d360d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-21 21:43:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/674205/1/check/dsvm-tempest-kvm/9b27975 : FAILURE in 39m 04s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/674205/1/check/dsvm-tempest-exe/7902a4a : FAILURE in 40m 18s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"3ed468227fb3927b7e186a077eb27f6df12a94b8","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-08-21 21:44:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/674205/1/iscsi/ : SUCCESS in 50m 42s\n- os-brick-smb http://cloudbase-ci.com/os-brick/674205/1/smb/ : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":1},{"id":"77d0498280225eaca7a783a4878a7c9384313f10","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-21 22:42:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/05/674205/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/c8f61c0 : FAILURE in 1h 50m 07s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/05/674205/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4397609 : SUCCESS in 1h 09m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"334f296d8a8db4d9f540d9ee6c89f42b98389252","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 23:19:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0eccd7c895b548008f7978b8fd177461 : SUCCESS in 3m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0e89dd416ef4c35a718776630744e5b : SUCCESS in 1h 48m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1ca1fa4a3d94d0e8af4fc9a481fb8ca : SUCCESS in 1h 27m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4a474bcccfd4d1fa1e947be913ee006 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee32a9f88e04f5092787d2a42bdbbc0 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ac4d341309745e2a1d1bcd14fc8b812 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ced3817fca84f3ab7767901293e5945 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7c8de32042f4b0988dc46d22f98fa55 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c67924cd5224fa0bd648125a2518e97 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71b59d41bc484ff9b9136de41dea756b : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e1acc522c3b200148bfb36fcf625a3dcced171c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:21:26.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c5fc2b24a42e2cce2437536603638ef4bedaf41d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-22 01:20:38.000000000","message":"Patch Set 1:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-674205-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"1d93fb1c3712bdad913f491d6dc718ab55b12734","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-22 05:40:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02bf967238c028d39eeef2fa94e22098057b9683","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-08-22 06:23:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/674205/2/check/dsvm-tempest-kvm/a886e88 : FAILURE in 35m 31s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/674205/2/check/dsvm-tempest-exe/15dc596 : FAILURE in 35m 04s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"1d2c632ad83b7b991a498ea812a4f59d1fb98239","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-08-22 06:29:28.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/674205/2/iscsi/ : FAILURE in 47m 52s\n- os-brick-smb http://cloudbase-ci.com/os-brick/674205/2/smb/ : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":2},{"id":"94f20618982f706ec28c8dbd1e5a870df4a30a19","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-22 07:01:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_XtremIO_FC_os-brick/876edf3/DellEMC_XtremIO_FC_os-brick/61 : SUCCESS in 51m 31s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_XtremIO_iSCSI_os-brick/736e664/DellEMC_XtremIO_iSCSI_os-brick/57 : FAILURE in 1h 20m 53s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"2eb42e7f308778fef9f7494585fd9c90f05164df","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-22 07:03:48.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/05/674205/2/check-cinder/Cinder-ISER-ISCSI/f317b0f : FAILURE in 1h 01m 56s\n- Cinder-ISER-LIO http://13.74.249.42/05/674205/2/check-cinder/Cinder-ISER-LIO/7bda7b4 : SUCCESS in 1h 04m 33s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"58cfc3bab7f2bec22b6bc18c0ff518660bb96bd7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-22 07:09:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_Unity_FC_os-brick/204c658/DellEMC_Unity_FC_os-brick/36 : SUCCESS in 1h 10m 48s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_Unity_iSCSI_os-brick/83d2d52/DellEMC_Unity_iSCSI_os-brick/35 : SUCCESS in 1h 29m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5a0f7427e0a87c25956f5338f117be23c4e972e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:24:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3046934859294a868c8d170706952051 : SUCCESS in 3m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f3f6e39a13794454a4fa59a6a7e4e5f0 : SUCCESS in 1h 42m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70eb186f5d7c497c864feb0b7040212b : SUCCESS in 1h 28m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/282c0d86154c44069fdc18e4d7302eeb : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ad5339ced2433883eb19fa37661d28 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63b9d673ddd04fb3b817e8d6d425ffaa : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c177b9f134ab4f7bbaf93a04b433585b : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afcb9e4815ef4d2893ea85d579ef8960 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0377ace2246a4908aa43a4fe15b17542 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5b45eb862644a2999b633c578ad45aa : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b7118433f4a7f2598a3e6880ad4f73fa872d1712","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-22 07:30:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/05/674205/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/9c4648b : FAILURE in 1h 50m 16s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/05/674205/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/a7e5ad6 : SUCCESS in 1h 03m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"358ae45b2eb478e8bbddd49d67ebdefde1894341","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-08-22 08:00:27.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/05/674205/2/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/115a674 : FAILURE in 33m 56s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/05/674205/2/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/7e09629 : FAILURE in 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2e86edae88c9a0c4a3c0477a9d927722b1af0897","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-08-22 09:06:59.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_SC_FC-os-brick/ea2e4eb/DellEMC_SC_FC-os-brick/25 : SUCCESS in 1h 11m 23s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_SC_iSCSI-os-brick/0fdbf6f/DellEMC_SC_iSCSI-os-brick/25 : SUCCESS in 1h 42m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"4ce819138e2273ed6ff8188d085200f4757fdf82","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-22 09:07:12.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/05/674205/2/check/kaminario-dsvm-tempest-full-iscsi/f588ac5 : SUCCESS in 1h 42m 50s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/05/674205/2/check/kaminario-dsvm-tempest-full-FC/15c94ad : SUCCESS in 1h 30m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a95b75d0662e71cfaffe53469ae606c5753a1aff","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-22 09:08:47.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-674205-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"447b0b2ec30432248bd8476abc51bc89b82bd8b2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-08-22 09:21:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_VNX_FC_os-brick/c882711/DellEMC_VNX_FC_os-brick/31 : SUCCESS in 59m 35s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_VNX_iSCSI_os-brick/2ec106d/DellEMC_VNX_iSCSI_os-brick/31 : SUCCESS in 1h 57m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"481ed11ad9145606d34976a008f62a5d5a07818a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-08-22 09:57:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_Unity_FC_os-brick/4d01e47/DellEMC_Unity_FC_os-brick/37 : SUCCESS in 1h 09m 38s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_Unity_iSCSI_os-brick/6fe35d9/DellEMC_Unity_iSCSI_os-brick/36 : SUCCESS in 1h 24m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fdca125c413fcc163f1e8538f9161d4c49600e97","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-08-22 10:24:37.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_XtremIO_FC_os-brick/10d05ad/DellEMC_XtremIO_FC_os-brick/62 : FAILURE in 1h 10m 09s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_XtremIO_iSCSI_os-brick/2cd5e40/DellEMC_XtremIO_iSCSI_os-brick/58 : FAILURE in 1h 45m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"3f4e604f494c6200fac27d3d9e6792160cba741d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-22 17:57:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/05/674205/2/check/DellEMC_VxFlexOS_os-brick/60e05a6/DellEMC_VxFlexOS_os-brick/28 : SUCCESS in 1h 28m 49s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"701757bed1f58e1e82372ed07e81dbaa7f89b743","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-08-22 22:14:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4a240180ce864579f53375a01fd2bfb50c609e6a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-23 12:03:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/05/674205/2/check/tempest-dsvm-full-xenial/2cb480a/ : SUCCESS in 2h 02m 15s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/05/674205/2/check/devstack-plugin-ceph-dsvm-xenial/43980f9/ : SUCCESS in 1h 50m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a260b2de5fe5f3527a1a7e7354996cabe596bad3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-28 16:12:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae7360a8b39bcf03a05c1cf666f5544671763ea6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 16:12:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a71fe3a85554748ab2bcdf59cdc0a72985f5b9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:19:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d96780d065c4175bfa917fdeebe2b63 : SUCCESS in 5m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/819681d386684a07894d2bac4111caf9 : SUCCESS in 1h 59m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6efe55ae68394bee900b156c332b2490 : SUCCESS in 1h 27m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19e38172a17242f7851c048a108dc8bb : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc322b7381f49a2bf77d00e2a02caad : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a07866456fb04d9ca6fd1baf6483b3aa : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ace377d7e7cf472f9698ca07f38019fd : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a9454fbc9a2423eac43e74dc291d57a : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae93142d40d423495fa2cae940ce119 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c572bcb8698345cc8c0a4f788bf645b9 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a6195d4f8b61f6d36bc2f83c9b4a2330029b1c88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:19:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"87c026fe4524fafe52ead8c766ee62890b946954","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:21:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47aed3103f9f4bb1aa789b3e03ff27eb : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2},{"id":"aa2c771156170c2634a0bbaed1614c89ca436f60","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 19:39:21.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/05/674205/2/check/check-dsvm-tempest-vz7-exe-minimal/d8d25a3 : EXCEPTION\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99c579c51ad477635fb750f747e13dbde0bee863","revisions":{"73e9c232757f4e400c4c9fca7e04407dbba727a6":{"kind":"REWORK","_number":1,"created":"2019-08-02 06:57:42.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/05/674205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/05/674205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/674205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/674205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/674205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/05/674205/1"}}},"commit":{"parents":[{"commit":"066ba564cceee9d17043b1dfea1699045329a020","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/066ba564cceee9d17043b1dfea1699045329a020"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:57:06.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:57:59.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: Ie1952c7b65542ac493e506c8916da0e1a856e838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73e9c232757f4e400c4c9fca7e04407dbba727a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73e9c232757f4e400c4c9fca7e04407dbba727a6"}]},"branch":"refs/heads/master"},"99c579c51ad477635fb750f747e13dbde0bee863":{"kind":"REWORK","_number":2,"created":"2019-08-22 05:40:10.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/05/674205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/05/674205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/674205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/674205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/674205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/05/674205/2"}}},"commit":{"parents":[{"commit":"68faaeb665e6842caab6414c7f9617d5bb341e20","subject":"Merge \"Add generate_connector_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/68faaeb665e6842caab6414c7f9617d5bb341e20"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:57:06.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-22 05:40:22.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: Ie1952c7b65542ac493e506c8916da0e1a856e838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/99c579c51ad477635fb750f747e13dbde0bee863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/99c579c51ad477635fb750f747e13dbde0bee863"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
