)]}'
{"id":"openstack%2Fos-brick~676433","triplet_id":"openstack%2Fos-brick~stable%2Fstein~I9d3a9dac13dcd585330e7b891c61b2626e5aabec","project":"openstack/os-brick","branch":"stable/stein","hashtags":[],"change_id":"I9d3a9dac13dcd585330e7b891c61b2626e5aabec","subject":"Check path alive before get scsi wwn","status":"MERGED","created":"2019-08-14 14:57:35.000000000","updated":"2019-09-03 21:06:52.000000000","submitted":"2019-09-03 21:04:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676433-1567544699555-26e757ed","meta_rev_id":"3ab0e0d29c56e3cc54c1ec48cae3440583b647d5","_number":676433,"virtual_id_number":676433,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-15 20:31:09.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-15 23:49:09.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"date":"2019-08-15 22:43:01.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-03 21:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-15 22:38:36.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2019-08-15 20:29:38.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-03 19:19:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":2,"date":"2019-09-03 19:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":1,"date":"2019-09-03 19:16:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-03 19:26:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 14:57:35.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-08-15 02:39:57.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 20:29:38.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-08-15 20:31:09.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 22:38:36.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-08-15 22:43:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 23:49:09.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-09-03 19:16:49.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-09-03 19:19:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-03 19:22:34.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-09-03 19:26:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-09-03 21:04:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b174d8eefb498a5fc8985084ed8af520e554b87f","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2019-08-14 14:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1b9def742b029bd4764b05ff872bffc52462ee8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 15:25:08.000000000","message":"Patch Set 1: Code-Review-1\n\nAll of these backports are missing the cherry-pick information in the commit message. Make sure to use \"git cherry-pick -x\" when backporting or the gerrit UI.\n\nAlso, really shouldn\u0027t backport to older branches until merged in the newer ones.","accounts_in_message":[],"_revision_number":1},{"id":"07cb7f99465f231d2ee29f6c499e770307cfbd0f","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-08-14 15:51:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/676433/1/iscsi/ : SUCCESS in 52m 28s\n- os-brick-smb http://cloudbase-ci.com/os-brick/676433/1/smb/ : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b391537cd51358482cc2dad7cb6fa45f50bc62ab","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-14 16:23:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/33/676433/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/0f3c097 : SUCCESS in 1h 19m 15s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/33/676433/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/77712df : SUCCESS in 1h 25m 11s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d87f04fc83894c1ecbdc879a4bd1b2de6e23160","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 16:49:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c8565e934364ceea042c501a61e501a : SUCCESS in 1h 48m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c11f0d06cdc40a3be01d1a989813c0c : SUCCESS in 1h 37m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a204dad0b7c949e181d9ecb0853ca8a4 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea1df6f5e1364ba89abf68e67ff06c2b : FAILURE in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b364db2f85604fd78ea3d89b8b793487 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a0f28c054b54c12bb0371a687b74783 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/510d36ad451a44c3a87daf77d3807ef4 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e048150d7ee5483ebab61fef74edb15a : SUCCESS in 3m 14s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/935bf425c38b4962818bcecceab74ac5 : SUCCESS in 1h 24m 36s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/969f4ada8b16460d9c078a81e220e544 : SUCCESS in 1h 30m 31s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b75b6873f1d14d1cad18fbb880d6c735 : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":1},{"id":"988455bc7f0177f8853d4728ea8334bb3734634b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-15 02:39:57.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/33/676433/1/check-cinder/Cinder-ISER-ISCSI/32232cd : SUCCESS in 1h 09m 21s\n- Cinder-ISER-LIO http://13.74.249.42/33/676433/1/check-cinder/Cinder-ISER-LIO/209c73a : FAILURE in 1h 11m 36s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"7b9b66b195a1140b0f00dddc0bed297f32793bb5","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2019-08-15 19:38:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30c3a234ff6c2d92a3a66d9f726f6f78842ccb6c","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-08-15 20:29:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/676433/2/iscsi/ : SUCCESS in 50m 30s\n- os-brick-smb http://cloudbase-ci.com/os-brick/676433/2/smb/ : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1081c6d014f6078fe4b16a521d21c5d3dc7db268","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-15 20:31:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/33/676433/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/6bbf84e : SUCCESS in 52m 48s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/33/676433/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/7fe1453 : FAILURE in 7m 10s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"126894d43a90338667d1baab997d364512c5a824","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 21:35:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b36cb3fc5eb54e70865bd0560e8f7e99 : SUCCESS in 1h 55m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e3ece27a99b419d8426b7a944076780 : SUCCESS in 1h 47m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65d7551ee4fe48e2ae4b9722b1a3ed2e : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b123f7a07e61477c8398dfd662ee3e4f : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8afee08a416344b385ce4af2ede126b3 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c95c1c36bc949a0bceefd3ddeac8216 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57319c0aff0344b6a5425eb250fed6fb : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b016a36fbf4fa5bf337ea408bbc00f : SUCCESS in 3m 59s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/01620fbf509445b9b3f0d2bdb2a2be15 : SUCCESS in 1h 30m 47s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/05113d2349b14e7cb0f8289f37a68e08 : SUCCESS in 1h 33m 13s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/365f971bb6bd4c5fae65421df2858a06 : SUCCESS in 1h 40m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8003baa052107af213c510c66b8f5a0d1582fc32","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-08-15 22:38:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://13.56.159.105/logs/33/676433/2/upstream-solidfire-check/cinder-solidfire-iSCSI/482b701/ : SUCCESS in 1h 02m 49s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"14f72a310a8a6ea2db58f4a1c9afd0fd74a5fcae","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-15 22:43:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://13.56.159.105/logs/33/676433/2/upstream-check/cinder-cDOT-iSCSI/deba0c0/ : SUCCESS in 48m 09s\n- cinder-cDOT-NFS http://13.56.159.105/logs/33/676433/2/upstream-check/cinder-cDOT-NFS/38d9bb1/ : SUCCESS in 52m 51s\n- cinder-cDOT-FCP http://13.56.159.105/logs/33/676433/2/upstream-check/cinder-cDOT-FCP/6eb4367/ : SUCCESS in 1h 07m 29s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"0d2b78c3154ccfb9643fa8e7b659bbfbdde71b69","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-08-15 23:49:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/33/676433/2/check/kaminario-dsvm-tempest-full-iscsi/164ed81 : SUCCESS in 2h 13m 34s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/33/676433/2/check/kaminario-dsvm-tempest-full-FC/7c0fae3 : SUCCESS in 1h 54m 26s","accounts_in_message":[],"_revision_number":2},{"id":"690c8ca2269fedf06019d67ebcd0a4d45103d42f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-09-03 19:16:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c4586ed2d42c0bc24efb94ae3a238b290fe64ee9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-03 19:19:18.000000000","message":"Patch Set 2: Code-Review+2\n\nCherry pick looks clean and good to backport.","accounts_in_message":[],"_revision_number":2},{"id":"7c7cffb7e764e15ba93b67a4039ece5120a4091b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-03 19:22:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0267af7f606049fd8686458cbea06888c45d0e3e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-03 19:26:45.000000000","message":"Patch Set 2: Workflow+1\n\nLooks like an appropriate backport.","accounts_in_message":[],"_revision_number":2},{"id":"47b5c76ac6ca531963c2e5317a16d576a0601efc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:26:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"763e41359d3873757f18fb8371d34117df428886","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:04:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ae6244b6717947eaa851e4ff1aa90e7f : SUCCESS in 1h 31m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68cf8ea03c594e888cc705b220fc6ae4 : SUCCESS in 1h 28m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a4a8b19c4fb40e6981d38e53c305431 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf9d671bd524002b34f054811b66357 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9af24d628db3478ab1960e036d768074 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e73d2e7e429430fb382152843e68c33 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36d520f4fdaa407cb915333625196571 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eec82cad8054569948060c4395b468d : SUCCESS in 4m 58s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/45792fd08b254b86b55e15a7718f2faa : SUCCESS in 1h 16m 45s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/845650e4c98442d0aae0fa8d3347c505 : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":2},{"id":"eb5ae73ebda9647ffb39f0d25f0b3653a7cb48a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:04:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3ab0e0d29c56e3cc54c1ec48cae3440583b647d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:06:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac95f9ab7104251b7add2efb56b9057 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a994dec147d1629084566ff6014f5dba1a60bfc","revisions":{"3cf26dbf6d6aa37cd5d6842f8ba2039f7d2a4112":{"kind":"REWORK","_number":1,"created":"2019-08-14 14:57:35.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/33/676433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/33/676433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/676433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/676433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/676433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/33/676433/1"}}},"commit":{"parents":[{"commit":"4c2b25391091b5ad4935e552f2be53c0cf3ee738","subject":"luks: Explicitly use the luks1 type to ensure LUKS v1 is used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4c2b25391091b5ad4935e552f2be53c0cf3ee738"}]}],"author":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-06-05 05:47:16.000000000","tz":480},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2019-08-14 14:57:37.000000000","tz":-240},"subject":"Check path alive before get scsi wwn","message":"Check path alive before get scsi wwn\n\nIn FC environment, when disconnect volume,\nif the first path in the loop is failed,\nit will lead get scsi wwn to failure.\nSo I think we need to add path validity check\nbefore get scsi wwn\n\nChange-Id: I9d3a9dac13dcd585330e7b891c61b2626e5aabec\nCloses-Bug: 1831621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3cf26dbf6d6aa37cd5d6842f8ba2039f7d2a4112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3cf26dbf6d6aa37cd5d6842f8ba2039f7d2a4112"}]},"branch":"refs/heads/stable/stein"},"7a994dec147d1629084566ff6014f5dba1a60bfc":{"kind":"REWORK","_number":2,"created":"2019-08-15 19:38:06.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/33/676433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/33/676433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/676433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/676433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/676433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/33/676433/2"}}},"commit":{"parents":[{"commit":"4c2b25391091b5ad4935e552f2be53c0cf3ee738","subject":"luks: Explicitly use the luks1 type to ensure LUKS v1 is used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4c2b25391091b5ad4935e552f2be53c0cf3ee738"}]}],"author":{"name":"zengjia","email":"zengjia@awcloud.com","date":"2019-06-05 05:47:16.000000000","tz":480},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2019-08-15 19:38:11.000000000","tz":-240},"subject":"Check path alive before get scsi wwn","message":"Check path alive before get scsi wwn\n\nIn FC environment, when disconnect volume,\nif the first path in the loop is failed,\nit will lead get scsi wwn to failure.\nSo I think we need to add path validity check\nbefore get scsi wwn\n\nChange-Id: I9d3a9dac13dcd585330e7b891c61b2626e5aabec\nCloses-Bug: 1831621\n(cherry picked from commit 3b8d5b7886b9a45df6b666ef594f4a337914856d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7a994dec147d1629084566ff6014f5dba1a60bfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7a994dec147d1629084566ff6014f5dba1a60bfc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
