)]}'
{"id":"openstack%2Fos-brick~684263","triplet_id":"openstack%2Fos-brick~master~I54a91dfcfae774b1051662291afe1586e59901ac","project":"openstack/os-brick","branch":"master","topic":"nvmeof","hashtags":[],"change_id":"I54a91dfcfae774b1051662291afe1586e59901ac","subject":"nvmeof: Fix broken UTs","status":"MERGED","created":"2019-09-24 09:58:07.000000000","updated":"2019-10-28 22:10:21.000000000","submitted":"2019-10-28 21:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684263-1572299356359-89326a6a","meta_rev_id":"ec2aef95ad6ff3019d9411a0ba8be8c94f7eb99d","_number":684263,"virtual_id_number":684263,"owner":{"_account_id":31019,"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","username":"pszczerx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-28 21:21:42.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-28 21:36:42.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-28 20:37:14.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 21:49:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-28 20:26:26.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2019-10-28 20:56:31.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-28 19:11:59.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-28 21:43:57.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"date":"2019-10-28 22:10:21.000000000","post_submit":true,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2019-10-28 21:21:41.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-28 21:44:49.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2019-10-28 21:37:45.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-10-28 18:44:33.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-28 12:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-28 12:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-28 12:57:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-28 13:19:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 10:30:25.000000000","updated_by":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"reviewer":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"state":"REVIEWER"},{"updated":"2019-09-24 10:47:07.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-24 11:58:46.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-24 12:18:56.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 18:08:35.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-10-17 05:50:54.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 13:19:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-28 18:44:33.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-10-28 19:11:59.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 20:26:26.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-10-28 20:37:14.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 20:56:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 21:21:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 21:21:42.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 21:36:42.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 21:37:45.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-10-28 21:43:57.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 21:44:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-10-28 21:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 22:10:21.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"}],"messages":[{"id":"a4124fdb7d64544291b1a1f1bf97d2d70f609a55","author":{"_account_id":31019,"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","username":"pszczerx"},"date":"2019-09-24 09:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c48fea709154be5022dcf8f8fad9da5e4c03a96","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2019-09-24 10:30:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9aaa78bc6de73073dd51463fa1adc00ec18f7a74","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-24 10:47:07.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/63/684263/1/check/check-dsvm-tempest-vz7-exe-minimal/1bb3d50 : FAILURE in 11m 32s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"e504ff32a7d84326df9f5b73dc60e163c4258e1a","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-09-24 11:41:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/684263/1/iscsi/ : SUCCESS in 48m 46s\n- os-brick-smb http://cloudbase-ci.com/os-brick/684263/1/smb/ : SUCCESS in 51m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9511b5dd80e0ed8775147976d53befacdff669b5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-24 11:58:46.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/63/684263/1/check/tempest-dsvm-full-xenial/34d2854/ : FAILURE in 1h 44m 59s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/63/684263/1/check/devstack-plugin-ceph-dsvm-xenial/009ac20/ : SUCCESS in 1h 58m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d54c9ed0002bdbf8adffc71c85587af549f93ed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 12:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/dfa88dac47574527a0971a6fe9911fd2 : SUCCESS in 1h 47m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8f09517f3dc4282b667e4f07802f579 : SUCCESS in 1h 55m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2eb8d7a12e14871bdd19fe150f8c237 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e3d4ab6fe24f589aba0e1ec0542546 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7b25376b4c04ad183150831b317a5ea : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4be69fa8ada140ee98a1dc7fc1cb59b7 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0263225f780c4233acac64f2cb74058f : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668d8e68dd57445abdfb9ba9dc768a33 : SUCCESS in 8m 06s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/40c9324da4384431939b3ee5fdcba2ad : SUCCESS in 1h 45m 53s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/878d1ee0c56a4e5688ba6c2575b6df0c : SUCCESS in 1h 18m 08s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/585daa96735243dd8bced142022b1468 : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0245cb65081381c8a76dabc66e200258be9def70","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2019-09-24 12:18:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/63/684263/1/check/dsvm-tempest-kvm/a68c001 : FAILURE in 40m 29s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/63/684263/1/check/dsvm-tempest-exe/bb4d985 : FAILURE in 50m 39s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"6e1f404eed15cbb3d9e992c409aedf25602af670","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-24 12:38:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/63/684263/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/ecf7d80 : SUCCESS in 52m 14s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/63/684263/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/6626bde : SUCCESS in 1h 14m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"760619f62c39d6d1b7c8c03e9e71de9709a50cbc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-09-24 12:45:11.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/63/684263/1/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/462e964 : FAILURE in 31m 30s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/63/684263/1/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/6e6e4a8 : FAILURE in 40m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0ee109f97d7102ebbef74932f87a84b4cc54085b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-09-24 12:52:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/63/684263/1/check/DellEMC_SC_FC-os-brick/8d161f7/DellEMC_SC_FC-os-brick/46 : FAILURE in 25m 38s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/63/684263/1/check/DellEMC_SC_iSCSI-os-brick/2debcca/DellEMC_SC_iSCSI-os-brick/45 : FAILURE in 16m 31s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"43b770695264ac0654549d56d3dc8f2abd63dd1d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-24 12:59:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/63/684263/1/check/DellEMC_VxFlexOS_os-brick/645bde6/DellEMC_VxFlexOS_os-brick/50 : FAILURE in 28m 40s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"a8e7b7769089bd963e281a5fbace3aa3abde0ea2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-24 13:07:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://54.183.250.52/logs/63/684263/1/upstream-check/cinder-cDOT-iSCSI/210694e/ : SUCCESS in 1h 00m 02s\n- cinder-cDOT-NFS http://54.183.250.52/logs/63/684263/1/upstream-check/cinder-cDOT-NFS/2af7e58/ : SUCCESS in 58m 44s\n- cinder-cDOT-FCP http://54.183.250.52/logs/63/684263/1/upstream-check/cinder-cDOT-FCP/fa1e38b/ : SUCCESS in 1h 04m 36s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"5c5e79ede95c115b1edfba8c3d77705ca021b227","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-09-24 13:07:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://54.183.250.52/logs/63/684263/1/upstream-solidfire-check/cinder-solidfire-iSCSI/b42219a/ : SUCCESS in 1h 04m 37s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"3d248b31100c01afdbd2f8c6c9b4fd54b380ee28","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-09-24 13:28:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/63/684263/1/check/DellEMC_XtremIO_FC_os-brick/6d58ec1/DellEMC_XtremIO_FC_os-brick/93 : SUCCESS in 51m 31s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/63/684263/1/check/DellEMC_XtremIO_iSCSI_os-brick/740954d/DellEMC_XtremIO_iSCSI_os-brick/88 : FAILURE in 1h 14m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"b26bed30ca6d4ff22b0803eec7cead5fe0301e97","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-24 13:34:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/63/684263/1/check-cinder/Cinder-ISER-ISCSI/b84081c : SUCCESS in 1h 37m 43s\n- Cinder-ISER-LIO http://13.74.249.42/63/684263/1/check-cinder/Cinder-ISER-LIO/c84a17d : SUCCESS in 1h 07m 52s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"23f06416fbb5c2c76fda59c79587a6e9105040b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-24 14:02:31.000000000","message":"Patch Set 1:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-684263-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"ae978bf767dca472ce9bd9ff005ed9f850a9d55d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-09-24 15:15:05.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/63/684263/1/check/DellEMC_VNX_FC_os-brick/7575cee/DellEMC_VNX_FC_os-brick/52 : FAILURE in 24m 36s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/63/684263/1/check/DellEMC_VNX_iSCSI_os-brick/fcbe781/DellEMC_VNX_iSCSI_os-brick/50 : SUCCESS in 1h 36m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"4352298df2d2653d82e6d607a6e49d247ea03725","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-09-24 17:04:53.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/63/684263/1/check/kaminario-dsvm-tempest-full-iscsi/6513286 : FAILURE in 3h 04m 34s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/63/684263/1/check/kaminario-dsvm-tempest-full-FC/cf91299 : SUCCESS in 1h 41m 26s","accounts_in_message":[],"_revision_number":1},{"id":"bddf3ee7dfd1d87d1e1edba977e02741563f8331","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 17:01:30.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good. Thanks for improving out testing.","accounts_in_message":[],"_revision_number":1},{"id":"6c3bc52184a47d6ae934a7fbe22a767ac7a13740","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-10-09 18:08:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fc0ade12f004a2bd2c2e3ba94eb978d1baf8f11","author":{"_account_id":31019,"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","username":"pszczerx"},"date":"2019-10-17 05:43:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"94f0099cc250ee009e0f1ace6380691a7f37a58a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-17 05:50:54.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/63/684263/2/check-cinder/Cinder-ISER-ISCSI/8f1890f : FAILURE in 7m 26s\n- Cinder-ISER-LIO http://13.74.249.42/63/684263/2/check-cinder/Cinder-ISER-LIO/d2e7b59 : FAILURE in 0s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"9ab4dc24429c6708fd4c7d481df8a04335390ae2","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-10-17 06:35:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/684263/2/iscsi/ : SUCCESS in 52m 28s\n- os-brick-smb http://cloudbase-ci.com/os-brick/684263/2/smb/ : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bae7139bfdbc80911d3baddf1fe0c62ebd744ec2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-17 06:50:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/63/684263/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/5480aa0 : SUCCESS in 57m 56s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/63/684263/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/3480c60 : SUCCESS in 1h 07m 09s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"b779fdc2ef9ea95981c73d8fa530e73bb10f061b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 08:05:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/36faa08343dc4bde8e6fccdff5adf25e : SUCCESS in 1h 40m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a71fc6adb9a4c3fa26a0598347bcfbd : SUCCESS in 1h 09m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e037dc517a9a493e9de50a5d3292072f : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3dba17c8dee41ed97de779ee2d9e059 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cd8f64ca0b7400dab0c5f6b3186cb79 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a80270d1fc3646bd84ecaa0308294d44 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0424b40ca81a48cc8ceb7e68c3c8cc71 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1196802ae184413a22ce59a9d5efd5e : SUCCESS in 7m 21s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/91bbc41ab4d54d25afd0f2760db21cf8 : SUCCESS in 1h 25m 49s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/a942c466d84841bdae0d48489e609ccc : SUCCESS in 2h 05m 25s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3d665f1fe6f941cb96b610c03b39baab : SUCCESS in 2h 20m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a4b1f62e972afecd1fbce5bcc116a5cb80321776","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-10-17 09:10:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/63/684263/2/upstream-solidfire-check/cinder-solidfire-iSCSI/23e1684/ : SUCCESS in 1h 04m 39s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"198a57f172b6e5bd8bd322701caab006edbe3bf4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-17 09:31:41.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-684263-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"62d122f79e6cb907a99f9aedf53461420b6d84d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-10-17 10:44:11.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/63/684263/2/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/ac31cac : FAILURE in 31m 58s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/63/684263/2/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/a69c6a6 : FAILURE in 31m 33s","accounts_in_message":[],"_revision_number":2},{"id":"681d8eb6465048d48f1c1408122a4c95f94f9be4","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-10-17 10:56:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/63/684263/2/check/DellEMC_SC_FC-os-brick/599ecf4/DellEMC_SC_FC-os-brick/51 : FAILURE in 26m 38s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/63/684263/2/check/DellEMC_SC_iSCSI-os-brick/63fdc74/DellEMC_SC_iSCSI-os-brick/50 : FAILURE in 26m 16s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"e04196b171b98f2608fd11228b0e7f31cfa521ec","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-17 11:21:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/63/684263/2/upstream-check/cinder-cDOT-iSCSI/6b619a2/ : SUCCESS in 3h 15m 32s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/63/684263/2/upstream-check/cinder-cDOT-NFS/e1a4567/ : FAILURE in 1h 21m 54s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/63/684263/2/upstream-check/cinder-cDOT-FCP/ac55b80/ : SUCCESS in 1h 20m 31s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d60a139577215778408ee774ba0d059e74719e9e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-10-17 12:37:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/63/684263/2/check/DellEMC_XtremIO_FC_os-brick/da33e87/DellEMC_XtremIO_FC_os-brick/103 : SUCCESS in 46m 59s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/63/684263/2/check/DellEMC_XtremIO_iSCSI_os-brick/8150728/DellEMC_XtremIO_iSCSI_os-brick/98 : FAILURE in 1h 18m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"e08a20758d8785af872d375f2fc135e961a9c097","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-10-17 13:45:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/63/684263/2/check/DellEMC_VNX_FC_os-brick/f2dabb3/DellEMC_VNX_FC_os-brick/58 : FAILURE in 24m 12s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/63/684263/2/check/DellEMC_VNX_iSCSI_os-brick/fde0bd8/DellEMC_VNX_iSCSI_os-brick/56 : FAILURE in 1h 38m 24s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"9f2d9c97d7f62490ad9d9791ae69e6b3d590ebfe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-17 14:26:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/63/684263/2/check/DellEMC_VxFlexOS_os-brick/03ac037/DellEMC_VxFlexOS_os-brick/52 : FAILURE in 31m 24s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"f69e67f163133ba14c0775c79117d1a205e35697","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-10-17 16:07:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/63/684263/2/check/kaminario-dsvm-tempest-full-iscsi/ae3f93c : SUCCESS in 2h 11m 13s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/63/684263/2/check/kaminario-dsvm-tempest-full-FC/30351c0 : SUCCESS in 1h 41m 07s","accounts_in_message":[],"_revision_number":2},{"id":"66d7583ab2bf4da4fbd1cb5bdbc71660ed2bb800","author":{"_account_id":31019,"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","username":"pszczerx"},"date":"2019-10-28 12:57:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"90110319c3f9bee625e2ee2d017c600c0d8ec5d3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 13:19:28.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"40ee7a105195f27b3d781d4f40c224f502e12649","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-10-28 13:50:58.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/684263/3/iscsi/ : FAILURE in 53m 10s\n- os-brick-smb http://cloudbase-ci.com/os-brick/684263/3/smb/ : FAILURE in 46m 52s","accounts_in_message":[],"_revision_number":3},{"id":"ee6ea310c905f18e3d78bb31025183dc2639e21f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-10-28 14:09:15.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_XtremIO_FC_os-brick/9f14661/DellEMC_XtremIO_FC_os-brick/18 : SUCCESS in 50m 52s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_XtremIO_iSCSI_os-brick/2026078/DellEMC_XtremIO_iSCSI_os-brick/18 : FAILURE in 1h 11m 33s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"0238cc25a744b62edc21e7bbbf4a32e3e55a8a5f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-28 14:13:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/63/684263/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/4e046a9 : SUCCESS in 1h 00m 18s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/63/684263/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/5f2a91f : SUCCESS in 1h 15m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"529d3788b27da898e9808c04c02d6e967ff58deb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-10-28 14:21:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_Unity_FC_os-brick/905dd09/DellEMC_Unity_FC_os-brick/17 : SUCCESS in 1h 05m 12s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_Unity_iSCSI_os-brick/920cbdf/DellEMC_Unity_iSCSI_os-brick/17 : SUCCESS in 1h 24m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"75918e79edfbdbfb773e4fb2aa86c495d7ddd32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 15:05:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/36f89af520a941a4aa477296aa45c7fe : FAILURE in 1h 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9005006763f54d74a4db71dd6702ffed : SUCCESS in 2h 05m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72a0105e90cf4a359473f7c8aa9951bd : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e47d7915534c34adb5991449ff66ac : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8697988f3c2f422a9628575848844865 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f989c091f6b845f997c3dc28de0d49f3 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35dba34a5aac4862bddf004687b2c766 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2202adf70494996a8b9d4522b426004 : SUCCESS in 4m 51s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/5a750d9307ea404aabe910099d3b2178 : SUCCESS in 1h 37m 33s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/9472749719424702bb563a7d6589fa72 : SUCCESS in 1h 15m 22s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/96eaf3f02e4d46d08bfff7ce0fb273cb : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a49a23538827713f7138804bfc3ce8a505add0b9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 17:56:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3651ab844d6adb9c4a88906a5c2a27373c6f5bc4","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-10-28 18:44:33.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/684263/3/iscsi/ : FAILURE in 47m 33s\n- os-brick-smb http://cloudbase-ci.com/os-brick/684263/3/smb/ : FAILURE in 41m 08s","accounts_in_message":[],"_revision_number":3},{"id":"29228281fc0647e23b0fcb0d2f3ea9d08dbac97d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-28 19:11:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/63/684263/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/a964f74 : SUCCESS in 1h 05m 52s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/63/684263/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/e95d17f : SUCCESS in 1h 15m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"ec768c124ec41da6f631820d4243edca8b129482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:59:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/67f21603293745b9a56d92c3562579b5 : SUCCESS in 1h 48m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/909c69e0a2824802b2e57d10b9e3440f : SUCCESS in 1h 54m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d006f78082246289de50e6e1b0dbfbf : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1d70ac56e54b109ad1bf1dd70790af : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6ba52aff7dc4a6cb8ecf42281c4736b : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf78e9777e841dd8eba504b78b0ad97 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0f11b683e044b3aa69b2a8eab435372 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62bdb9ab4b4a4629b5706b10f0f64119 : SUCCESS in 7m 05s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/68292bcb6b8e43b0b99680b7de1d8451 : SUCCESS in 1h 22m 08s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/b7e5282d9fd649afba292f8c4289dd45 : SUCCESS in 1h 37m 04s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d623e5cf1bb245518f2384199c8b9657 : SUCCESS in 1h 58m 07s","accounts_in_message":[],"_revision_number":3},{"id":"7856d41fabc7d93edb483cb891f439d0d5d998a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:59:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e9883725403d58a0ba4f1f1a515edb0a0f68ed8d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-10-28 20:26:26.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_SC_FC-os-brick/3931756/DellEMC_SC_FC-os-brick/13 : FAILURE in 22m 21s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_SC_iSCSI-os-brick/0b5f42f/DellEMC_SC_iSCSI-os-brick/11 : FAILURE in 27m 12s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"7abe946830f72823085d80c29047088b1937ba75","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-10-28 20:37:14.000000000","message":"Patch Set 3:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/63/684263/3/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/685ca1d : FAILURE in 35m 27s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/63/684263/3/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/7b9a084 : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":3},{"id":"37eca1edc7e496fdec6f128591b401d41604d46b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-28 20:56:31.000000000","message":"Patch Set 3:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-684263-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"221a9e82f7b35f209d24374b611cb4dad5e2f85d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-28 21:21:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/63/684263/3/upstream-check/cinder-cDOT-iSCSI/4f3b160/ : SUCCESS in 57m 13s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/63/684263/3/upstream-check/cinder-cDOT-NFS/fdeaafc/ : FAILURE in 1h 22m 23s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/63/684263/3/upstream-check/cinder-cDOT-FCP/43e9198/ : FAILURE in 1h 15m 58s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"e70d0a25a02a3190bb6b372c4b015be7b823bf61","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-10-28 21:21:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_XtremIO_FC_os-brick/5bb0efe/DellEMC_XtremIO_FC_os-brick/19 : SUCCESS in 49m 36s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_XtremIO_iSCSI_os-brick/5ba02ad/DellEMC_XtremIO_iSCSI_os-brick/19 : FAILURE in 1h 22m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"a4a4fdfd9e8cc535d8a706f6047326c0df223565","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-28 21:36:42.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_VxFlexOS_os-brick/17dbf07/DellEMC_VxFlexOS_os-brick/67 : SUCCESS in 1h 37m 27s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"6c658f1c167a0931eed058e9f8ea80cbed6c22f8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-10-28 21:37:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/63/684263/3/upstream-solidfire-check/cinder-solidfire-iSCSI/8ec609c/ : SUCCESS in 1h 38m 17s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ebdc99b6c93010f27257638dcd7049c68e4a896","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-10-28 21:43:57.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_VNX_FC_os-brick/afda859/DellEMC_VNX_FC_os-brick/9 : SUCCESS in 52m 47s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_VNX_iSCSI_os-brick/300c3c1/DellEMC_VNX_iSCSI_os-brick/9 : FAILURE in 1h 44m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"499c30b2a622d10434a45b0bc265fa075040155c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-10-28 21:44:49.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_Unity_FC_os-brick/d055853/DellEMC_Unity_FC_os-brick/18 : SUCCESS in 1h 11m 56s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/63/684263/3/check/DellEMC_Unity_iSCSI_os-brick/8ee1e83/DellEMC_Unity_iSCSI_os-brick/18 : SUCCESS in 1h 45m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"72704da6e4676720de9e1337e5665dee43c98f15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 21:49:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cbe7f8a73816bdb20f32fa1764c1123ec21c43ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 21:49:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d66e9a82508f4e2cb786faac521b285c : SUCCESS in 1h 34m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/012ff8ed45984c8bb9b9656aa72d8bf3 : SUCCESS in 1h 34m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50ef0066543842dda0fe38c4aa38dabf : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f0fea19d45415cb963e6b347e78727 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94da26eace1b447c86fb5882e5a96dbe : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f991e0f6bcd345578781d81ffa59ecd7 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0462b5416b94868aad14224e02ab55f : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fac55dbfe1a4182876d0fc97be74ba7 : SUCCESS in 7m 18s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/d6a820a0939e464b8540741af79a946a : SUCCESS in 1h 09m 39s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/090d7a6069514924a1e4a01d92025876 : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":3},{"id":"0bad1adcc06b98fd265495112244efcf7883ba80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 21:50:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6e1ccf19574846b732790e2e502aea : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3},{"id":"b48e4b565488d72b8989dcff1e6d8b2b47273edb","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-10-28 22:10:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/63/684263/3/check/kaminario-dsvm-tempest-full-iscsi/958539a : SUCCESS in 2h 10m 59s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/63/684263/3/check/kaminario-dsvm-tempest-full-FC/69acc7c : SUCCESS in 1h 40m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87171abef8bf2336f15ce3a7949f77d7999e11b7","revisions":{"d4ace818b57ec069184031c600f13ec17728508c":{"kind":"REWORK","_number":1,"created":"2019-09-24 09:58:07.000000000","uploader":{"_account_id":31019,"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","username":"pszczerx"},"ref":"refs/changes/63/684263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/63/684263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/63/684263/1"}}},"commit":{"parents":[{"commit":"db4ce9e5b490cb32fdbb5b41b857bd308023c706","subject":"nvmeof: Use subnqn to disconnect a volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/db4ce9e5b490cb32fdbb5b41b857bd308023c706"}]}],"author":{"name":"Szczerbik, Przemyslaw","email":"przemyslawx.szczerbik@intel.com","date":"2019-09-24 09:50:07.000000000","tz":60},"committer":{"name":"Szczerbik, Przemyslaw","email":"przemyslawx.szczerbik@intel.com","date":"2019-09-24 09:54:17.000000000","tz":60},"subject":"nvmeof: Fix broken UTs","message":"nvmeof: Fix broken UTs\n\nThis patch fixes broken nvmeof UTs. A few tests\ncalled \u0027asert_called_once_with\u0027 method on mock object,\ninstead of calling \u0027assert_called_once_with\u0027.\nThe mistake went undetected, because mock objects\nwere not configured to use auto-speccing.\nTo avoid the same problem in the future, configure all\nmocks to use autospec.\n\nAdditionally, replace \u0027assert_not_called\u0027 method which\nis not available on older versions of Python and caused\nlower-constraints testenv to fail.\n\nChange-Id: I54a91dfcfae774b1051662291afe1586e59901ac\nSigned-off-by: Szczerbik, Przemyslaw \u003cprzemyslawx.szczerbik@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d4ace818b57ec069184031c600f13ec17728508c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d4ace818b57ec069184031c600f13ec17728508c"}]},"branch":"refs/heads/master"},"1a6008806d3259b6053c1a251409f37ed576634b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-17 05:43:00.000000000","uploader":{"_account_id":31019,"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","username":"pszczerx"},"ref":"refs/changes/63/684263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/63/684263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/63/684263/2"}}},"commit":{"parents":[{"commit":"6ea276c22e204531653110089759d800fd4f7bbf","subject":"nvmeof: Use subnqn to disconnect a volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6ea276c22e204531653110089759d800fd4f7bbf"}]}],"author":{"name":"Szczerbik, Przemyslaw","email":"przemyslawx.szczerbik@intel.com","date":"2019-09-24 09:50:07.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","date":"2019-10-17 05:43:00.000000000","tz":0},"subject":"nvmeof: Fix broken UTs","message":"nvmeof: Fix broken UTs\n\nThis patch fixes broken nvmeof UTs. A few tests\ncalled \u0027asert_called_once_with\u0027 method on mock object,\ninstead of calling \u0027assert_called_once_with\u0027.\nThe mistake went undetected, because mock objects\nwere not configured to use auto-speccing.\nTo avoid the same problem in the future, configure all\nmocks to use autospec.\n\nAdditionally, replace \u0027assert_not_called\u0027 method which\nis not available on older versions of Python and caused\nlower-constraints testenv to fail.\n\nChange-Id: I54a91dfcfae774b1051662291afe1586e59901ac\nSigned-off-by: Szczerbik, Przemyslaw \u003cprzemyslawx.szczerbik@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1a6008806d3259b6053c1a251409f37ed576634b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1a6008806d3259b6053c1a251409f37ed576634b"}]},"branch":"refs/heads/master"},"87171abef8bf2336f15ce3a7949f77d7999e11b7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-28 12:57:28.000000000","uploader":{"_account_id":31019,"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","username":"pszczerx"},"ref":"refs/changes/63/684263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/63/684263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/63/684263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/63/684263/3"}}},"commit":{"parents":[{"commit":"707a8d5941cb6aa2e0d1c1eb0325b66d15a0c4f2","subject":"Merge \"Require oslo.privsep 1.32.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/707a8d5941cb6aa2e0d1c1eb0325b66d15a0c4f2"}]}],"author":{"name":"Szczerbik, Przemyslaw","email":"przemyslawx.szczerbik@intel.com","date":"2019-09-24 09:50:07.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslawx.szczerbik@intel.com","date":"2019-10-28 12:57:28.000000000","tz":0},"subject":"nvmeof: Fix broken UTs","message":"nvmeof: Fix broken UTs\n\nThis patch fixes broken nvmeof UTs. A few tests\ncalled \u0027asert_called_once_with\u0027 method on mock object,\ninstead of calling \u0027assert_called_once_with\u0027.\nThe mistake went undetected, because mock objects\nwere not configured to use auto-speccing.\nTo avoid the same problem in the future, configure all\nmocks to use autospec.\n\nAdditionally, replace \u0027assert_not_called\u0027 method which\nis not available on older versions of Python and caused\nlower-constraints testenv to fail.\n\nChange-Id: I54a91dfcfae774b1051662291afe1586e59901ac\nSigned-off-by: Szczerbik, Przemyslaw \u003cprzemyslawx.szczerbik@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/87171abef8bf2336f15ce3a7949f77d7999e11b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/87171abef8bf2336f15ce3a7949f77d7999e11b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
