)]}'
{"id":"openstack%2Fos-brick~690640","triplet_id":"openstack%2Fos-brick~master~I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9","project":"openstack/os-brick","branch":"master","topic":"FC-fix-broad-scan","hashtags":[],"change_id":"I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9","subject":"Fix FC scan too broad","status":"MERGED","created":"2019-10-23 17:06:46.000000000","updated":"2019-11-14 03:57:52.000000000","submitted":"2019-11-14 03:56:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690640-1573703796607-a4121484","meta_rev_id":"178e498511049fe1ce1fd00f8045dabc54774d54","_number":690640,"virtual_id_number":690640,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-11-11 22:56:04.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-11 17:49:03.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-11 17:39:29.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 03:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2019-11-11 20:27:29.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2019-11-11 17:29:12.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-11 16:45:12.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-11 18:33:32.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-11-11 15:49:58.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"date":"2019-11-13 21:34:47.000000000","_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-13 17:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":2,"date":"2019-11-11 17:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":2,"date":"2019-11-14 02:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":1,"date":"2019-11-14 02:17:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 18:19:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-23 19:02:27.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-10-24 06:37:42.000000000","updated_by":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"reviewer":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"state":"REVIEWER"},{"updated":"2019-10-24 20:34:19.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-10-24 21:05:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-10-24 21:25:35.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-24 22:17:19.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-24 23:11:06.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2019-11-07 15:52:15.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2019-11-11 15:49:58.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-11-11 16:45:12.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-11 17:29:12.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-11 17:30:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-11-11 17:39:29.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-11 17:49:03.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-11 18:33:32.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-11 20:27:29.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2019-11-11 22:56:04.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-13 17:32:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-14 02:17:15.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-11-14 03:56:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7970dad6d8290eb537ecc8788ea3d259c34d9109","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-10-23 17:06:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"492ddbea25846172ea869e32c75d155e6ab168a0","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-10-23 18:16:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/690640/1/iscsi/ : SUCCESS in 1h 09m 02s\n- os-brick-smb http://cloudbase-ci.com/os-brick/690640/1/smb/ : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"49aa56c2e7d8618688ebd4c98c4ee884fef4370a","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-23 18:18:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/40/690640/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f2797f6 : SUCCESS in 1h 03m 00s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/40/690640/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/1eb42ce : SUCCESS in 1h 11m 15s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"5423bbc12474f757ab43f105562a862c3095834e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-23 18:19:21.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck cinder-mlnx\u0027.\n\n- Cinder-ISER-ISCSI http://13.74.249.42/40/690640/1/check-cinder/Cinder-ISER-ISCSI/7778b2b : FAILURE in 33m 26s\n- Cinder-ISER-LIO http://13.74.249.42/40/690640/1/check-cinder/Cinder-ISER-LIO/491a912 : FAILURE in 31m 24s\n\nTo re-run the job post \u0027recheck cinder-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c1ff11c3cdaa1b99c34bd008721d929117fab3c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-10-23 18:49:07.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_Unity_FC_os-brick/6a9e98a/DellEMC_Unity_FC_os-brick/6 : SUCCESS in 1h 12m 02s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_Unity_iSCSI_os-brick/28a316c/DellEMC_Unity_iSCSI_os-brick/6 : SUCCESS in 1h 42m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"538830701710a412afc57a8af40c08b19f65d2dc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-10-23 18:56:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_XtremIO_FC_os-brick/3ae7005/DellEMC_XtremIO_FC_os-brick/7 : SUCCESS in 49m 40s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_XtremIO_iSCSI_os-brick/c70fa57/DellEMC_XtremIO_iSCSI_os-brick/7 : FAILURE in 1h 18m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"4cdf51725dc544702b02745c696ad8d97884df36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 19:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a7772e0df7c04e42a7d362615adbf60f : SUCCESS in 1h 38m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8403448e82cf4e15bcb3c7d3b6a1912e : SUCCESS in 1h 41m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e51ce12de11a4b808a991fb4e3af58a9 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b20f8af2ba14aedb6b36dd087d3b2a0 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/072047843c10411b9501a23f469331a8 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25c23b3b86aa4438ba3d332fa273006a : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b8610244ffc45a0aa2eb0203f260f98 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c90cad93b2d7439eb67c8d0202c11b02 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/972c15177a534e7089646330b146e89c : SUCCESS in 4m 35s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/40b7f074b45a4122ad6b7270fec3570f : SUCCESS in 1h 46m 34s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/80cc21702b094eb68452cd2f1f4d3aea : SUCCESS in 1h 43m 57s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8d1683d5613e4276be0b9343d873b5de : SUCCESS in 1h 46m 28s","accounts_in_message":[],"_revision_number":1},{"id":"080fc223c3ea521dd3dfcd238c9d7d801652a79b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-10-23 19:27:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_SC_FC-os-brick/654ccbf/DellEMC_SC_FC-os-brick/7 : FAILURE in 21m 32s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_SC_iSCSI-os-brick/1ef09e6/DellEMC_SC_iSCSI-os-brick/5 : FAILURE in 26m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"87401da7cf1cd877c17f014fe64d7fa2b6de7371","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-10-23 19:50:39.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/40/690640/1/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/c8cb152 : FAILURE in 45m 18s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/40/690640/1/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/128e212 : FAILURE in 33m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d52c0eedc9b81f427412387fe76e9971886c14c2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-23 20:09:23.000000000","message":"Patch Set 1:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-40-690640-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"6686a1e1a20cadb08a8cd795e5e9b3a1ae4882c1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-23 20:13:59.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/40/690640/1/check/tempest-dsvm-full-xenial/d32ec26/ : FAILURE in 2h 59m 45s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/40/690640/1/check/devstack-plugin-ceph-dsvm-xenial/7ffd2d1/ : FAILURE in 3h 06m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c39eb18631fcd6d4c4f21dfc7dfbc38b013cd02d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-23 20:29:38.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_VxFlexOS_os-brick/ccf5892/DellEMC_VxFlexOS_os-brick/61 : SUCCESS in 1h 28m 16s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"7d56386c424c188b0aa062912a454e5689152d7a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-10-23 20:47:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_VNX_FC_os-brick/0d40e53/DellEMC_VNX_FC_os-brick/3 : SUCCESS in 1h 00m 47s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_VNX_iSCSI_os-brick/c02697d/DellEMC_VNX_iSCSI_os-brick/3 : FAILURE in 1h 46m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"532ed213db422f905d575a59f7444369462b2bef","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-10-23 21:11:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/40/690640/1/check/kaminario-dsvm-tempest-full-iscsi/9ff1c72 : SUCCESS in 2h 09m 40s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/40/690640/1/check/kaminario-dsvm-tempest-full-FC/1096d20 : SUCCESS in 1h 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9ac6e85a65cf0cbf57f101167bafdb4b903e1a86","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-10-23 21:21:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/40/690640/1/upstream-solidfire-check/cinder-solidfire-iSCSI/977e4db/ : SUCCESS in 2h 20m 13s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e68a69364a9c55ff60a75857ffed2bbff76e5d8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-23 21:54:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/40/690640/1/upstream-check/cinder-cDOT-iSCSI/4688df1/ : SUCCESS in 2h 02m 30s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/40/690640/1/upstream-check/cinder-cDOT-NFS/832d797/ : FAILURE in 19m 58s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/40/690640/1/upstream-check/cinder-cDOT-FCP/453b388/ : FAILURE in 2h 53m 30s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"30dec507fa358e9c60fa27665be0a13ce463d692","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-10-23 22:49:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_XtremIO_FC_os-brick/d7b1834/DellEMC_XtremIO_FC_os-brick/10 : SUCCESS in 48m 59s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_XtremIO_iSCSI_os-brick/961cd29/DellEMC_XtremIO_iSCSI_os-brick/10 : FAILURE in 1h 17m 37s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"67719284771988b72ac8cf611abcad6ca53b2b2b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-10-23 23:12:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_Unity_FC_os-brick/7945929/DellEMC_Unity_FC_os-brick/9 : SUCCESS in 1h 12m 48s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/40/690640/1/check/DellEMC_Unity_iSCSI_os-brick/f5f475f/DellEMC_Unity_iSCSI_os-brick/9 : SUCCESS in 1h 26m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"cde6112cbadb4b6c6e49d449b8c848ed1dc194f4","author":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"date":"2019-10-24 06:37:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f330c38e367ace961ddbff4759357c1d984a18bc","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-10-24 15:51:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b4d7b99dc3b230c56f51fcfa379238725238223a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-10-24 17:11:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64e281bbd1e77311454c693f20c47b3d789882f5","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-10-24 18:15:59.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/690640/2/iscsi/ : FAILURE in 1h 02m 44s\n- os-brick-smb http://cloudbase-ci.com/os-brick/690640/2/smb/ : FAILURE in 1h 04m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ae7728c6df01a20be5b128d65ed6626349bc5937","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-24 19:02:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/40/690640/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/d4f5ffd : SUCCESS in 1h 05m 17s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/40/690640/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/b22e1a1 : FAILURE in 1h 51m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b80c4eb326f8446342978786365e7fdf4243d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:16:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/08a8affdb85d4bd4b83141eade18058d : SUCCESS in 1h 33m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f2c92ff6f2a4d829dc87f8deb3c201c : SUCCESS in 1h 39m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb69c6f1342e4fb3b145fe0b5d754673 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba50890fa7441f8995387d60906c2b4 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8fff4207aad4c659f93aa6494d06888 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c9e5aa0d15746348e76ef2ec8969341 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09076f05f65a46ea8b3049785e82a136 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f6617ea52d495ca100aa14dd11fcae : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e769ee1a5a6b4568aa6acdb6faec6e34 : SUCCESS in 4m 58s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/71efa3df47654973aea6d37a8550631d : SUCCESS in 1h 31m 42s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/5bc74480cc2f49ed8bc6a9781a4b3140 : SUCCESS in 1h 55m 36s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/63b08c011ba54956b1414c3c1cb2658e : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":2},{"id":"854a5adb0272b91944be5803667fd81a508b9619","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-10-24 19:51:44.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/40/690640/2/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/19cbc3c : FAILURE in 32m 23s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/40/690640/2/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/74137c4 : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6a197c9de4113807e5ee7b2a43ffc1b6f6faed0c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-24 20:15:25.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/40/690640/2/check/tempest-dsvm-full-xenial/15a643c/ : FAILURE in 2h 53m 00s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/40/690640/2/check/devstack-plugin-ceph-dsvm-xenial/293765d/ : FAILURE in 3h 03m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e023ab52a1c3a518fed3fb9f566263b6a7f3803","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-10-24 20:34:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/40/690640/2/upstream-solidfire-check/cinder-solidfire-iSCSI/7a3982e/ : SUCCESS in 1h 12m 36s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"89554c0a1bdce7133947dee841e36a4fbdef4233","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-24 20:43:16.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-40-690640-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"971059261201c02050c2f93137827408dd806c7c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-10-24 20:46:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_SC_FC-os-brick/0c9e6af/DellEMC_SC_FC-os-brick/11 : FAILURE in 21m 45s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_SC_iSCSI-os-brick/cfddf8f/DellEMC_SC_iSCSI-os-brick/9 : FAILURE in 29m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"b33fe003c3240ed561a9779a022b3d9033a09e03","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-10-24 21:03:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_XtremIO_FC_os-brick/a29e96e/DellEMC_XtremIO_FC_os-brick/14 : SUCCESS in 49m 30s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_XtremIO_iSCSI_os-brick/9e35c69/DellEMC_XtremIO_iSCSI_os-brick/14 : FAILURE in 1h 14m 48s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"7d36ec17e4a7782223ab7e9dd36d77cc644bbbf5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-10-24 21:05:47.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_Unity_FC_os-brick/ce5d509/DellEMC_Unity_FC_os-brick/13 : SUCCESS in 1h 13m 53s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_Unity_iSCSI_os-brick/07fcf45/DellEMC_Unity_iSCSI_os-brick/13 : SUCCESS in 1h 33m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"276df0f75371b7a1b85e34d57597930e79b738bf","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-24 21:25:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/40/690640/2/upstream-check/cinder-cDOT-iSCSI/bf10441/ : SUCCESS in 1h 02m 46s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/40/690640/2/upstream-check/cinder-cDOT-NFS/b525d72/ : FAILURE in 2h 03m 56s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/40/690640/2/upstream-check/cinder-cDOT-FCP/4e495be/ : FAILURE in 1h 29m 19s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"80dca185a1400ec858c9a706d75d027164403470","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-10-24 22:00:35.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_VNX_FC_os-brick/44b4e84/DellEMC_VNX_FC_os-brick/7 : SUCCESS in 58m 20s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_VNX_iSCSI_os-brick/5c21a5f/DellEMC_VNX_iSCSI_os-brick/7 : SUCCESS in 1h 42m 54s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"52feb46f07a629ea50e5022b0065b8d28e87bc8b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-24 22:17:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/40/690640/2/check/DellEMC_VxFlexOS_os-brick/4236d3f/DellEMC_VxFlexOS_os-brick/65 : SUCCESS in 1h 30m 15s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"6155a2ac55b7038d04a8cdd8b1165f406fe319ac","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2019-10-24 23:11:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://54.209.116.144/40/690640/2/check/kaminario-dsvm-tempest-full-iscsi/37b075a : SUCCESS in 2h 12m 13s\n- kaminario-dsvm-tempest-full-FC http://54.209.116.144/40/690640/2/check/kaminario-dsvm-tempest-full-FC/3df42cd : SUCCESS in 1h 39m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ca18f5fbaf9492a77a609364ce51d5626b803b0f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-11-07 15:52:15.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nDo you know which drivers don\u0027t return an initiator_target_map for FC connections?   This is bad.","accounts_in_message":[],"_revision_number":2},{"id":"308673a7412257c7b6e410515d51c40f7fa11992","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-11 04:31:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nCode looks good to me.  Caught a few typos in comments, and one old comment that I don\u0027t think is accurate any more.","accounts_in_message":[],"_revision_number":2},{"id":"0bfc1f765f9544fe84c724d919a48c0e4664457e","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-11-11 14:48:43.000000000","message":"Patch Set 2:\n\n(6 comments)\n\nThank you for the reviews!!","accounts_in_message":[],"_revision_number":2},{"id":"0767730b45bf6b913dcf7a2e20df067bc2382433","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-11-11 14:54:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bab122e0cfe9a0a5db77edee3767b362ba8b528e","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-11-11 15:49:58.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/690640/3/iscsi/ : FAILURE in 53m 58s\n- os-brick-smb http://cloudbase-ci.com/os-brick/690640/3/smb/ : FAILURE in 55m 19s","accounts_in_message":[],"_revision_number":3},{"id":"74c59e50345a714df32d0a9145517db93a9c7aed","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-11-11 16:05:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_XtremIO_FC_os-brick/98a9282/DellEMC_XtremIO_FC_os-brick/26 : SUCCESS in 49m 19s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_XtremIO_iSCSI_os-brick/f2feae7/DellEMC_XtremIO_iSCSI_os-brick/26 : FAILURE in 1h 11m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"bdbb7f4b9bac55c5d254420c254a60b5dccb6648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:36:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/29862e4b12f34296b07b5e0608164071 : SUCCESS in 1h 35m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f52177a550df422fbb5da6add924b81e : SUCCESS in 1h 23m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b39da8e1b75842dcafee9e1b93e27651 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1ab0de28ee45dd8b5d9ebc2aef9476 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05a557c989854ab9a029b1514f913159 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1224038a7b9a4f2a9ce240319c292e93 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d571ce7858644c30aef8a86ab1b14573 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2236592dd7f47e897cdd70d4d00c4f8 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86cbe8ac8ea8447291e0cb8d10bfa6ea : SUCCESS in 5m 46s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/c92cd4824897420591d827ea8833fe9b : SUCCESS in 1h 26m 00s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/9864500334414e9fa64136f439bd329b : SUCCESS in 1h 30m 57s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bfd05e6f7b6b453c989b7e4dbb9c94f1 : SUCCESS in 1h 27m 06s","accounts_in_message":[],"_revision_number":3},{"id":"75f0654bb7d581d2499d821e89750e83aa50803b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-11-11 16:45:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/40/690640/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/80c1588 : FAILURE in 1h 50m 41s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/40/690640/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c9bc35d : FAILURE in 1h 09m 23s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"961c6991497f6858d5ede4d6e5666140dcf57298","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-11-11 17:29:12.000000000","message":"Patch Set 3:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-40-690640-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"a48e843791bce74ae9306a2af27e9a9b224cfff2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-11 17:30:16.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"17c5890397ddbd07b7e71e7f6138ed330d1d1735","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-11-11 17:39:29.000000000","message":"Patch Set 3:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/40/690640/3/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/635ef68 : FAILURE in 31m 43s\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/40/690640/3/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/2335a76 : FAILURE in 32m 36s","accounts_in_message":[],"_revision_number":3},{"id":"a8c592ab87411fc618cf1eedf1e834728892c31b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-11-11 17:49:03.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/40/690640/3/check/tempest-dsvm-full-xenial/6d6ce79/ : SUCCESS in 2h 54m 09s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/40/690640/3/check/devstack-plugin-ceph-dsvm-xenial/4d1b0fb/ : SUCCESS in 2h 47m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ee9cd80e016c9666fc3f6b20007977330b2f409","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-11-11 18:33:32.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_VNX_FC_os-brick/809af30/DellEMC_VNX_FC_os-brick/13 : SUCCESS in 58m 28s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_VNX_iSCSI_os-brick/c0e9322/DellEMC_VNX_iSCSI_os-brick/13 : SUCCESS in 1h 41m 59s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"8ab776838a6d5124d894b7f9d299ffbeadfe21ed","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-11-11 20:27:29.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_SC_FC-os-brick/29a6ac3/DellEMC_SC_FC-os-brick/17 : SUCCESS in 1h 16m 52s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_SC_iSCSI-os-brick/cba28fe/DellEMC_SC_iSCSI-os-brick/15 : FAILURE in 2h 51m 55s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"ca911827510219a6eb041cf40bd6c4f3fb004f9f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-11-11 22:56:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_XtremIO_FC_os-brick/692ceef/DellEMC_XtremIO_FC_os-brick/29 : SUCCESS in 49m 23s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/40/690640/3/check/DellEMC_XtremIO_iSCSI_os-brick/51bbec5/DellEMC_XtremIO_iSCSI_os-brick/29 : FAILURE in 1h 15m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"6b0f88b2ff8f73e1af36858774cfc5bff02553af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-13 17:32:10.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nLooks reasonable to me. Holding off on +W to give others a chance to review, but I don\u0027t think we need to wait too long.","accounts_in_message":[],"_revision_number":3},{"id":"bb821ee30f07ff63199706cdb5feedb3675f072b","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-11-13 21:34:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38bb735b551e251c01c753ee26b47ab3ca060b34","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-11-14 02:17:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nTypo can be fixed in a follow-up patch.","accounts_in_message":[],"_revision_number":3},{"id":"f466ec8ac10cbda394869cc7eec19c4f5afd85a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 02:17:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"05f624b0707d6e2ca985b3aebf7b9de0266e5b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 03:56:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/37d04568db4544419e4c41e7e9f3145a : SUCCESS in 1h 37m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2549ee6080794120ab86861e521c1df7 : SUCCESS in 1h 22m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81e51acd1b12400fabf0a928ed1a1b39 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda10b0855914d468dd5ef1a2dafd6ba : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb9d26c82fc14c6e84a49c226dad4f89 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/025d7bf3ff314fd498c2c4deac0209dc : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c65a0c1380504d07b36b5115968fa407 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d57bc26d25f4bf294f4fbe130ddcb19 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7111a3100f5e40d49b95a7e43d799e34 : SUCCESS in 4m 00s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/8f700c9049a64ea196f910d8457a1d54 : SUCCESS in 1h 27m 47s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14250a5e37c04ea89d817e6df77ff410 : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ef5b9343d179ca5985a8c388d6d08502d9f6f3d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 03:56:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3281e6879b510b7d420ec023b5ff7a6bb9c9dc8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 03:57:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fb556cf27b4e7c94f142caf14b4ce2 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/830811fc882148f4967deb9ac8434723 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"708733e49598145eb8e898d486723dc3df00a52f","revisions":{"47f04ec859345b9ce62e3f1b3a6c42ec996bae03":{"kind":"REWORK","_number":1,"created":"2019-10-23 17:06:46.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/40/690640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/40/690640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/40/690640/1"}}},"commit":{"parents":[{"commit":"cc17adb70f969fe25ef7aa512a8405035e4197ca","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cc17adb70f969fe25ef7aa512a8405035e4197ca"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-10-14 14:42:22.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-10-23 16:23:57.000000000","tz":120},"subject":"Fix FC scan too broad","message":"Fix FC scan too broad\n\nAfter some changes to the FC connector we have introduced a regression\non the way we do the scans, and we end up scanning using wildcards even\nthough we shouldn\u0027t.\n\nThe targets in the \"initiator_target_map\" don\u0027t mean that they are all\nconnected, so we must take that into account.\n\nWith the current code, if we have the following connections:\n\nHBA host7 ---- SWITCH ---- port W (channel 0, target 2)\n          \\--- SWITCH ---- port X (channel 0, target 3)\n\nHBA host8 ---- SWITCH ---- port Y (channel 0, target 2)\n          \\--- SWITCH ---- port Z (channel 0, target 3)\n\nWe will end up with the following scans 8 scans for LUN L:\n\n  - - L \u003e host7\n  - - L \u003e host7\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n  - - L \u003e host8\n  - - L \u003e host8\n\nWhich correspond to the responses from _get_hba_channel_scsi_target like\nthis:\n\n               port Y          port Z          port W          port X\n  host7 ... [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]\n  host8 ... [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]\n\nAnd we should only be doing 4 scans:\n\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n\nThe FC initiator is automatically populating sysfs with the target\nports, so the default will be to not use wildcards and always look for\nthe target ports in sysfs to limit our scanning.\n\nIn case a driver doesn\u0027t want this behavior because they know all connections\nshould exist, then it can pass True as value for the \"force_scan\" key in\nthe connection information dictionary.\n\nClose-Bug: #1849504\nRelated-Bug: #1765000\nRelated-Bug: #1828440\nChange-Id: I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/47f04ec859345b9ce62e3f1b3a6c42ec996bae03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/47f04ec859345b9ce62e3f1b3a6c42ec996bae03"}]},"branch":"refs/heads/master"},"500907e130ddf4f558d865f4e890b555fa0b380e":{"kind":"REWORK","_number":2,"created":"2019-10-24 17:11:04.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/40/690640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/40/690640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/40/690640/2"}}},"commit":{"parents":[{"commit":"cc17adb70f969fe25ef7aa512a8405035e4197ca","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cc17adb70f969fe25ef7aa512a8405035e4197ca"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-10-14 14:42:22.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-10-24 16:40:23.000000000","tz":120},"subject":"Fix FC scan too broad","message":"Fix FC scan too broad\n\nAfter some changes to the FC connector we have introduced a regression\non the way we do the scans, and we end up scanning using wildcards even\nthough we shouldn\u0027t.\n\nThe targets in the \"initiator_target_map\" don\u0027t mean that they are all\nconnected, so we must take that into account.\n\nWith the current code, if we have the following connections:\n\nHBA host7 ---- SWITCH ---- port W (channel 0, target 2)\n          \\--- SWITCH ---- port X (channel 0, target 3)\n\nHBA host8 ---- SWITCH ---- port Y (channel 0, target 2)\n          \\--- SWITCH ---- port Z (channel 0, target 3)\n\nWe will end up with the following scans 8 scans for LUN L:\n\n  - - L \u003e host7\n  - - L \u003e host7\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n  - - L \u003e host8\n  - - L \u003e host8\n\nWhich correspond to the responses from _get_hba_channel_scsi_target like\nthis:\n\n               port Y          port Z          port W          port X\n  host7 ... [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]\n  host8 ... [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]\n\nAnd we should only be doing 4 scans:\n\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n\nMost storage arrays get their target ports automatically detected by the\nLinux FC initiator and sysfs gets populated with that information, but\nthere are some that don\u0027t.\n\nWe\u0027ll do a narrow scan using the channel, target, and LUN for the former\nand a wider scan for the latter.\n\nIf all paths to a former type of array were down on the system boot the\narray could look like it\u0027s of the latter type and make us bring us\nunwanted volumes into the system by doing a broad scan.\n\nTo prevent this from happening Cinder drivers can use the\n\"enable_wildcard_scan\" key in the connection information to let us know\nthey don\u0027t want us to do broad scans even if no target ports are found\n(because they know the cause is there\u0027s no connection).\n\nClose-Bug: #1849504\nRelated-Bug: #1765000\nRelated-Bug: #1828440\nChange-Id: I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/500907e130ddf4f558d865f4e890b555fa0b380e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/500907e130ddf4f558d865f4e890b555fa0b380e"}]},"branch":"refs/heads/master"},"708733e49598145eb8e898d486723dc3df00a52f":{"kind":"REWORK","_number":3,"created":"2019-11-11 14:54:15.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/40/690640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/40/690640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/690640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/40/690640/3"}}},"commit":{"parents":[{"commit":"cc17adb70f969fe25ef7aa512a8405035e4197ca","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cc17adb70f969fe25ef7aa512a8405035e4197ca"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-10-14 14:42:22.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-11-11 14:47:59.000000000","tz":60},"subject":"Fix FC scan too broad","message":"Fix FC scan too broad\n\nAfter some changes to the FC connector we have introduced a regression\non the way we do the scans, and we end up scanning using wildcards even\nthough we shouldn\u0027t.\n\nThe targets in the \"initiator_target_map\" don\u0027t mean that they are all\nconnected, so we must take that into account.\n\nWith the current code, if we have the following connections:\n\nHBA host7 ---- SWITCH ---- port W (channel 0, target 2)\n          \\--- SWITCH ---- port X (channel 0, target 3)\n\nHBA host8 ---- SWITCH ---- port Y (channel 0, target 2)\n          \\--- SWITCH ---- port Z (channel 0, target 3)\n\nWe will end up with the following scans 8 scans for LUN L:\n\n  - - L \u003e host7\n  - - L \u003e host7\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n  - - L \u003e host8\n  - - L \u003e host8\n\nWhich correspond to the responses from _get_hba_channel_scsi_target like\nthis:\n\n               port Y          port Z          port W          port X\n  host7 ... [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]\n  host8 ... [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]\n\nAnd we should only be doing 4 scans:\n\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n\nMost storage arrays get their target ports automatically detected by the\nLinux FC initiator and sysfs gets populated with that information, but\nthere are some that don\u0027t.\n\nWe\u0027ll do a narrow scan using the channel, target, and LUN for the former\nand a wider scan for the latter.\n\nIf all paths to a former type of array were down on the system boot the\narray could look like it\u0027s of the latter type and make us bring us\nunwanted volumes into the system by doing a broad scan.\n\nTo prevent this from happening Cinder drivers can use the\n\"enable_wildcard_scan\" key in the connection information to let us know\nthey don\u0027t want us to do broad scans even if no target ports are found\n(because they know the cause is there\u0027s no connection).\n\nClose-Bug: #1849504\nRelated-Bug: #1765000\nRelated-Bug: #1828440\nChange-Id: I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/708733e49598145eb8e898d486723dc3df00a52f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/708733e49598145eb8e898d486723dc3df00a52f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
