)]}'
{"id":"openstack%2Fos-brick~696384","triplet_id":"openstack%2Fos-brick~stable%2Fqueens~I7ca0bf5fac20a370097f1c39ce005a1b23814459","project":"openstack/os-brick","branch":"stable/queens","topic":"bug/1849504","hashtags":[],"change_id":"I7ca0bf5fac20a370097f1c39ce005a1b23814459","subject":"Handle None value \u0027inititator_target_map\u0027","status":"MERGED","created":"2019-11-27 18:32:38.000000000","updated":"2019-12-03 18:23:29.000000000","submitted":"2019-12-03 18:23:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696384-1575397409114-73dee3f6","meta_rev_id":"365b6c01dfe459d8409d364ac95089b67c8a3017","_number":696384,"virtual_id_number":696384,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"date":"2019-11-27 20:12:19.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-11-27 19:56:23.000000000","_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-27 22:55:01.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-27 22:35:37.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 18:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-27 20:30:34.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-11-27 19:48:27.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-03 16:02:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-03 15:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-03 16:02:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"date":"2019-12-03 15:34:19.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-27 18:32:38.000000000","_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 18:32:38.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2019-11-27 19:48:27.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-11-27 19:56:23.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 20:12:19.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 20:30:34.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-11-27 22:35:37.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 22:55:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 15:56:20.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-12-03 16:02:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-12-03 18:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ddc7b44102f0e8ff3801751f697436d97836a3e","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-11-27 18:32:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6e2917bd4ea58b0974d52d1c405b9c524854c3c","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-11-27 19:48:27.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/696384/1/iscsi/ : FAILURE in 1h 15m 18s\n- os-brick-smb http://cloudbase-ci.com/os-brick/696384/1/smb/ : FAILURE in 1h 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ea1a5a116e6f871d8948a67c3de6b5fe20a793f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:56:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ac9689fec48b4757b708163c7f941d2e : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8aa5df531f148a78c7d35486f5fdcc1 : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bcc5b3886f14d91990f6bfb8e53c37c : SUCCESS in 3m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d4107479383f49eaaac083b9269702cd : SUCCESS in 3m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/59d5fd72ec5043ff8b71b76be2b0169f : SUCCESS in 1h 18m 26s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/b1f579cf29a0456cbfc1872958586fc5 : SUCCESS in 1h 04m 18s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/2206c646ba0f46409d96aa1debe6ccf8 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"46902be8990176a251d97a5a43b65e9c6eac01cb","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-11-27 19:56:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/84/696384/1/upstream-eseries-check/cinder-eseries-iSCSI/9fb4129/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/84/696384/1/upstream-eseries-check/cinder-eseries-FCP/bf85b82/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":1},{"id":"f1f62d3fd834b175eb81224b4252cf4c0fe2388e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-11-27 20:12:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/84/696384/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/ec77459 : FAILURE in 9m 14s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/84/696384/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/fb05f09 : FAILURE in 12m 02s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"ff0de1a92f6023bf914877100bb5624ad82c4ea1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-11-27 20:30:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/84/696384/1/upstream-solidfire-check/cinder-solidfire-iSCSI/7f5111d/ : FAILURE in 34m 08s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"5142acf067abf2e0e986272a3827aeb06d21be8b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-11-27 22:35:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/84/696384/1/upstream-check/cinder-cDOT-iSCSI/d5fb6dd/ : FAILURE in 1h 20m 48s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/84/696384/1/upstream-check/cinder-cDOT-iSCSI-IPv6/0a25973/ : NOT_REGISTERED\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/84/696384/1/upstream-check/cinder-cDOT-NFS/d691a08/ : FAILURE in 51m 08s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/84/696384/1/upstream-check/cinder-cDOT-NFS-IPv6/5117720/ : FAILURE in 2h 39m 18s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/84/696384/1/upstream-check/cinder-cDOT-FCP/2191bc3/ : FAILURE in 1h 17m 10s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e7a838a3b25494fa07760446fb357267f6acfa24","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-11-27 22:55:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/84/696384/1/check/tempest-dsvm-full-xenial/e004dab/ : SUCCESS in 2h 40m 49s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/84/696384/1/check/devstack-plugin-ceph-dsvm-xenial/8f582a0/ : SUCCESS in 2h 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfcceb81b76c94df8002d8478f1e1019b879dc43","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-28 03:19:22.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t think this change is in Rocky.","accounts_in_message":[],"_revision_number":1},{"id":"79d3dca2da7914c5ae436e3f4f4c3d1326732ea8","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-11-28 11:03:26.000000000","message":"Patch Set 1: Workflow-1\n\nYou are right!  I thought it was because I needed it to do a clean backport of the FC patch...","accounts_in_message":[],"_revision_number":1},{"id":"009b7269b3cf6e5eb337768774ed9cdfcdfe73d5","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-11-28 11:20:25.000000000","message":"Patch Set 1:\n\n@rosmaita Here\u0027s the backport for Rocky, thanks for catching this https://review.opendev.org/696521","accounts_in_message":[],"_revision_number":1},{"id":"857280f176111838229879b5b4b975db4972a24d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-02 13:35:37.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"6c26084bdfb8e3c249f26eb86c540be3d8c0514e","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-12-03 15:34:19.000000000","message":"Patch Set 1: -Workflow\n\nhttps://review.opendev.org/#/c/696521/ is in the process of merging","accounts_in_message":[],"_revision_number":1},{"id":"90134e16cb2edaa7731b2c939f56686a67a35dcf","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-12-03 15:56:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3a76737065de2cd0a6081fe0bd1675f71b5284e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-03 16:02:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nRocky has merged and this is a good backport for Queens.","accounts_in_message":[],"_revision_number":1},{"id":"e6faffdfbd509781fbd4749d931a65dba54aa9ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 16:02:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbff2358c00fcf3f7dcef6223e2ef16cfaa07fcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:23:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/73c614c64ed64b2aa36b093cdad42070 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6cc9a6b30c45aabe3805a4441a1156 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90f922ad5fae4af0981d4f360370944c : SUCCESS in 3m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe271204b5964159b9c90b26e809bdc1 : SUCCESS in 3m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/25a8834c3a7e41b0abba55d10dde17f9 : SUCCESS in 1h 36m 52s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/ec112bcbb8a54cd6b3b4530584f6aebc : SUCCESS in 1h 28m 40s","accounts_in_message":[],"_revision_number":1},{"id":"22fcb6a048b407cfaa77b437156b0df7b75d8b5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:23:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4e616e3a07009143ed8c840ec38b7c937180d07","revisions":{"c4e616e3a07009143ed8c840ec38b7c937180d07":{"kind":"REWORK","_number":1,"created":"2019-11-27 18:32:38.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/84/696384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/696384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/696384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/696384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/696384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/696384/1"}}},"commit":{"parents":[{"commit":"8d3e40b7cd0ee73cc422afddc0ad37a23673ef73","subject":"Check path alive before get scsi wwn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8d3e40b7cd0ee73cc422afddc0ad37a23673ef73"}]}],"author":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2018-01-30 11:08:57.000000000","tz":-300},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-11-27 18:31:55.000000000","tz":60},"subject":"Handle None value \u0027inititator_target_map\u0027","message":"Handle None value \u0027inititator_target_map\u0027\n\nShould check not only if \u0027initiator_target_map\u0027 is in conn_props but\nalso its value is not None.\n\nChange-Id: I7ca0bf5fac20a370097f1c39ce005a1b23814459\nCloses-Bug: #1746218\n(cherry picked from commit 53877a95bc9f4d40315f876c076a7a32817ea904)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c4e616e3a07009143ed8c840ec38b7c937180d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c4e616e3a07009143ed8c840ec38b7c937180d07"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
