)]}'
{"id":"openstack%2Fos-brick~696385","triplet_id":"openstack%2Fos-brick~stable%2Fqueens~I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9","project":"openstack/os-brick","branch":"stable/queens","topic":"bug/1849504","hashtags":[],"change_id":"I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9","subject":"Fix FC scan too broad","status":"MERGED","created":"2019-11-27 18:32:38.000000000","updated":"2019-12-04 15:23:46.000000000","submitted":"2019-12-04 15:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696385-1575472947066-758459c9","meta_rev_id":"cf0961bf0276bfb3438d0fc1e69acdaa652722ad","_number":696385,"virtual_id_number":696385,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-27 20:41:50.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-11-27 20:16:13.000000000","_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-27 22:58:55.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-27 23:04:09.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 15:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-27 21:13:58.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-11-27 19:48:41.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-28 02:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-03 15:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-04 13:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":1,"date":"2019-12-04 09:52:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 19:48:41.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-11-27 20:16:13.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 20:41:50.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 21:13:58.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-11-27 22:58:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-27 23:04:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-28 02:58:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-12-04 13:43:19.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-12-04 15:22:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e8712fe4b3c9c7c471b0bac3f1aa585e22eb737","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-11-27 18:32:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e7376d5b27888f5b8f64e08e6e33bdc63384ed7","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-11-27 19:48:41.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/696385/1/iscsi/ : FAILURE in 1h 14m 59s\n- os-brick-smb http://cloudbase-ci.com/os-brick/696385/1/smb/ : FAILURE in 1h 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7d675fe1de8ab373f5b787296f23e04146a69283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:07:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a11b6c0f581d46d9bf30069c7d155840 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa06112687441eb92b786657894dbc1 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dee59432ad814ae293ec70443d72c3ac : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ce4aa29a46da490b8db417245bec6ab3 : SUCCESS in 3m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/107fc7c97ad94e8b9105c6e5e16e7e77 : SUCCESS in 1h 28m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c008a1de93c04727925be44cb271a83b : SUCCESS in 3m 45s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/be9bf0afe94c4d938335165851807d0b : SUCCESS in 1h 05m 01s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/3f800932e5d84ae7b58cc2a498b02829 : SUCCESS in 1h 05m 43s","accounts_in_message":[],"_revision_number":1},{"id":"60645e2bd7af7229d18d32d20d109ba3d2a0994e","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-11-27 20:16:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/85/696385/1/upstream-eseries-check/cinder-eseries-iSCSI/6d205e6/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/85/696385/1/upstream-eseries-check/cinder-eseries-FCP/2589312/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":1},{"id":"77736d72e74b5486ec83e06e88f1d42852808e12","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-11-27 20:41:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/85/696385/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/5618bc1 : FAILURE in 8m 41s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/85/696385/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/39dda6f : FAILURE in 11m 55s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"aaeade1929f05fc560e5ef9637b8f37cbcbd4a2e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-11-27 21:13:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/85/696385/1/upstream-solidfire-check/cinder-solidfire-iSCSI/7e61733/ : FAILURE in 57m 40s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"c8d917e7597fe67f5a38b44982a59a08de58c1b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-11-27 22:58:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/85/696385/1/check/tempest-dsvm-full-xenial/bb60e6a/ : SUCCESS in 2h 41m 59s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/85/696385/1/check/devstack-plugin-ceph-dsvm-xenial/c69a3c2/ : SUCCESS in 2h 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9d02e9866d163a257a332f238a9c30041ec56bf","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-11-27 23:04:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/85/696385/1/upstream-check/cinder-cDOT-iSCSI/1a88a21/ : FAILURE in 1h 03m 49s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/85/696385/1/upstream-check/cinder-cDOT-iSCSI-IPv6/81606ba/ : NOT_REGISTERED\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/85/696385/1/upstream-check/cinder-cDOT-NFS/01a7dd1/ : FAILURE in 55m 44s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/85/696385/1/upstream-check/cinder-cDOT-NFS-IPv6/e268518/ : FAILURE in 2h 48m 00s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/85/696385/1/upstream-check/cinder-cDOT-FCP/8fb3c1a/ : FAILURE in 1h 18m 59s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"8be4b93e3d294b82f56678e8d07fe780fb71e24d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-28 02:58:03.000000000","message":"Patch Set 1: Code-Review+2\n\nThis meets the backport criteria.  It addresses a real problem for FC users, and has no impact on others.","accounts_in_message":[],"_revision_number":1},{"id":"f9794d4b05efb44ce972ee698235f37fcb40c09d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-12-03 15:22:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b317de8e6278920920d39846840d03fbfa3d11bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 16:02:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eff1c12f161f9e9cf70fc338d6d2bf51662b78c8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-12-03 16:48:03.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"824995a439a93868d7c2cf892f3f2d7a90729fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:23:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/015fb13138b241218cc052ac8f9d6933 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf9036e60ac41bc9648da95f4b64273 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd49a42807474ae58273ca4be3078af3 : SUCCESS in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c67e80bebbe4e13879b1f1f4f3f7f21 : SUCCESS in 6m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc2c0342881e4a5784c6d8403e93a167 : SUCCESS in 1h 17m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0942d4679780474496e2a38cad6510f5 : SUCCESS in 4m 32s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/605121c372714c7dbf29a9c37871c15c : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a01c55341e1018fa50589ee0f706e6ceed9a7128","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-12-04 09:52:56.000000000","message":"Patch Set 1: Workflow+1\n\nRocky patch merged, so we can let this one merge as well.","accounts_in_message":[],"_revision_number":1},{"id":"35ea81f03c8eba991833bba035a429e6b3b9ca13","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-12-04 13:30:50.000000000","message":"Removed Workflow-1 by Eric Harney \u003ceharney@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"e5bfe62b554be4d6dba53bf5093942569869c498","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-12-04 13:43:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1016496ae0ad570ff71fa257edce222de11241c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 13:43:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27cfdd4579edbfbb75d5ff19b11e66f3d1463ce8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:22:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8d32f6ff6b7f4a228c4e7359e7116147 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5f8734b272d4178897e12940f70ce6c : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/885db9fd71184a7da2b216d918f5e4cf : SUCCESS in 8m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3215c41f1f4b4edf979ae99052ffe9d9 : SUCCESS in 5m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a9c8bc4dedb4142b974c7b10aa8e319 : SUCCESS in 1h 33m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72ffd2c3e7da42e4a9b802483c8affc4 : SUCCESS in 4m 25s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/aee07f63a6344bc29daacb2d657cf925 : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d5257cc63b79f29c1c9878002c7997eae2ebb498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:22:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dc70aabcc5e0f65713e86a1bc22e2d647326b486","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:23:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/149233d30eb64f62a750cd7d40b7abb2 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e46deeaa109f46686171e9af6ea12e94eb9e1f7","revisions":{"2e46deeaa109f46686171e9af6ea12e94eb9e1f7":{"kind":"REWORK","_number":1,"created":"2019-11-27 18:32:38.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/85/696385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/85/696385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/85/696385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/85/696385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/85/696385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/85/696385/1"}}},"commit":{"parents":[{"commit":"c4e616e3a07009143ed8c840ec38b7c937180d07","subject":"Handle None value \u0027inititator_target_map\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c4e616e3a07009143ed8c840ec38b7c937180d07"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-10-14 14:42:22.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-11-27 18:32:09.000000000","tz":60},"subject":"Fix FC scan too broad","message":"Fix FC scan too broad\n\nAfter some changes to the FC connector we have introduced a regression\non the way we do the scans, and we end up scanning using wildcards even\nthough we shouldn\u0027t.\n\nThe targets in the \"initiator_target_map\" don\u0027t mean that they are all\nconnected, so we must take that into account.\n\nWith the current code, if we have the following connections:\n\nHBA host7 ---- SWITCH ---- port W (channel 0, target 2)\n          \\--- SWITCH ---- port X (channel 0, target 3)\n\nHBA host8 ---- SWITCH ---- port Y (channel 0, target 2)\n          \\--- SWITCH ---- port Z (channel 0, target 3)\n\nWe will end up with the following scans 8 scans for LUN L:\n\n  - - L \u003e host7\n  - - L \u003e host7\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n  - - L \u003e host8\n  - - L \u003e host8\n\nWhich correspond to the responses from _get_hba_channel_scsi_target like\nthis:\n\n               port Y          port Z          port W          port X\n  host7 ... [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]\n  host8 ... [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]\n\nAnd we should only be doing 4 scans:\n\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n\nMost storage arrays get their target ports automatically detected by the\nLinux FC initiator and sysfs gets populated with that information, but\nthere are some that don\u0027t.\n\nWe\u0027ll do a narrow scan using the channel, target, and LUN for the former\nand a wider scan for the latter.\n\nIf all paths to a former type of array were down on the system boot the\narray could look like it\u0027s of the latter type and make us bring us\nunwanted volumes into the system by doing a broad scan.\n\nTo prevent this from happening Cinder drivers can use the\n\"enable_wildcard_scan\" key in the connection information to let us know\nthey don\u0027t want us to do broad scans even if no target ports are found\n(because they know the cause is there\u0027s no connection).\n\nClose-Bug: #1849504\nRelated-Bug: #1765000\nRelated-Bug: #1828440\nChange-Id: I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9\n(cherry picked from commit 708733e49598145eb8e898d486723dc3df00a52f)\n(cherry picked from commit c76b6dc514b8e17e1211e4a7465a7e2a635adea3)\n(cherry picked from commit 8933503041529978b9fd816e1788b1b0c343b540)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2e46deeaa109f46686171e9af6ea12e94eb9e1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2e46deeaa109f46686171e9af6ea12e94eb9e1f7"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
