)]}'
{"id":"openstack%2Fos-brick~697115","triplet_id":"openstack%2Fos-brick~stable%2Ftrain~Ib9c8ebae7a6051e18538920139fecd123682a474","project":"openstack/os-brick","branch":"stable/train","hashtags":[],"change_id":"Ib9c8ebae7a6051e18538920139fecd123682a474","subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","status":"MERGED","created":"2019-12-03 15:39:28.000000000","updated":"2019-12-09 23:13:28.000000000","submitted":"2019-12-09 23:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697115-1575933126040-7198a097","meta_rev_id":"09ffd2f86da6e2df5fd6147e6a03b480208ddb87","_number":697115,"virtual_id_number":697115,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-12-03 16:52:27.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-12-03 18:38:10.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-04 01:37:49.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 23:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-03 18:56:56.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-12-03 16:33:21.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-03 15:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-09 21:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-09 21:14:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 15:46:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-03 16:33:21.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-12-03 16:52:27.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 18:38:10.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 18:56:56.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-12-04 01:37:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-09 21:14:48.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-12-09 23:12:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be53cb8e81a17bb5e5c79e46b6a64764c250bf6e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-03 15:39:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2de3b10e0feaf9ec61baef47635124bc9d239202","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-03 15:46:32.000000000","message":"Patch Set 1: Code-Review+2\n\nGood to backport.","accounts_in_message":[],"_revision_number":1},{"id":"09dbf78a93cf6c608c33eb64d58c3544ec8ef98b","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-03 16:33:21.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697115/1/iscsi/ : FAILURE in 53m 33s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697115/1/smb/ : FAILURE in 52m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1d4e1b35c5a5684bb4e69ae2e4e16cabe88202d9","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-03 16:52:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/15/697115/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/b0ccfca : SUCCESS in 1h 01m 52s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/15/697115/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/c849023 : SUCCESS in 1h 12m 44s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"411d9e92575ad8c45a3c5a2ac581c7f3948fe1c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:45:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/46c595c3858f490d929fa16a8e9cc44b : SUCCESS in 1h 35m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dd10860c1d343dab3298d5c23b08941 : SUCCESS in 1h 58m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/747e5c5b473e4d08a1bf4b16ce92c403 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e0031c5eb5430a8447e88b5bb14b9e : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5c2f1c98d5e40139d799c67fa27ee26 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a1105edc8714844b24d4ad164915280 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35c61f54d6b44e6b81c9956edc30da5e : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8ba05ba06d412ea1a621eab57ef4d0 : SUCCESS in 7m 11s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/8f9c4713b6724724a3faa7963b5a7ef6 : SUCCESS in 1h 24m 15s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/9d711b0dfd5d40a8ae01575d07a06d90 : SUCCESS in 2h 01m 08s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/084a4746e73d49cebabcceb4f7a2ffea : SUCCESS in 1h 34m 06s","accounts_in_message":[],"_revision_number":1},{"id":"06e5900ab39aba97ce66c33bcee619d291f00b44","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-03 18:38:10.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/15/697115/1/check/tempest-dsvm-full-xenial/b223285/ : FAILURE in 2h 58m 07s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/15/697115/1/check/devstack-plugin-ceph-dsvm-xenial/057ca1f/ : SUCCESS in 2h 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d827b3f22e754f6a5f8a6def6745260c0ebc12ca","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-03 18:56:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/15/697115/1/upstream-solidfire-check/cinder-solidfire-iSCSI/3e7e0af/ : SUCCESS in 1h 11m 01s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"0852b1a97cab5b80c7d7e5fee023e812b38729c7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-04 01:37:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/15/697115/1/upstream-check/cinder-cDOT-iSCSI/22e524d/ : SUCCESS in 1h 00m 48s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/15/697115/1/upstream-check/cinder-cDOT-iSCSI-IPv6/88a9a63/ : FAILURE in 2h 31m 54s (non-voting)\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/15/697115/1/upstream-check/cinder-cDOT-NFS/0ec6ca3/ : SUCCESS in 58m 53s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/15/697115/1/upstream-check/cinder-cDOT-NFS-IPv6/e4de25a/ : SUCCESS in 2h 52m 06s (non-voting)\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/15/697115/1/upstream-check/cinder-cDOT-FCP/ea7d034/ : SUCCESS in 2h 04m 36s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fdde96a611e3fb59068415e9043cfc0059084189","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-09 21:14:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMeets backport criteria.","accounts_in_message":[],"_revision_number":1},{"id":"75b0c9c3cd785827a2da6c9d31307c1fe9346d58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 21:15:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3774f67f8a3b34b41dc83357b0165e3d07da24e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:12:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/84cc1c6d329a4a1fa28a70ecc6fe0a61 : SUCCESS in 1h 34m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d3715e7a1314dd19c2605d2bd131376 : SUCCESS in 1h 28m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04b41859421540af8c2d68db0db34635 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed8d126b10714eeca433cea19c77914c : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5189cea221834ea3845524245ec591d5 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/092cd4c504e742bfb22dabf56a10c645 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de6a22b6a7ae4b9a9676eb89ea2cad84 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4740319e4d34a808f6a3c2c9f82a3b2 : SUCCESS in 7m 37s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/896012baf49941b6afcb223b1557a8e8 : SUCCESS in 1h 27m 01s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/391b183a9d2a4dffb4fa5572fec38a58 : SUCCESS in 1h 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a2af7cfdbc6e83bd93b54965c278c1ea7768311f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:12:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"09ffd2f86da6e2df5fd6147e6a03b480208ddb87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:13:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5070dd62de534e9d872eb974afc60755 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8ed21e7d25f67c6d2bfe71515723f68ce15b887","revisions":{"d8ed21e7d25f67c6d2bfe71515723f68ce15b887":{"kind":"REWORK","_number":1,"created":"2019-12-03 15:39:28.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/15/697115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/15/697115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/15/697115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/15/697115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/15/697115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/15/697115/1"}}},"commit":{"parents":[{"commit":"c76b6dc514b8e17e1211e4a7465a7e2a635adea3","subject":"Fix FC scan too broad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c76b6dc514b8e17e1211e4a7465a7e2a635adea3"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-21 20:01:50.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-03 15:39:21.000000000","tz":0},"subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","message":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate\n\nBug #1820007 documents failures to find /dev/disk/by-id/ symlinks\nassociated with encrypted volumes both in real world and CI\nenvironments. These failures appear to be due to udev on these slow or\noverloaded hosts failing to populate the required /dev/disk/by-id/\nsymlinks in time after the iSCSI volume has been connected.\n\nThis change seeks to avoid such failures by simply decorating\n_get_device_link with the @utils.retry to hopefully allow udev time to\ncreate the required symlinks under /dev/disk/by-id/.\n\nCloses-Bug: #1820007\nChange-Id: Ib9c8ebae7a6051e18538920139fecd123682a474\n(cherry picked from commit 331316827a60aa50ee26be08a06027525f5194ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d8ed21e7d25f67c6d2bfe71515723f68ce15b887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d8ed21e7d25f67c6d2bfe71515723f68ce15b887"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
