)]}'
{"id":"openstack%2Fos-brick~697116","triplet_id":"openstack%2Fos-brick~stable%2Fstein~Ib9c8ebae7a6051e18538920139fecd123682a474","project":"openstack/os-brick","branch":"stable/stein","hashtags":[],"change_id":"Ib9c8ebae7a6051e18538920139fecd123682a474","subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","status":"MERGED","created":"2019-12-03 15:39:39.000000000","updated":"2019-12-12 15:53:41.000000000","submitted":"2019-12-12 15:51:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697116-1576165893727-6763ac6d","meta_rev_id":"bb2c142d69c70850f9b5762777191df1a05d16ae","_number":697116,"virtual_id_number":697116,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-12-11 14:49:22.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-12-12 07:40:48.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 15:51:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 16:09:35.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-12-11 14:29:27.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"date":"2019-12-11 13:37:52.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-12 01:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":2,"date":"2019-12-12 13:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":1,"date":"2019-12-12 13:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 15:40:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-11 14:29:27.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-12-11 14:49:22.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 16:09:35.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-12-12 01:38:23.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-12-12 07:40:48.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-12 13:59:30.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2019-12-12 15:51:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"101f8a8ff1612c17cf26a96af2f9c959646908c1","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-03 15:39:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf89a58d55bb2b7ca6a0c04cea77e9ee9e4c8503","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-03 15:40:18.000000000","message":"Patch Set 1: Code-Review-2\n\nNeeds to merge to later branches first.","accounts_in_message":[],"_revision_number":1},{"id":"0a490e2d5ac5fee122cd208544e4d7033707889c","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-03 16:27:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697116/1/iscsi/ : FAILURE in 47m 25s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697116/1/smb/ : FAILURE in 47m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8ed45c25ddb154f7da3b4ec31c8f76685e6957b5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-03 16:55:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/16/697116/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/b28258a : FAILURE in 1h 14m 40s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/16/697116/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/f4b5244 : SUCCESS in 1h 11m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"582eef8c36be3abc9afdc19f8fc2b1fbfacd0564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:21:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/462bc46063bf446bab5e0475ab48e7c7 : SUCCESS in 1h 33m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff343fefdd384f0785854f1c1be31bab : SUCCESS in 1h 09m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0148f9d4a9a45cfad21c45cf33461ca : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb71b600b1f4708911cb06153bb1c05 : POST_FAILURE in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/374a0b48768f4908add0c71d63196e6b : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9cd09d7ea644289bf86f0a117b2dbed : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dba90e37ff814ba2aec85b7537f6a85f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6275f9bf46e43aa8f1143cf63a47cb4 : SUCCESS in 5m 40s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/555e61f7539341b0bc4a270aaaeda734 : SUCCESS in 1h 28m 08s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/4837ee4153ab4af18859d99238d6b679 : SUCCESS in 1h 20m 29s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/93b011f067a44c14bb1951285185afb0 : SUCCESS in 1h 36m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ca425583286aa9b0f32c9db8cd91faf6c88bfc10","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-03 18:47:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/697116/1/check/tempest-dsvm-full-xenial/98a0ba9/ : SUCCESS in 2h 50m 00s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/697116/1/check/devstack-plugin-ceph-dsvm-xenial/7f606d0/ : FAILURE in 3h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffa0196293f23728030be45a12f259d530cb8da7","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-10 08:39:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"491a89317e52da8ad9cdbcb81c5ae7005871fc37","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-10 09:34:41.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697116/1/iscsi/ : FAILURE in 48m 50s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697116/1/smb/ : FAILURE in 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"99f48f7eff2fd084f978497639e0ead350917bf7","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-10 09:43:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/16/697116/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/910569f : SUCCESS in 54m 07s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/16/697116/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/dc72412 : SUCCESS in 1h 04m 26s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"a4322a621dc679b67d13d57e1dc7fd5152e37069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 10:24:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a48e0c6297443dd8147c2d5e0ac7df2 : SUCCESS in 1h 31m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fd23dd270554d30a5079db9c18371a4 : SUCCESS in 1h 27m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fff1e7b3724547b7ae3cceded8fbf17a : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731ea8bd7a0748798ff71f1c6d93707d : FAILURE in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e65ac80d8bf428abb9e6cc0606efb2a : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b9146e452048b98b24c853292483e7 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/053f6cf0698d4c49a941cffc83d9ac5d : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab66d27501a84a3b91177a75370f0a23 : SUCCESS in 3m 42s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/4024ba3c490a45a6952cf929a20f8114 : SUCCESS in 1h 17m 20s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/d7414386e8554ca89e192c740d496afe : SUCCESS in 1h 33m 54s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d4493e5d6c8f47e393c464b86ae02565 : SUCCESS in 1h 39m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0a704f73857526bb3df5c5a1367d3b70154692f9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-10 13:30:51.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/#/c/697115/ has merged to train","accounts_in_message":[],"_revision_number":1},{"id":"06d9714c2948519b40ed84be1969f05b58d66f05","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-10 13:48:22.000000000","message":"Removed Code-Review-2 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"dbebcf2ffd1a4506ff120e5aa9977062fbc5dc84","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-11 12:45:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/697116/1/check/tempest-dsvm-full-xenial/d771c5d/ : SUCCESS in 2h 46m 46s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/697116/1/check/devstack-plugin-ceph-dsvm-xenial/5b97857/ : SUCCESS in 2h 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb0423a071e16ea13e14a84bb8f3c43caaa44637","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-11 12:52:04.000000000","message":"Patch Set 1:\n\npep8 is failing on W503.  In master \u0026 train, in tox.ini we have:\nignore \u003d E251,W503,W504\nThe older stable branches have:\nignore \u003d E251\n(except ocata, which has ignore \u003d E123,E125,E251,E265,H405,H904)\n\nI think we should backport adding W503 and W504 to the ignore list, or we\u0027ll continue to have this problem with backports.","accounts_in_message":[],"_revision_number":1},{"id":"f7ab52ac70e48908fb94ac9ce878413a983ec354","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-11 12:56:00.000000000","message":"Patch Set 1:\n\nPatch backporting ignore of W503,W504 is https://review.opendev.org/#/c/698471/","accounts_in_message":[],"_revision_number":1},{"id":"108ed7c91c4c5c53a8368ad52fa9281949667a82","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-11 13:37:52.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"54b126ae1e5ddcee1db8fcfb754654e7ebc1bea2","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-11 14:29:27.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697116/2/iscsi/ : FAILURE in 47m 25s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697116/2/smb/ : FAILURE in 51m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5ce24dce53803260f093ab0d3354b22965b71acf","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-11 14:49:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/16/697116/2/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/98c5d49 : SUCCESS in 55m 57s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/16/697116/2/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/aa413c6 : SUCCESS in 1h 11m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"0fbf5aecc42bff25a8e8023f309e8537f4f06eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 14:59:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b093713434e04dc5b8ad8d6abec517bc : SUCCESS in 1h 19m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52f7e09943ec452d91cba437c6fe2738 : SUCCESS in 1h 18m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e8d08c7003b41c4bb05b6773b70d650 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394e936589e74051aa41800b36e7b771 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/310d356cc92f422f93129ac899a07dab : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f2b87b107ef444aae7e3d8f3ca2b648 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e2fb74f50004abbae2e436433f01745 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5752be4ea2c5435dbc2356033d32319b : SUCCESS in 3m 53s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/0341b412b2764547979b6288e0c44b83 : SUCCESS in 1h 08m 52s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/6c108a71133340cf8ac8aac735226ce2 : SUCCESS in 1h 17m 03s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc9632a5a8ad4f56aee738f5bc7487d4 : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6e55a67ebe19f202a01b734ad82541ff72cf47d7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-11 16:09:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/16/697116/2/upstream-solidfire-check/cinder-solidfire-iSCSI/d7e8558/ : SUCCESS in 1h 09m 20s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"4ac30ba6628c00972a78d65afaf8e1d25f3c1882","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-12 01:38:23.000000000","message":"Patch Set 2: Code-Review+2\n\nMakes sense to backport this.","accounts_in_message":[],"_revision_number":2},{"id":"8f99c54c73f02c33a05779bd321a214df00d62db","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-12 07:40:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/697116/2/check/tempest-dsvm-full-xenial/e9553c4/ : SUCCESS in 2h 47m 58s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/697116/2/check/devstack-plugin-ceph-dsvm-xenial/ef2b030/ : SUCCESS in 2h 51m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b1d3bf93ca607b03f80f7ee938bd72d3f3c4bef","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-12-12 13:59:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"773db5ed8630aad8d6e7fc04585b2ef16ba41d74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:59:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"305ea3e8690a6bacb1b22a437dbdfe598be0b23c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:51:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bbafa9c8b40a4687967720af946b498c : SUCCESS in 1h 46m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7394b239361b409699dacb16c270db1f : SUCCESS in 1h 21m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e106d30528924979b64ffc081ab03d85 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b388e411f774eedae5927867021bbe5 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/153c5c167387402bb14531c37c7df270 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d319404cf38455aaecc092d34ec2b4f : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4181df7939b14a4babeacc7810e121b4 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a07091d64394ac1b813835fe5f1c0d4 : SUCCESS in 6m 07s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/1a8b4dd6b3e64a66ba9ceb5695737a33 : SUCCESS in 1h 48m 04s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f7319e61a84b46fb8235741e287ab895 : SUCCESS in 1h 39m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c27dcdcc7c2df034de3ffd675514b805618dadff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:51:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bb2c142d69c70850f9b5762777191df1a05d16ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:53:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650eda861a3f4409854caadf7d6d783d : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1f84dfd897d35053d39c22ae484f6586d1268bd","revisions":{"7ffa3fb7d9d497fd3848ab0313ba8aa84136a06e":{"kind":"REWORK","_number":1,"created":"2019-12-03 15:39:39.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/16/697116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/16/697116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/697116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/697116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/697116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/16/697116/1"}}},"commit":{"parents":[{"commit":"15160ab610680f78617166e11fe04337de6ee847","subject":"Merge \"Remove VxFlexOS connector external dependencies\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/15160ab610680f78617166e11fe04337de6ee847"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-21 20:01:50.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-03 15:39:33.000000000","tz":0},"subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","message":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate\n\nBug #1820007 documents failures to find /dev/disk/by-id/ symlinks\nassociated with encrypted volumes both in real world and CI\nenvironments. These failures appear to be due to udev on these slow or\noverloaded hosts failing to populate the required /dev/disk/by-id/\nsymlinks in time after the iSCSI volume has been connected.\n\nThis change seeks to avoid such failures by simply decorating\n_get_device_link with the @utils.retry to hopefully allow udev time to\ncreate the required symlinks under /dev/disk/by-id/.\n\nCloses-Bug: #1820007\nChange-Id: Ib9c8ebae7a6051e18538920139fecd123682a474\n(cherry picked from commit 331316827a60aa50ee26be08a06027525f5194ef)\n(cherry picked from commit d8ed21e7d25f67c6d2bfe71515723f68ce15b887)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7ffa3fb7d9d497fd3848ab0313ba8aa84136a06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7ffa3fb7d9d497fd3848ab0313ba8aa84136a06e"}]},"branch":"refs/heads/stable/stein"},"d1f84dfd897d35053d39c22ae484f6586d1268bd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-11 13:37:52.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/16/697116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/16/697116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/697116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/697116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/697116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/16/697116/2"}}},"commit":{"parents":[{"commit":"61346f10029fd38613e5ae7593660405ecbac71f","subject":"Ignore pep8 W503/W504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/61346f10029fd38613e5ae7593660405ecbac71f"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-21 20:01:50.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-11 13:37:21.000000000","tz":0},"subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","message":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate\n\nBug #1820007 documents failures to find /dev/disk/by-id/ symlinks\nassociated with encrypted volumes both in real world and CI\nenvironments. These failures appear to be due to udev on these slow or\noverloaded hosts failing to populate the required /dev/disk/by-id/\nsymlinks in time after the iSCSI volume has been connected.\n\nThis change seeks to avoid such failures by simply decorating\n_get_device_link with the @utils.retry to hopefully allow udev time to\ncreate the required symlinks under /dev/disk/by-id/.\n\nCloses-Bug: #1820007\nChange-Id: Ib9c8ebae7a6051e18538920139fecd123682a474\n(cherry picked from commit 331316827a60aa50ee26be08a06027525f5194ef)\n(cherry picked from commit d8ed21e7d25f67c6d2bfe71515723f68ce15b887)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d1f84dfd897d35053d39c22ae484f6586d1268bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d1f84dfd897d35053d39c22ae484f6586d1268bd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
