)]}'
{"id":"openstack%2Fos-brick~697117","triplet_id":"openstack%2Fos-brick~stable%2Frocky~Ib9c8ebae7a6051e18538920139fecd123682a474","project":"openstack/os-brick","branch":"stable/rocky","hashtags":[],"change_id":"Ib9c8ebae7a6051e18538920139fecd123682a474","subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","status":"MERGED","created":"2019-12-03 15:39:58.000000000","updated":"2019-12-14 16:59:29.000000000","submitted":"2019-12-14 16:59:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697117-1576342769774-bbef5f2b","meta_rev_id":"a711cdc91c1a2b533f8978074589ce7fb38e11ff","_number":697117,"virtual_id_number":697117,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-12-14 14:15:13.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-12-14 15:29:21.000000000","_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-14 16:42:08.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-14 13:52:34.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2019-12-14 05:51:16.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-14 16:59:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-14 16:31:03.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-12-14 16:06:37.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-13 21:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-13 21:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-12-13 21:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 21:48:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-12-13 21:55:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-14 05:51:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-14 14:15:13.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-14 15:29:21.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-14 16:06:37.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-12-14 16:31:03.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-12-14 16:42:08.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-14 16:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3343a57d5dc65f490804467acba2ea4c1f877a80","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-03 15:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0165b57dd3ab26492d353818cc3648586de08264","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-03 15:40:28.000000000","message":"Patch Set 1: Code-Review-2\n\nNeeds to merge to later branches first.","accounts_in_message":[],"_revision_number":1},{"id":"b6807d6397069a306f6eb650887c4529ea5b8151","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-03 17:05:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697117/1/iscsi/ : FAILURE in 37m 45s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697117/1/smb/ : FAILURE in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6c1880e0fd36c5856ce3678e08192d9a708af949","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-03 17:20:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/17/697117/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/ce1b510 : FAILURE in 21m 59s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/17/697117/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/fdd5680 : FAILURE in 24m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"ffacb4acdf3f0d33bacddd7218753986acb27e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:22:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d31c4abdfd6f4450946e102b70f370d8 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d950a0bdfccb4e6897d2cddeb0826e57 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9398d28e69b546d8a2b7620e484bba5e : SUCCESS in 4m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/34590497c06744dfbe603a4355af28a2 : SUCCESS in 5m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/454efce82c9e42f098e610a5fffcc4a5 : SUCCESS in 1h 26m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6532f5ad0b354e908bf82da2e24cfb76 : SUCCESS in 6m 01s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/2299e99fb4114517aa167edd59b5cc23 : SUCCESS in 1h 15m 28s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/7a5f473feb744f49b9638b8e48be6cf2 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a2e9fa8eda57b69e972afa0eae4d1813b19854b0","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-12-03 17:24:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/17/697117/1/upstream-eseries-check/cinder-eseries-iSCSI/d864259/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/17/697117/1/upstream-eseries-check/cinder-eseries-FCP/99a6357/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":1},{"id":"15d19202bc7209624c646d1c28508e222ca60d20","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-03 18:28:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/1/check/tempest-dsvm-full-xenial/626e1e6/ : SUCCESS in 2h 42m 15s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/1/check/devstack-plugin-ceph-dsvm-xenial/adda7a6/ : FAILURE in 2h 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f99aebbfb22b7c05b3a756208c1554cbf1d75708","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-03 18:32:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/17/697117/1/upstream-solidfire-check/cinder-solidfire-iSCSI/dadec59/ : SUCCESS in 1h 07m 37s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"d8f8a4a4cdc06e7cee63ea3e8a998988c50c11d7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-04 01:25:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/17/697117/1/upstream-check/cinder-cDOT-iSCSI/4091ed9/ : SUCCESS in 54m 06s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/17/697117/1/upstream-check/cinder-cDOT-iSCSI-IPv6/0d2e5b0/ : FAILURE in 2h 11m 27s (non-voting)\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/17/697117/1/upstream-check/cinder-cDOT-NFS/4bfda77/ : SUCCESS in 57m 03s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/17/697117/1/upstream-check/cinder-cDOT-NFS-IPv6/95ce910/ : SUCCESS in 3h 05m 40s (non-voting)\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/17/697117/1/upstream-check/cinder-cDOT-FCP/b4e76ab/ : SUCCESS in 1h 42m 09s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"410dde874008afb3f5c5bfa31921c3d78d3ee236","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-11 17:33:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"88ee3fc69ed37b79d6099254f172d15981579e95","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-11 17:36:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b7ddcee7212a77d8b625b4b5a14d951c8e1bc611","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-11 18:17:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/17/697117/3/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/6c4ed64 : FAILURE in 21m 05s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/17/697117/3/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/1979e89 : FAILURE in 23m 39s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"4cb9fc424b05d32851ac32c98f57ccb8aa59fb48","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-11 19:01:49.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697117/3/iscsi/ : FAILURE in 42m 12s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697117/3/smb/ : FAILURE in 37m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4a85356da6bcc5b24b855a197afb871c0466368d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:12:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f280405f258544dc801ac53b27bcc3c2 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85ca8236a6443619baf274863a2c81f : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea3c101963e0427a966773b693171024 : SUCCESS in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d486dd8cb5ae4b6ba382f542214c5a3a : SUCCESS in 4m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a0abaf304e040ed995b7d4acdb6496b : SUCCESS in 1h 29m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/841e8014dc004e8cb0586fe89ed1bc72 : SUCCESS in 8m 00s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/22d99904df7a4808a5143cabb81777ed : SUCCESS in 1h 21m 47s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/8192ce1653b4456fb7302415683e0f21 : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":3},{"id":"1c73f89cc3020eb32cfef273e7ff6b48c2eedebe","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-12-11 19:13:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/17/697117/3/upstream-eseries-check/cinder-eseries-iSCSI/583976c/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/17/697117/3/upstream-eseries-check/cinder-eseries-FCP/61e0ce5/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":3},{"id":"b70d7287e7da1f2d74fa5eaf001b3880e68fffe2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-11 20:23:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/17/697117/3/upstream-solidfire-check/cinder-solidfire-iSCSI/40dbcb3/ : SUCCESS in 1h 09m 32s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"0acc6a5b563ce53f1d675f74c4abf4992f5c37eb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-11 23:05:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/17/697117/3/upstream-check/cinder-cDOT-iSCSI/c013384/ : SUCCESS in 57m 15s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/17/697117/3/upstream-check/cinder-cDOT-iSCSI-IPv6/9606c9b/ : FAILURE in 3h 51m 11s (non-voting)\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/17/697117/3/upstream-check/cinder-cDOT-NFS/61d62f9/ : SUCCESS in 58m 40s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/17/697117/3/upstream-check/cinder-cDOT-NFS-IPv6/5716162/ : FAILURE in 3h 50m 04s (non-voting)\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/17/697117/3/upstream-check/cinder-cDOT-FCP/27b2072/ : FAILURE in 34m 08s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"954d82a0cc33398bed846c93598cbfae08a3ffc1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-12 10:13:48.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/3/check/tempest-dsvm-full-xenial/74e3b13/ : FAILURE in 2h 58m 57s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/3/check/devstack-plugin-ceph-dsvm-xenial/8ed5187/ : SUCCESS in 2h 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4e145eac05778b8af5082d94b02c8c22582c602","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-12 16:28:31.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nChange has merged into stable/stein","accounts_in_message":[],"_revision_number":3},{"id":"e0a4993a112ac189e2b488183a7af829a56aa1a0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-12 17:42:48.000000000","message":"Patch Set 3: -Code-Review\n\nAnother reason to not do these backports until they are ready. That does mess with some tooling, but nothing to critical. Would be good to fix if you can.","accounts_in_message":[],"_revision_number":3},{"id":"5c22171ab8c1217b775c00c2876beceebefbd4fd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 17:39:06.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"adc14cb07aa6dae91e04309436f7bb5ff930f62f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-13 17:59:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/17/697117/4/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/85c283c : FAILURE in 19m 43s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/17/697117/4/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/26fe403 : FAILURE in 19m 56s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"c90f2202b99e690d2a605c49905fcbcc1d60ab2e","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-13 18:56:48.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697117/4/iscsi/ : FAILURE in 47m 05s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697117/4/smb/ : FAILURE in 44m 51s","accounts_in_message":[],"_revision_number":4},{"id":"e06b541af23bde5ef97cf8085021e10aef5da52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 19:19:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0f29584f7d114129a8d2c97c49871fab : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5213b3b2727e4f5c8c7fb9606edfe17f : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68a5173167824b11a2b313aaeaae19e7 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b3a7bdbf52ae4016928c62176b7bc9a4 : SUCCESS in 4m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d567decf6b524fa4aabcd6f8f1e90517 : SUCCESS in 1h 29m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a721170492f6405481c810fa57825bba : SUCCESS in 6m 25s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/450aefca58f24388b3de4167c4e88474 : SUCCESS in 1h 38m 10s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/f42a0ee8a93247e180ca1be8e1c49df1 : SUCCESS in 1h 32m 13s","accounts_in_message":[],"_revision_number":4},{"id":"7067c69c224f83e528906ea20970ea8ead9d4949","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 21:48:07.000000000","message":"Patch Set 4: Code-Review+2\n\nAll I did was fix the stein hash on the commit message, so I think it\u0027s OK to vote on it.  This change meets backport criteria and it\u0027s a clean backport into rocky.","accounts_in_message":[],"_revision_number":4},{"id":"d861f076ae65d779e3f2caf86eea14580075c20f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 21:55:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"60430cfbf93cbe1d2c3e686d52944d4a85ec134d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 21:55:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b48b310d2199d42ed8a7d2bad95c719b4544aac7","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-12-13 22:18:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/17/697117/4/upstream-eseries-check/cinder-eseries-iSCSI/712ae43/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/17/697117/4/upstream-eseries-check/cinder-eseries-FCP/40ff54f/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":4},{"id":"6bf19adf806c66a76d5030e441114c02c568290f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-13 23:26:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/17/697117/4/upstream-solidfire-check/cinder-solidfire-iSCSI/708b51f/ : SUCCESS in 1h 08m 31s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"40ed49cedca0f27da24d70892402c5e7f82bcfde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 23:43:46.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b224b10447f246c8a7d8e4162e8e725f : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6325c9ebcfa34c78875c7c1fd65589f0 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cb6f11006ba45c982003ac8e5160bfc : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6d7e1e8354254c71acb79a0647e5df76 : SUCCESS in 4m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a823b0a37d034667a96f979359a8cf8a : SUCCESS in 1h 39m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20a47987940c4d1f93365356643453c4 : SUCCESS in 6m 58s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/91ba449276ae4575966996c20e4ea4bb : FAILURE in 1h 25m 46s","accounts_in_message":[],"_revision_number":4},{"id":"a7a25fe700df364642805ffd6cb8b02d38cc64f7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-14 05:51:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/17/697117/4/upstream-check/cinder-cDOT-iSCSI/5aa2ac6/ : SUCCESS in 59m 22s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/17/697117/4/upstream-check/cinder-cDOT-iSCSI-IPv6/0269737/ : FAILURE in 3h 51m 58s (non-voting)\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/17/697117/4/upstream-check/cinder-cDOT-NFS/45a26a6/ : SUCCESS in 58m 23s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/17/697117/4/upstream-check/cinder-cDOT-NFS-IPv6/997f52d/ : FAILURE in 3h 51m 25s (non-voting)\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/17/697117/4/upstream-check/cinder-cDOT-FCP/07820c5/ : FAILURE in 1h 56m 22s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"84de8e9de1ba7c28a724e65c9ae370f54acb9b90","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-14 10:55:00.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/4/check/tempest-dsvm-full-xenial/31e8728/ : FAILURE in 2h 50m 01s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/4/check/devstack-plugin-ceph-dsvm-xenial/97019d4/ : FAILURE in 2h 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ebf9c38d0ef1cbbf5a034b220e2a801b9cb769fe","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-14 13:52:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"74398063328996509b7affd168c79b05987786d0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-14 14:15:13.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/17/697117/4/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/564558a : FAILURE in 22m 23s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/17/697117/4/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/8a45265 : FAILURE in 21m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"9fa98d79219a95156adde515a471ffe13647ef1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 15:28:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/abbed31b4edc443686b24036cecc1ef0 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415cd745a4954dcabadd7fa6a98e4cd0 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9360acb0ddc04349bf456609b2ccdc1c : SUCCESS in 3m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/78d26796cae64e00b3ce7e7a51cf7126 : SUCCESS in 4m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5eac987b51d24359bc557c5bf0d153bf : SUCCESS in 1h 30m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07752d3845e84bfd970196d67187f938 : SUCCESS in 5m 53s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/02c8a3a9e5b34a94b6377865f450f621 : SUCCESS in 1h 10m 28s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/8b3a1f99670e4bd3b7527c805602127c : SUCCESS in 1h 27m 41s","accounts_in_message":[],"_revision_number":4},{"id":"5312a82915c49239a8bd0ad1c6ca522d614dc9ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 15:28:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"af5abf6a78a11273b7d35bfa5f3664ae89083461","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-12-14 15:29:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/17/697117/4/upstream-eseries-check/cinder-eseries-iSCSI/31c8e59/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/17/697117/4/upstream-eseries-check/cinder-eseries-FCP/66efb29/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":4},{"id":"e839cab500b1bc840f8b988cefed29f62ba726a2","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-14 16:06:37.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697117/4/iscsi/ : FAILURE in 2h 13m 36s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697117/4/smb/ : FAILURE in 2h 00m 34s","accounts_in_message":[],"_revision_number":4},{"id":"297b7889745ebd95797041d6c6ba6d0bf8db9028","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-14 16:31:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/17/697117/4/upstream-solidfire-check/cinder-solidfire-iSCSI/6cbdf52/ : FAILURE in 1h 01m 41s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"f6dbd3d517f3f4dc57f588e99186b303211a3a9b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-14 16:42:08.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/4/check/tempest-dsvm-full-xenial/fd23590/ : SUCCESS in 2h 44m 03s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/697117/4/check/devstack-plugin-ceph-dsvm-xenial/9afbdf4/ : SUCCESS in 2h 48m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d4db4671c633eeba061fe92e892c26db375e35a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 16:59:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a711cdc91c1a2b533f8978074589ce7fb38e11ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 16:59:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8be7aa64a7564718b7a08ef46981efad : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e0d2de3a7d448af9c0ca347f763f442 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62ceb563581a47ed97557603849abb61 : SUCCESS in 4m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f404aabffb554fba95ded28904533109 : SUCCESS in 3m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/38ac6a83d12c434a85e9defd5bf673c7 : SUCCESS in 1h 25m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66ea2da621974d6989bbf278e7cc2c26 : SUCCESS in 6m 18s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/4a3e861d592a437993a131b22cefeddd : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1f811c412a172d5456d37cc748bc217fc4718cbf","revisions":{"ed9feedfa3f9a434a0bbd48953262fbb2999465b":{"kind":"REWORK","_number":1,"created":"2019-12-03 15:39:58.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/17/697117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/697117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/697117/1"}}},"commit":{"parents":[{"commit":"22765cb4aac2c960b86399199dd95bb644bad717","subject":"Check path alive before get scsi wwn","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/22765cb4aac2c960b86399199dd95bb644bad717"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-21 20:01:50.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-03 15:39:45.000000000","tz":0},"subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","message":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate\n\nBug #1820007 documents failures to find /dev/disk/by-id/ symlinks\nassociated with encrypted volumes both in real world and CI\nenvironments. These failures appear to be due to udev on these slow or\noverloaded hosts failing to populate the required /dev/disk/by-id/\nsymlinks in time after the iSCSI volume has been connected.\n\nThis change seeks to avoid such failures by simply decorating\n_get_device_link with the @utils.retry to hopefully allow udev time to\ncreate the required symlinks under /dev/disk/by-id/.\n\nCloses-Bug: #1820007\nChange-Id: Ib9c8ebae7a6051e18538920139fecd123682a474\n(cherry picked from commit 331316827a60aa50ee26be08a06027525f5194ef)\n(cherry picked from commit d8ed21e7d25f67c6d2bfe71515723f68ce15b887)\n(cherry picked from commit 7ffa3fb7d9d497fd3848ab0313ba8aa84136a06e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ed9feedfa3f9a434a0bbd48953262fbb2999465b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ed9feedfa3f9a434a0bbd48953262fbb2999465b"}]},"branch":"refs/heads/stable/rocky"},"0b764328728a4b6057377abc00c26fcfda5e16a5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-11 17:33:50.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/17/697117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/697117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/697117/2"}}},"commit":{"parents":[{"commit":"6cbcda0c27cfea1d6d4eafb6445e35dcb3b1bcb3","subject":"Ignore pep8 W503/W504","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6cbcda0c27cfea1d6d4eafb6445e35dcb3b1bcb3"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-21 20:01:50.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-11 17:33:33.000000000","tz":0},"subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","message":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate\n\nBug #1820007 documents failures to find /dev/disk/by-id/ symlinks\nassociated with encrypted volumes both in real world and CI\nenvironments. These failures appear to be due to udev on these slow or\noverloaded hosts failing to populate the required /dev/disk/by-id/\nsymlinks in time after the iSCSI volume has been connected.\n\nThis change seeks to avoid such failures by simply decorating\n_get_device_link with the @utils.retry to hopefully allow udev time to\ncreate the required symlinks under /dev/disk/by-id/.\n\nCloses-Bug: #1820007\nChange-Id: Ib9c8ebae7a6051e18538920139fecd123682a474\n(cherry picked from commit 331316827a60aa50ee26be08a06027525f5194ef)\n(cherry picked from commit d8ed21e7d25f67c6d2bfe71515723f68ce15b887)\n(cherry picked from commit 7ffa3fb7d9d497fd3848ab0313ba8aa84136a06e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0b764328728a4b6057377abc00c26fcfda5e16a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0b764328728a4b6057377abc00c26fcfda5e16a5"}]},"branch":"refs/heads/stable/rocky"},"eb480a04837307648ac0ba03ee175bc126bbc53a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-11 17:36:43.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/17/697117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/697117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/697117/3"}}},"commit":{"parents":[{"commit":"34a3ca23b3184e9b356877432d0f44c836705a70","subject":"Fix FC scan too broad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/34a3ca23b3184e9b356877432d0f44c836705a70"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-21 20:01:50.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-11 17:36:36.000000000","tz":0},"subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","message":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate\n\nBug #1820007 documents failures to find /dev/disk/by-id/ symlinks\nassociated with encrypted volumes both in real world and CI\nenvironments. These failures appear to be due to udev on these slow or\noverloaded hosts failing to populate the required /dev/disk/by-id/\nsymlinks in time after the iSCSI volume has been connected.\n\nThis change seeks to avoid such failures by simply decorating\n_get_device_link with the @utils.retry to hopefully allow udev time to\ncreate the required symlinks under /dev/disk/by-id/.\n\nCloses-Bug: #1820007\nChange-Id: Ib9c8ebae7a6051e18538920139fecd123682a474\n(cherry picked from commit 331316827a60aa50ee26be08a06027525f5194ef)\n(cherry picked from commit d8ed21e7d25f67c6d2bfe71515723f68ce15b887)\n(cherry picked from commit 7ffa3fb7d9d497fd3848ab0313ba8aa84136a06e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/eb480a04837307648ac0ba03ee175bc126bbc53a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/eb480a04837307648ac0ba03ee175bc126bbc53a"}]},"branch":"refs/heads/stable/rocky"},"1f811c412a172d5456d37cc748bc217fc4718cbf":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-12-13 17:39:06.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/17/697117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/697117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/697117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/697117/4"}}},"commit":{"parents":[{"commit":"34a3ca23b3184e9b356877432d0f44c836705a70","subject":"Fix FC scan too broad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/34a3ca23b3184e9b356877432d0f44c836705a70"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-21 20:01:50.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-12-13 17:38:45.000000000","tz":0},"subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","message":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate\n\nBug #1820007 documents failures to find /dev/disk/by-id/ symlinks\nassociated with encrypted volumes both in real world and CI\nenvironments. These failures appear to be due to udev on these slow or\noverloaded hosts failing to populate the required /dev/disk/by-id/\nsymlinks in time after the iSCSI volume has been connected.\n\nThis change seeks to avoid such failures by simply decorating\n_get_device_link with the @utils.retry to hopefully allow udev time to\ncreate the required symlinks under /dev/disk/by-id/.\n\nCloses-Bug: #1820007\nChange-Id: Ib9c8ebae7a6051e18538920139fecd123682a474\n(cherry picked from commit 331316827a60aa50ee26be08a06027525f5194ef)\n(cherry picked from commit d8ed21e7d25f67c6d2bfe71515723f68ce15b887)\n(cherry picked from commit d1f84dfd897d35053d39c22ae484f6586d1268bd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1f811c412a172d5456d37cc748bc217fc4718cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1f811c412a172d5456d37cc748bc217fc4718cbf"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
