)]}'
{"id":"openstack%2Fos-brick~697133","triplet_id":"openstack%2Fos-brick~stable%2Frocky~I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9","project":"openstack/os-brick","branch":"stable/rocky","topic":"bug/1849504","hashtags":[],"change_id":"I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9","subject":"Fix FC scan too broad","status":"MERGED","created":"2019-12-03 16:46:17.000000000","updated":"2019-12-04 04:54:30.000000000","submitted":"2019-12-04 04:53:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697133-1575435205217-aa01f648","meta_rev_id":"3eae58dbb63da09f208b5d6e0e8eced99247018f","_number":697133,"virtual_id_number":697133,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-12-03 17:38:20.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-12-03 18:26:39.000000000","_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-03 19:38:51.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"date":"2019-12-03 23:33:46.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 04:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-03 19:34:19.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2019-12-03 18:19:12.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-03 16:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-04 03:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-04 03:17:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-03 16:48:36.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 16:48:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-03 16:48:36.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-12-03 17:38:20.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 18:19:12.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2019-12-03 18:26:39.000000000","updated_by":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 19:34:19.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2019-12-03 19:38:51.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 23:33:46.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 03:17:31.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-12-04 04:53:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d6004d6465e9d9c0b3603bb8bf3faae0ca6960a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2019-12-03 16:46:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4703e5e8eb7ec1f82a76fd6a5e778d550e6e23ed","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-03 16:48:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5f0e700c662a34fbc97fca2b61375e80bcc5577","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-03 17:38:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/33/697133/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/3750069 : FAILURE in 22m 55s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/33/697133/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/36e52f4 : FAILURE in 22m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"cabb3c424a9fcf4341f2f53a530177a658964fd3","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-03 18:19:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/697133/1/iscsi/ : FAILURE in 37m 05s\n- os-brick-smb http://cloudbase-ci.com/os-brick/697133/1/smb/ : FAILURE in 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"426bcccf94c97c6ffb07bbcc1f0caa5a79d08258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:24:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/75ece3bb187b41a39bac396f5e96c2e7 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9186c6cc51147008f5b4de78e0f4a31 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3770b0f8ecc54c73af39028500f38b4c : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e31e7b87463b4b5d8a821499e6fbf83e : SUCCESS in 5m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bee6bce984cf4be883dcda73b66ee083 : SUCCESS in 1h 30m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71369e0b014645e982dffb66d6198fcc : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d3778d0f73549df928d3d2ec0d90fec : SUCCESS in 6m 27s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/df61562797f34ff29e6d16434b169b93 : SUCCESS in 1h 06m 56s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/3d0108ec339d4679bbff25ad54e1897d : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5acbb5feaced2d01a82983f6c866320ab18456a3","author":{"_account_id":24230,"name":"NetApp E-Series CI","email":"ng-openstack-eseries-ci@netapp.com","username":"netapp-eseries-ci","tags":["SERVICE_USER"]},"date":"2019-12-03 18:26:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-eseries-iSCSI http://logs.openstack.netapp.com/logs/33/697133/1/upstream-eseries-check/cinder-eseries-iSCSI/3559711/ : NOT_REGISTERED (non-voting)\n- cinder-eseries-FCP http://logs.openstack.netapp.com/logs/33/697133/1/upstream-eseries-check/cinder-eseries-FCP/fc4e6f8/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-netapp-eseries-ci\u0027 or \u0027netapp-eseries-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_Eseries_CI","accounts_in_message":[],"_revision_number":1},{"id":"3c8ab72ca8c9398f5c2adf4939a875603fabc377","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-03 19:34:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/33/697133/1/upstream-solidfire-check/cinder-solidfire-iSCSI/8d77ebe/ : SUCCESS in 1h 00m 27s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"950ba1a51ad76327f7b1e77cc2ec79a868293cae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-03 19:38:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/33/697133/1/check/tempest-dsvm-full-xenial/b67b6ee/ : SUCCESS in 2h 45m 09s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/33/697133/1/check/devstack-plugin-ceph-dsvm-xenial/0da2576/ : SUCCESS in 2h 38m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0225eb724fdbdf96efb52590999c78865d0cfd0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-03 23:33:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/33/697133/1/upstream-check/cinder-cDOT-iSCSI/ec6a4c8/ : SUCCESS in 58m 43s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/33/697133/1/upstream-check/cinder-cDOT-iSCSI-IPv6/293fb48/ : FAILURE in 2h 33m 20s (non-voting)\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/33/697133/1/upstream-check/cinder-cDOT-NFS/5dddc48/ : SUCCESS in 50m 46s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/33/697133/1/upstream-check/cinder-cDOT-NFS-IPv6/a0e9ca7/ : FAILURE in 3h 02m 18s (non-voting)\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/33/697133/1/upstream-check/cinder-cDOT-FCP/93fbaf8/ : SUCCESS in 1h 46m 46s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"edfa9fdf8501882224202bce996073a98a7f35a7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-04 03:17:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70274f01e7f0477e8a26a054e73345e10f29f5cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 03:17:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3efe3cfc76952da53b450058205e3477daa01128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 04:53:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c59950d6e16a42a6ba75746a20ceaf85 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4c0a5bada14558b2c5110f0ce77f13 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f92aa21901694addafc3a92b094d2e5a : SUCCESS in 3m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/249674353dae42db9f4d0b03ac804fea : SUCCESS in 4m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1ac137c2bbbb472bbe9958aa934ad631 : SUCCESS in 1h 34m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1165c7b70e924db39b5cadd8c486298d : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d5d575ca63f49d780b636e3abd467b9 : SUCCESS in 6m 25s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/ab4d2335a1bc4b5cadbc1155796a3332 : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0a1deba49d031b1b6bda2a9edaef29410958f887","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 04:53:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b6ba20e8c7416ae3008b2d8df1457a5177c81b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 04:54:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58626ee4108d41a1ad6c73f8c917f050 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34a3ca23b3184e9b356877432d0f44c836705a70","revisions":{"34a3ca23b3184e9b356877432d0f44c836705a70":{"kind":"REWORK","_number":1,"created":"2019-12-03 16:46:17.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/33/697133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/33/697133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/697133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/697133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/33/697133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/33/697133/1"}}},"commit":{"parents":[{"commit":"a15ede0f06ba441b3480354fb01191cde7a358a3","subject":"Handle None value \u0027inititator_target_map\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a15ede0f06ba441b3480354fb01191cde7a358a3"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-10-14 14:42:22.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2019-12-03 16:45:04.000000000","tz":60},"subject":"Fix FC scan too broad","message":"Fix FC scan too broad\n\nAfter some changes to the FC connector we have introduced a regression\non the way we do the scans, and we end up scanning using wildcards even\nthough we shouldn\u0027t.\n\nThe targets in the \"initiator_target_map\" don\u0027t mean that they are all\nconnected, so we must take that into account.\n\nWith the current code, if we have the following connections:\n\nHBA host7 ---- SWITCH ---- port W (channel 0, target 2)\n          \\--- SWITCH ---- port X (channel 0, target 3)\n\nHBA host8 ---- SWITCH ---- port Y (channel 0, target 2)\n          \\--- SWITCH ---- port Z (channel 0, target 3)\n\nWe will end up with the following scans 8 scans for LUN L:\n\n  - - L \u003e host7\n  - - L \u003e host7\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n  - - L \u003e host8\n  - - L \u003e host8\n\nWhich correspond to the responses from _get_hba_channel_scsi_target like\nthis:\n\n               port Y          port Z          port W          port X\n  host7 ... [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]\n  host8 ... [\u00270\u0027,\u00272\u0027,L]     [\u00270\u0027,\u00273\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]     [\u0027-\u0027,\u0027-\u0027,L]\n\nAnd we should only be doing 4 scans:\n\n  0 2 L \u003e host7\n  0 3 L \u003e host7\n  0 2 L \u003e host8\n  0 3 L \u003e host8\n\nMost storage arrays get their target ports automatically detected by the\nLinux FC initiator and sysfs gets populated with that information, but\nthere are some that don\u0027t.\n\nWe\u0027ll do a narrow scan using the channel, target, and LUN for the former\nand a wider scan for the latter.\n\nIf all paths to a former type of array were down on the system boot the\narray could look like it\u0027s of the latter type and make us bring us\nunwanted volumes into the system by doing a broad scan.\n\nTo prevent this from happening Cinder drivers can use the\n\"enable_wildcard_scan\" key in the connection information to let us know\nthey don\u0027t want us to do broad scans even if no target ports are found\n(because they know the cause is there\u0027s no connection).\n\nClose-Bug: #1849504\nRelated-Bug: #1765000\nRelated-Bug: #1828440\nChange-Id: I5dbefaff43fb902b15117b443fc92f7b6a6ad8c9\n(cherry picked from commit 708733e49598145eb8e898d486723dc3df00a52f)\n(cherry picked from commit c76b6dc514b8e17e1211e4a7465a7e2a635adea3)\n(cherry picked from commit 8933503041529978b9fd816e1788b1b0c343b540)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/34a3ca23b3184e9b356877432d0f44c836705a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/34a3ca23b3184e9b356877432d0f44c836705a70"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
