)]}'
{"id":"openstack%2Fos-brick~700265","triplet_id":"openstack%2Fos-brick~master~Ib13a3fe3f1a103b5e24a7c5bb7838ffdae2bf198","project":"openstack/os-brick","branch":"master","topic":"hacking","hashtags":[],"change_id":"Ib13a3fe3f1a103b5e24a7c5bb7838ffdae2bf198","subject":"Raise hacking version to 2.0.0","status":"MERGED","created":"2019-12-21 09:11:39.000000000","updated":"2020-02-21 12:27:14.000000000","submitted":"2020-02-21 12:24:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700265-1582287890922-0cc33de8","meta_rev_id":"8d11dc97147652dcd1e272ee1347ea2f2b72685c","_number":700265,"virtual_id_number":700265,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-20 14:50:15.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-21 05:50:04.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-02-21 12:15:42.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-21 11:03:13.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-21 15:32:19.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-21 12:24:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-21 01:09:52.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-02-21 11:55:36.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-21 01:24:00.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-21 09:30:11.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-20 22:57:42.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-20 23:17:53.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-02-21 10:16:55.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-19 13:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-18 10:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-02-20 16:12:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-19 13:41:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-21 15:32:19.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 10:38:54.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-02-19 13:41:50.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-02-20 16:12:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-20 22:57:42.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-20 23:17:53.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-02-21 01:09:52.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-02-21 01:24:00.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 05:50:04.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 09:30:11.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 10:16:55.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-02-21 11:03:13.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 11:55:36.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 12:15:42.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-21 12:24:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4856a4ae465606307308fcf4ac688e315840868","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-21 09:11:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6912526f7b199c92e9ab07046fbca608bce3a4f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-21 09:19:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/700265/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/1525f15 : FAILURE in 7m 26s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/700265/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/ddc1830 : FAILURE in 1m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"1e4d00f612fc502f102af6373726b6aa5d2ebcfd","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-21 10:09:37.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/700265/1/iscsi/ : SUCCESS in 57m 30s\n- os-brick-smb http://cloudbase-ci.com/os-brick/700265/1/smb/ : FAILURE in 54m 52s","accounts_in_message":[],"_revision_number":1},{"id":"238240628d1ffeb830f3e1c58f6befa3a9b7feaf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-21 10:11:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-ISER-LIO http://13.74.249.42/refs/changes/65/700265/1/Cinder-ISER-LIO : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"f372dfd9b2315030b9d6e735f3e0b85a3af249f7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-21 10:37:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/tempest-dsvm-full-xenial/28a0c96/ : FAILURE in 1h 24m 57s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/devstack-plugin-ceph-dsvm-xenial/7dd936e/ : FAILURE in 53m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34d5ade81157948045d52c0fa58afbb6e89671f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 11:00:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b5ac9e33f5f0422587a347cef25305c1 : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbfebde2120c46e7a7ec48f742497996 : SUCCESS in 1h 25m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6334e0f081ba44da97d50c510740d0b5 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34cee1446f0243ad88c7531287a472b4 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3f0e4362115441b81dfce48ebbda8d9 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a486216ee483422ca62c2fb0ba7cee9c : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ccd656839349b0b65cc8130ab951b3 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dcce9c0a83b439c8f69a9dc11d05101 : SUCCESS in 5m 53s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/2985e43c4a984278bbe0e8ae4bdf1344 : SUCCESS in 1h 38m 11s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/4f73602bbff944ed90c351c9dfaf9a41 : SUCCESS in 1h 34m 20s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/82ce486600e945a99f0dec9dc48ebacb : FAILURE in 1h 46m 51s","accounts_in_message":[],"_revision_number":1},{"id":"fe0b51105e3d4a58a9c8e9ef2334546e9d83fd43","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-21 11:10:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-ISER-LIO http://13.74.249.42/refs/changes/65/700265/1/Cinder-ISER-LIO : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4e3897a4d57a1e6be25dc704f4093cfc7cd9ac11","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-21 13:48:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e158a7794db8f6734f029e4faa7e48dd83c4eea8","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-21 13:57:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap http://openstack-logs.purestorage.com/65/700265/1/check/PureISCSIDriver-tempest-dsvm-xenial-aio-multipath-chap/f5398fe : FAILURE in 7m 59s\n- PureFCDriver-tempest-dsvm-xenial-aio-multipath http://openstack-logs.purestorage.com/65/700265/1/check/PureFCDriver-tempest-dsvm-xenial-aio-multipath/4d9faa0 : FAILURE in 8m 49s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"807aa1fe92dd96c3264c77b037d3d7de2357c7db","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2019-12-21 14:55:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/700265/1/iscsi/ : SUCCESS in 1h 05m 49s\n- os-brick-smb http://cloudbase-ci.com/os-brick/700265/1/smb/ : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2faaca17a6a2dc1eda8aa776e0f1e47282264a42","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-21 15:16:10.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/tempest-dsvm-full-xenial/7c44be1/ : FAILURE in 1h 26m 40s\n- devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/devstack-plugin-ceph-dsvm-xenial/99419ad/ : FAILURE in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"371f79fc8438c25d1261a958764cf5ba6930e4e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-21 15:24:25.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-ISER-LIO http://13.74.249.42/refs/changes/65/700265/1/Cinder-ISER-LIO : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"816f25f771eb1c65534f98c0d9994e6d9982ee28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 15:29:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4aebbd24112c414bb562850994a879f5 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34ec2d67242e41a199537444d35db4f2 : SUCCESS in 1h 26m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12b6f4df0a8d40aa81870cc5290aa260 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0884ab6d32154f5e925ad2f5393f94dd : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b03e30a626b24594b95ad057fb149853 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2196ab2da34f42d9b1a2512e9f050a73 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d453ba28e649b8ba3ea2cb7325a54e : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e12eeb2301a94b208247fd0248592e6a : SUCCESS in 4m 26s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/6f8003ee99ca48499c866e29b23f4129 : SUCCESS in 1h 13m 51s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/3aea18fbcef342f6ae3f28e81612af66 : SUCCESS in 1h 16m 23s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eefebd551f614e4a95d09b445145df90 : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8f6593badc369009d9289ae43186fabce01425de","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-12-21 15:32:19.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- 3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/65/700265/1/check/3par-iscsi-driver-master-client-pip-ssa02-src-os-brick-dsvm/7ae8911 : NOT_REGISTERED\n- 3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm http://54.201.44.218/65/700265/1/check/3par-fc-driver-master-client-pip-ssa02-src-os-brick-dsvm/93a6c55 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"b841bc03483e7d90cbc7f218dcb96a5f9a01e7d4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-12-21 15:43:36.000000000","message":"Patch Set 1:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-700265-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"b5d23dbbef642dc30ca50bfb34366c073ee48e9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-21 16:45:33.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-ISER-LIO http://13.74.249.42/refs/changes/65/700265/1/Cinder-ISER-LIO : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4b4905c12fd5399a798861bcedd005b1284947ad","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2019-12-21 17:41:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/65/700265/1/upstream-solidfire-check/cinder-solidfire-iSCSI/5c1d8ee/ : SUCCESS in 2h 11m 48s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"ffb5039f6f77f019b10342b3000abf9c5b7e0f8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-21 18:42:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-ISER-LIO http://13.74.249.42/refs/changes/65/700265/1/Cinder-ISER-LIO : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"745f38bc723ef1fd2891d9654544beb955556869","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-21 22:07:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-iSCSI/8a31802/ : SUCCESS in 52m 02s\n- cinder-cDOT-iSCSI-IPv6 http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-iSCSI-IPv6/2c98e78/ : FAILURE in 2h 35m 23s (non-voting)\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-NFS/34a6f1e/ : SUCCESS in 51m 08s\n- cinder-cDOT-NFS-IPv6 http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-NFS-IPv6/0eb2c0f/ : FAILURE in 3h 26m 33s (non-voting)\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-FCP/ce28561/ : FAILURE in 38m 41s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d3f1efe07515c10f2cc009e9b83a1799eada14f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2019-12-21 22:49:52.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_SC_FC-os-brick/0194810/DellEMC_SC_FC-os-brick/47 : SUCCESS in 1h 12m 11s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_SC_iSCSI-os-brick/f1c4af9/DellEMC_SC_iSCSI-os-brick/45 : SUCCESS in 1h 37m 13s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"f62023a9aa1e38e8cc32601fa013de46b21633a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-22 01:13:07.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-ISER-LIO http://13.74.249.42/refs/changes/65/700265/1/Cinder-ISER-LIO : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f76893b94e7e3a34a7e538399d55919eaa86de7b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-12-22 04:50:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_VNX_FC_os-brick/afc5547/DellEMC_VNX_FC_os-brick/43 : SUCCESS in 59m 55s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_VNX_iSCSI_os-brick/4a846d8/DellEMC_VNX_iSCSI_os-brick/42 : FAILURE in 2h 54m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"55446b7a76e1ed75799d7aa3526d8d22a72934d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-22 05:52:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-ISER-LIO http://13.74.249.42/refs/changes/65/700265/1/Cinder-ISER-LIO : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7b9d4f9bc9a3bfa970ee0b9060ff78dedd226efd","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-02-18 10:38:54.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for the cleaning this up, Sean","accounts_in_message":[],"_revision_number":1},{"id":"c17298663571f2a4f3ebc8f08fe19cae71542d16","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-02-19 13:41:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, Sean.  This will help keep us consistent across components.","accounts_in_message":[],"_revision_number":1},{"id":"5c4b7f1fb5a656b987259e745cf34a2d4b123d5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:42:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93893a5c24053443cc91544e130f6312ca3e2def","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 15:16:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cffe9e9baff046a6820b6cec07820e77 : SUCCESS in 4m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b8a6f1ad2fd473392166a10495feb9f : SUCCESS in 1h 20m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51a59ea6b61f4d6692326546785a3097 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73332f976c24122bf2ffd50ab962f96 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd8d2c14d824a229c935950024d404b : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bcc14b50d4042849c68cd9dbda37902 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a5a02e38e146fd8c2da43c4ac05b2f : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52738313aa6421f905fb5370b16749d : SUCCESS in 7m 23s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/be53c180c8c94462a05eae87a9aa127e : FAILURE in 1h 30m 02s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4cbfebb09954f5c807d7dd73f80bbdc : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"42112de1b2acf9f0fc6506005b5ebe0298b8d540","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-20 14:50:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cf2e90ab8b1bca6db9d7b7d06dc0d58da9393cc1","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-02-20 15:44:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/700265/1/iscsi/ : FAILURE in 53m 49s\n- os-brick-smb http://cloudbase-ci.com/os-brick/700265/1/smb/ : FAILURE in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4bf88d3d9b2270c532fd5f55b8ac128a9b6748ff","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-20 15:55:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/65/700265/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4bdf432 : SUCCESS in 49m 30s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/65/700265/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/57474c9 : SUCCESS in 1h 05m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"5c503a6b4a2648e3d576979222ffa58bb50bf1d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-20 16:07:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9ab6bf4a6d6560da7fbc69dd0b4694f7383e407b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 16:12:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d595ab1a2c4f21fb8e67d178efb66c74d363593e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 16:33:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/16f75fca110347bc8f556732400f1d2f : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84591296478049789981bd2d1de43a63 : SUCCESS in 1h 32m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5576377097ce46c780bec2629b72bde1 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a93f99609264096931383c8b5460bf6 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5df295d6a0b143259dfe69e630f8012c : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3242709725da44eaba7d11a1fa8a8891 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e96ebd867248f099163e54f7b8bc1e : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55fa1c0b713c4b02a081f1fd9d6329b3 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bfe8fb754e7463d855b2366136c9518 : SUCCESS in 11m 01s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/ddc6f4423d8543819ca656cecbe9b63a : SUCCESS in 1h 25m 40s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/f8a7bddeb9e74ffeb5a9f9757078384d : SUCCESS in 1h 14m 56s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/771929ae066c4116b60244a29406817b : FAILURE in 1h 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"00fefc56f36af66e4853fb1cea67f8bfa72e6a99","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-20 17:51:29.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/tempest-dsvm-full-bionic/1e70eda/ : SUCCESS in 1h 38m 51s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/devstack-plugin-ceph-dsvm-bionic/eae4704/ : FAILURE in 2h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"908d672d5f153a40f6e53bb40a486bdb37b07155","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-02-20 19:14:46.000000000","message":"Patch Set 1:\n\ntest failures are timeouts not related to this patch","accounts_in_message":[],"_revision_number":1},{"id":"8c06d8200c08f4024f658144707f8f174f6fde7a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-02-20 19:14:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d29d862e194169de95ad304a6f0e309620b2633","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-20 19:29:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"80b0025c334b5d4a9e959d4528743d319aef0181","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-02-20 20:09:18.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/700265/1/iscsi/ : FAILURE in 53m 35s\n- os-brick-smb http://cloudbase-ci.com/os-brick/700265/1/smb/ : FAILURE in 54m 02s","accounts_in_message":[],"_revision_number":1},{"id":"061237fcd04470123f2803bf3f7850638326940f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-20 20:21:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/65/700265/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/2aa36b5 : SUCCESS in 52m 38s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/65/700265/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ea89334 : SUCCESS in 1h 05m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"4421bacdbe6c24a8c6b0e480086cb3721258d75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 20:44:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e32112a5a88408885a165d8f856acf3 : SUCCESS in 3m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a07596eb718c431baa5ec4c342369c06 : SUCCESS in 1h 15m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00b3930af3214770b4d97f0fbdd0ff18 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f92e57503849d99fa601b67e0b81db : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a0b80722db40ac955a28ff229a98a1 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67dbf71b5d474b7b8e2fe3919c52c7cd : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cecda22b0064ba983bb1868484e2a95 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55de149a32c34ac9b665a131ae4ce6a9 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/947aeb6681154c61b763d48cb2a66e91 : SUCCESS in 6m 36s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/279d4a6c16d8422c8752fd58087c75d2 : SUCCESS in 1h 14m 14s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/b3fa5b68446e44289de464155e8b7b80 : SUCCESS in 1h 21m 35s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/34455631bf674ee691747482ae9c5cbd : SUCCESS in 1h 28m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ec818af1f0828b8597431b958494cd0a74f86d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 20:44:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7180f13c9b418eda1af9ac9b6f980e8b059dc619","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-20 21:27:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/tempest-dsvm-full-bionic/e4ee489/ : SUCCESS in 2h 08m 25s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/devstack-plugin-ceph-dsvm-bionic/5e177e6/ : SUCCESS in 1h 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ec4ec4717e7b763e67216e6c4adff01300406ddb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-20 21:44:05.000000000","message":"Patch Set 1:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-700265-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"d0f54d359e81ba2ddedac42a5c746e37fd29013e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-20 22:04:06.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"506813d190cf6855516070f6f4ac60581ad4e0ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 22:27:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/66e4b383635140b6a321159ff92be366 : SUCCESS in 5m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/483c82deaa944ed5acc64fcc7ca67ca0 : SUCCESS in 1h 28m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8763413c86d7498cb18c1af5514aff9d : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa368c70b71f4e90a9d1c0d9e480bd7b : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fc05cea43b0466aa4a12af542f20c17 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa3a95185ad64fcab3b29595993afcbd : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc57a06f01e4544a4888c79f94f5e66 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c02c43a14de45f2bedb20736c5429cc : SUCCESS in 7m 31s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/69cb166d7b5b467f816ab2f2b2bcb1d2 : SUCCESS in 1h 40m 21s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f82e375a428490dad6a65695906e012 : FAILURE in 1h 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"365909599284fe867f5446795926606e3a7da39b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-20 22:57:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-iSCSI/3b362d3/ : FAILURE in 1h 32m 51s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-NFS/9297488/ : FAILURE in 1h 30m 48s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/65/700265/1/upstream-check/cinder-cDOT-FCP/d17b883/ : FAILURE in 2h 12m 42s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"568109f4686930459d34027092ccf070be28ac44","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-20 23:17:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/65/700265/1/upstream-solidfire-check/cinder-solidfire-iSCSI/3f9fba3/ : FAILURE in 2h 32m 45s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"adfee90f58334db515cf9c77ce2269bf313d9312","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-21 00:11:54.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b97b8f371d9749572ee68b1be708b4bc78f26828","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-02-21 01:09:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_SC_FC-os-brick/905e4be/DellEMC_SC_FC-os-brick/61 : FAILURE in 1h 12m 34s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_SC_iSCSI-os-brick/eee44c2/DellEMC_SC_iSCSI-os-brick/59 : FAILURE in 2h 05m 28s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"a63e4585af91db525e0c0c29c07e691b2d1c6495","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-02-21 01:24:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_VNX_FC_os-brick/c384af0/DellEMC_VNX_FC_os-brick/55 : FAILURE in 1h 07m 30s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_VNX_iSCSI_os-brick/43cb558/DellEMC_VNX_iSCSI_os-brick/54 : SUCCESS in 2h 02m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e29913dfa0336bb4d059a7300080a2f07f018535","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-21 02:23:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"58fc4959f7a5394836577f1e067f6cba7b5e72eb","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-02-21 05:50:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_XtremIO_FC_os-brick/6836bec/DellEMC_XtremIO_FC_os-brick/78 : FAILURE in 10m 26s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/65/700265/1/check/DellEMC_XtremIO_iSCSI_os-brick/4856559/DellEMC_XtremIO_iSCSI_os-brick/77 : FAILURE in 11m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"2f5658d3dc8132fcd6358996abc7bd124b92bbf4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-21 07:02:56.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"81278c8e18a4c4e694d87f4fb9b412633bbc6cc3","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-02-21 09:21:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5e69649a9022bcd8295bae95f8914e98ccc3bef","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-21 09:30:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/65/700265/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/196fddd : FAILURE in 51s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/65/700265/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/05719ec : FAILURE in 8m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"25e6c83359a3d09d013faddb40d70dd9db6f5105","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-02-21 10:16:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/700265/1/iscsi/ : FAILURE in 55m 01s\n- os-brick-smb http://cloudbase-ci.com/os-brick/700265/1/smb/ : FAILURE in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"90e60b6e0838ff318b69d463ec7d2485366a8425","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-21 10:35:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a7aff880fe1760325cf758cda520e8e07ee78d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 10:59:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dfff9653ce834938be1c04de601552a4 : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a75721836e245efa20277e98824bb8c : SUCCESS in 1h 33m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a77cee2b09b458f8b9dbb77e0dfb197 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaeca14f4da748caa5244506cd78b40b : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e4d7826794a4650bf962c3eb9bc4a9c : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64cea0c301f6459c910725f845a80469 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f18ab769edd34609a797bbf58fea9db6 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f300174a064e3682af84e34ba9ef37 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/725c13337a9b463094aa96243951e79b : SUCCESS in 6m 24s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/e7c062599b3b4ac7b7ba3940ad61c543 : SUCCESS in 1h 10m 55s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/651d0b1537be4052a3915274f93d5b89 : SUCCESS in 1h 24m 06s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b2ee45f4e1bb4bcf879b589c6d84b77f : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"13143c11233d1731d8000f5449936bda0c606275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 10:59:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8075a89a8a6b29fa98f821c713c4bece8e04aeb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-21 11:03:13.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/tempest-dsvm-full-bionic/9c869fa/ : SUCCESS in 1h 37m 36s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/65/700265/1/check/devstack-plugin-ceph-dsvm-bionic/aba443c/ : SUCCESS in 1h 38m 31s","accounts_in_message":[],"_revision_number":1},{"id":"57d094b9caad54fef6ebdb3055fbc4e4a308a5bc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-21 11:55:36.000000000","message":"Patch Set 1:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-700265-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"ea70b69078eab17ed3af629703602f3576be306e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-21 12:15:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/65/700265/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"1669cc7314f29f11b9518aa0670b5572a55c8fec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 12:24:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f8d3394c154d23633ca02093887c6d495f1de653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 12:24:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/575e07146c9140ef8a3c7c7420ca37de : SUCCESS in 4m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7606c8943dda4d5f9d31f58ee9cc9027 : SUCCESS in 1h 21m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86e6d5e7dc6644af8e3d20573771830f : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd46145380645c5b92ec1afd84d27d5 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3653df586c4041f6ba53e89b348fa8d4 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3d05fe34f5d426ebfb2359664fa24e4 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91f36f2c2f14adf8582f209979bca58 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ce0b82eca2c4fab8ce742fe126bd499 : SUCCESS in 7m 59s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/104ce594fa604294ab75444d12f1eb7f : SUCCESS in 1h 21m 35s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8b124097600040a799d7ae5b3e95dc29 : SUCCESS in 1h 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e6df72e4e7e438e5906bac9e4be5c791b9b21b28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 12:27:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d7a3dc5de94e52ac86dad39459b319 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edb36f01854e432c9a6fdff8415465c3 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1448c7ff622b97f821a00d0ce2c9c667136f6ccd","revisions":{"1448c7ff622b97f821a00d0ce2c9c667136f6ccd":{"kind":"REWORK","_number":1,"created":"2019-12-21 09:11:39.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/65/700265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/65/700265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/65/700265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/65/700265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/65/700265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/65/700265/1"}}},"commit":{"parents":[{"commit":"df2a3969b7e191d40430e6c37f28dcf31d4b1da1","subject":"Merge \"Remove Python 2.7 support from testing and gates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/df2a3969b7e191d40430e6c37f28dcf31d4b1da1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-21 09:11:27.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-21 09:11:27.000000000","tz":-360},"subject":"Raise hacking version to 2.0.0","message":"Raise hacking version to 2.0.0\n\nWe\u0027ve kept hacking capped for a long time now. This raises the hacking\npackage version to the latest release and fixes the issues that it\nfound.\n\nChange-Id: Ib13a3fe3f1a103b5e24a7c5bb7838ffdae2bf198\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1448c7ff622b97f821a00d0ce2c9c667136f6ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1448c7ff622b97f821a00d0ce2c9c667136f6ccd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
