)]}'
{"id":"openstack%2Fos-brick~707644","triplet_id":"openstack%2Fos-brick~stable%2Ftrain~Idc281be7cf88eeeeefe260877a1fc275d94f2bed","project":"openstack/os-brick","branch":"stable/train","hashtags":[],"change_id":"Idc281be7cf88eeeeefe260877a1fc275d94f2bed","subject":"Skip cryptsetup password quality checking","status":"MERGED","created":"2020-02-13 14:12:39.000000000","updated":"2020-02-19 17:11:08.000000000","submitted":"2020-02-18 17:43:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707644-1582047781352-798fc2db","meta_rev_id":"ea890b7e8c4f8a5f8af569377c044b50534ab33e","_number":707644,"virtual_id_number":707644,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-13 14:14:35.000000000","_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-18 16:35:33.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-02-13 20:43:11.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-18 17:38:17.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-02-18 13:59:55.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"date":"2020-02-18 17:01:11.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 17:43:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"date":"2020-02-18 16:50:06.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-02-18 14:59:02.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":2,"date":"2020-02-13 14:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-18 10:32:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-13 18:44:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-18 10:32:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 14:14:35.000000000","updated_by":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"reviewer":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"state":"REVIEWER"},{"updated":"2020-02-13 14:32:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-13 18:44:52.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2020-02-13 20:43:11.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 10:32:21.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-02-18 13:59:55.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-02-18 14:59:02.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-02-18 16:35:33.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 16:50:06.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-02-18 17:01:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 17:38:17.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 17:43:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7786ff020bc0f811925d9c957bed3ff2be62b8b2","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-02-13 14:12:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b0407ec5a3d53dbe0cb757f0b317ef7213f9bcb","author":{"_account_id":30112,"name":"Aditi Pai Dukle","email":"Aditi.Dukle1@ibm.com","username":"aditidukle","inactive":true},"date":"2020-02-13 14:14:35.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":1},{"id":"a37b38b0c9c7e8ff81d57d4ef4f2152b34a304ce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-13 14:32:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"563918c00bc3df35d55cd727a01ee1ac9e8c8f2c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-13 14:47:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"19674be9dc82813372bab498b59279232a738b71","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-02-13 15:06:45.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/707644/1/iscsi/ : FAILURE in 53m 37s\n- os-brick-smb http://cloudbase-ci.com/os-brick/707644/1/smb/ : FAILURE in 52m 34s","accounts_in_message":[],"_revision_number":1},{"id":"56f255ba205db10a83dfefb539e30f28c7e4b14d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-13 15:47:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/44/707644/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c0ea66e : FAILURE in 1h 14m 11s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/44/707644/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/523a9dc : FAILURE in 1m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"43acd0f3c1b5ec81be278ed1bb684ce48d268736","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-13 16:02:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9d3496d8e23208cee59b93077c8a4b77e97108ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:05:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/271ab4bf101e421ea0da8912edc137f1 : SUCCESS in 1h 38m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ac7691e1d504d499dc993afa6e5d870 : SUCCESS in 1h 20m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50881b8e6f7548c1a9c6e2f0328611c1 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f62e29294ef49bc9cb7315181259313 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5673cb95cc134ba982a757cfaa4eea2f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c10fda67b1804a4c9adbb13a29b88920 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8070f48e1bde493393ac7d5c647f2c16 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3d105d07d14151941e509292d98156 : SUCCESS in 6m 17s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/07c3e2bdc4d9497db458610887ffcc26 : SUCCESS in 1h 45m 53s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/b7336249c43f40d4906866e80bfc9bd3 : SUCCESS in 1h 31m 10s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/476a86b1785049ff9e42cf427e623623 : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d0f1dc1a62429451ed95646db654ad7c15c2c53f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-13 16:59:47.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/82452f118cde42e1b2def36c5bec22e7 : TIMED_OUT in 2h 37m 17s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/44/707644/1/check/devstack-plugin-ceph-dsvm-bionic/afbb2bb/ : SUCCESS in 2h 06m 47s","accounts_in_message":[],"_revision_number":1},{"id":"baab4dcd87f5860e046b034c3738e17bfc32dc60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-13 17:51:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c9c8790e53fd1addc1a06c37a93f9cf7b3516663","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-02-13 18:44:43.000000000","message":"Patch Set 1:\n\nClean backport, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"326bacd5c8cbe7af3ba462dd7e95e0211c752870","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-02-13 18:44:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6658a1cdf04196adb9deb3249817589812346413","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-13 20:03:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/44/707644/1/upstream-check/cinder-cDOT-iSCSI/ac7bdcf/ : SUCCESS in 52m 51s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/44/707644/1/upstream-check/cinder-cDOT-NFS/c35e91f/ : FAILURE in 3h 57m 41s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/44/707644/1/upstream-check/cinder-cDOT-FCP/78d9601/ : SUCCESS in 1h 23m 35s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4091f81ecd7c11db7f1191507aa943dc52ce77d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-13 20:03:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/44/707644/1/upstream-solidfire-check/cinder-solidfire-iSCSI/9116e5e/ : FAILURE in 3h 57m 52s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"48639555de9a18ccb53de4303bf25685fa9e4fdb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-13 20:43:11.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"38a87ee6fe74a4b930d6da72d5abf6bcbc528bef","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-02-18 10:32:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad4e75d0600a247f853171bbc8ec18c312ce3257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 10:32:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8643b796d6cf791cbb7ab8319014fe7db0d334cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 12:19:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/55057734fd1c48a99d415575cae8a7d4 : SUCCESS in 1h 32m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99027018aab84273ab47d32c29710a12 : FAILURE in 1h 20m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b45e718611fa460783c0aa5f12a6894a : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0406985194b342579bd3e5619f51b744 : SUCCESS in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c04372af88e41b3b522c657058e6872 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e7623597b514e29b7933309459100f5 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29a3247c953946f4bb46b8f40000cbcd : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a81e6235bf4bcf93bea80add2e682c : SUCCESS in 9m 04s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/daae024413ea486695f558e50c10cb97 : SUCCESS in 1h 32m 58s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/21718a5c50574a97af14c9cf06b72fb0 : SUCCESS in 1h 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3b688876d992ae81f626dffdb6ed2d34735847a6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-18 13:59:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ebae870d010da21a68e3bb376b7c5a00a9a18788","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-02-18 14:59:02.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/707644/1/iscsi/ : FAILURE in 58m 46s\n- os-brick-smb http://cloudbase-ci.com/os-brick/707644/1/smb/ : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"871005eb55ba8e784536499b2edcda28fa3c0ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 15:44:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/45b0af3cae6c4d43831a840ee1b6b560 : SUCCESS in 1h 38m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ca6b1f183c143f98246c3cfa8ae8afc : SUCCESS in 1h 24m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f3141ebe02a4139a2e139d0a4d60574 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ada0562c084a1db7a94d07464894b3 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2de28b78ad494ba49711049a668d8d1b : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22812116a65f404d83b96295d530374e : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d1ea6eeaa75421aa5e1ed09baacafec : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8da8b07563c44a095d3533ba6da08cc : SUCCESS in 8m 15s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/e0cbfa7fd6be4842934abb0bc23f9d2f : SUCCESS in 1h 27m 48s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/960b213a5e1544ec92868104a66ec2eb : SUCCESS in 1h 35m 25s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/75144cc5e8dc4a0781421fcbb6cd9fcf : SUCCESS in 1h 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fff3a76d8746e7f556f1c03bcba53bd9ec6dbfff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 15:44:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ebfe1acb6c47a8e769e49429c2b904e4f8ad94","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-18 16:35:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/44/707644/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/de895a6 : SUCCESS in 1h 00m 10s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/44/707644/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/892d038 : SUCCESS in 1h 20m 56s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ba414bd6a0e8230b613187b4d76c676592dfba1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-02-18 16:50:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/44/707644/1/upstream-solidfire-check/cinder-solidfire-iSCSI/428369d/ : SUCCESS in 1h 05m 08s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"9a84c2189c421ef7a24595cc379b394c72185a98","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-18 17:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/44/707644/1/upstream-check/cinder-cDOT-iSCSI/6fc6225/ : SUCCESS in 59m 05s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/44/707644/1/upstream-check/cinder-cDOT-NFS/a9b9e54/ : SUCCESS in 58m 24s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/44/707644/1/upstream-check/cinder-cDOT-FCP/c90faa2/ : SUCCESS in 1h 16m 21s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ffa073135ada0f42cbdd3d419f3213b2a808e405","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-18 17:38:17.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/ac82622f8f684efea23b3aecf0a145a4 : TIMED_OUT in 2h 41m 46s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/44/707644/1/check/devstack-plugin-ceph-dsvm-bionic/ad4073a/ : SUCCESS in 2h 24m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3132a6c2beb63c2d333250b2a32d0f6d2985150d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:43:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6f0057ad93c1ac9dfd20de6323c31caad73e6c35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:43:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3779d0a2e8f8483ba52b8f35a0ebd460 : SUCCESS in 1h 38m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbe7db10ee854ee9807742552faf1505 : SUCCESS in 1h 32m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81af666c3a534210b023dbb8d6be3abf : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b15625d2cb4b67ab59beb10b673709 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/183ab96a429c48cba82c7ce6f38c611e : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a352d5b9e3642c288e20d99cb688c3f : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63314c45550448dabae36877af678202 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913e5d2efb844b6e8dcc935980186671 : SUCCESS in 7m 18s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/c171776adcc24474959f58c994152e61 : SUCCESS in 1h 39m 22s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf0bd3479cb243aa9b475bf617e30b9a : SUCCESS in 1h 46m 38s","accounts_in_message":[],"_revision_number":1},{"id":"90740c5e319148daec84022a7270dabb34d19332","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:45:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d623036c6294d82b675ef1f02447e8e : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"04f8b51df618c18a1f5fb87ecbab256dceab3ffc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-18 18:15:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/44/707644/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ea890b7e8c4f8a5f8af569377c044b50534ab33e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-02-19 17:11:08.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 0db2285387e32230471c81eef057c74436a9c82f","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6461c993bc24a07af0b4240aaccab50eaa018707","revisions":{"6461c993bc24a07af0b4240aaccab50eaa018707":{"kind":"REWORK","_number":1,"created":"2020-02-13 14:12:39.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/44/707644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/44/707644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/707644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/707644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/707644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/44/707644/1"}}},"commit":{"parents":[{"commit":"d8ed21e7d25f67c6d2bfe71515723f68ce15b887","subject":"iscsi: Add _get_device_link retry when waiting for /dev/disk/by-id/ to populate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d8ed21e7d25f67c6d2bfe71515723f68ce15b887"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-01-28 14:12:55.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-02-13 14:12:39.000000000","tz":0},"subject":"Skip cryptsetup password quality checking","message":"Skip cryptsetup password quality checking\n\nLUKS password quality checking is not useful\nsince we only use long hex strings for passwords.\n\nNot skipping this means that we have to install\ncracklib-dicts for cryptsetup to work, which is\nunnecessary weight.\n\nCloses-Bug: #1861120\n\nChange-Id: Idc281be7cf88eeeeefe260877a1fc275d94f2bed\n(cherry picked from commit afb7beb7ce98824d5fb789ad8e7577cebae4e41c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6461c993bc24a07af0b4240aaccab50eaa018707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6461c993bc24a07af0b4240aaccab50eaa018707"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
