)]}'
{"id":"openstack%2Fos-brick~713082","triplet_id":"openstack%2Fos-brick~stable%2Fstein~If31d120955eb216823a55005fdd3d24870aa6b9a","project":"openstack/os-brick","branch":"stable/stein","hashtags":[],"change_id":"If31d120955eb216823a55005fdd3d24870aa6b9a","subject":"Read mounts from /proc/mounts instead of running mount","status":"MERGED","created":"2020-03-13 23:57:35.000000000","updated":"2020-07-29 18:14:26.000000000","submitted":"2020-05-20 02:08:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713082-1589940490109-278ba6fa","meta_rev_id":"ea5f4142be768a9431d7f3c5d3b1d394e8cd3e68","_number":713082,"virtual_id_number":713082,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-03-14 04:07:26.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-14 02:37:52.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 02:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"date":"2020-03-14 01:54:50.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-16 09:29:48.000000000","_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"date":"2020-03-14 03:31:10.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-14 02:42:33.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-03-14 00:05:34.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-20 00:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-23 08:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-03-26 12:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-25 16:27:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-20 00:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"value":0,"date":"2020-03-13 23:58:12.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 23:58:12.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-03-14 00:05:34.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-03-14 01:54:50.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-14 02:37:52.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-14 02:42:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-03-14 03:31:10.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-14 04:07:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-16 09:29:48.000000000","updated_by":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"reviewer":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"state":"REVIEWER"},{"updated":"2020-03-23 08:43:40.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-03-25 16:27:06.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2020-03-26 12:54:38.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2020-05-20 00:31:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-20 02:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2ac3107d71282a78f6967843eb2c92ad7646077","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2020-03-13 23:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5277e388391525712e746007cf055632ac46ae1b","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-03-14 00:05:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/713082/1/iscsi/ : FAILURE in 7m 17s\n- os-brick-smb http://cloudbase-ci.com/os-brick/713082/1/smb/ : FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d31844c9860f6f8ae768e473f8eebad27a0964b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-14 00:05:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-tgtadm : ABORTED\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-lioadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"24d2938039e43539e7f146fb7a687029ac14304f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-14 00:36:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-tgtadm : FAILURE\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-lioadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c2fab8b70c128da5281229da9f7cc8814b196bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 01:42:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2429395960cd40edb98f9d9fa415e413 : SUCCESS in 1h 42m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2957a125399c4367a15e2994a4ecc593 : SUCCESS in 1h 23m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fdfbb3b4dbd4f4aa4540913ae3bac3f : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ad6bdbd61c4075bb6f931df43868b3 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/774d819b4cd74feba1d17aa59ad1a9ee : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd7e8b838424e6bac8a654cbcfc1db2 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aeb931377af44f0cbf57f18d537eebb4 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d93e365fd744ca0ac72dafcd8c63e2c : SUCCESS in 4m 28s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/e127672371234ab099ea2a111832285f : SUCCESS in 1h 12m 19s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/5768302148fc4cad8946f6756cd146cd : SUCCESS in 1h 14m 42s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/acd3b7dd84bc4e78be2e6276a29c834f : SUCCESS in 1h 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e28f001065815c5ac16fac390747f1db2973999e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-14 01:54:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/82/713082/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/20d2042 : FAILURE in 1h 57m 00s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/82/713082/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ba77547 : UNSTABLE in 1h 12m 35s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"81786260a47a8407ef88764212ad8ce544732ead","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-14 02:37:52.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/82/713082/1/check/tempest-dsvm-full-bionic/7872443/ : TIMED_OUT in 2h 36m 26s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/82/713082/1/check/devstack-plugin-ceph-dsvm-bionic/9253165/ : FAILURE in 1h 42m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3e99253369c9fd73a099545c7fadc487cc04d423","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-03-14 02:42:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI http://logs.openstack.netapp.com/logs/82/713082/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f4a4ad9/ : SUCCESS in 1h 00m 00s\n\nComment \u0027run-netapp-solidfire-ci\u0027 or \u0027netapp-solidfire-ci-recheck\u0027 or \u0027run solidfire\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"34b067f0d9a589e83f1f501a7ceb5c1c9d9c717b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-14 03:13:59.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-tgtadm : FAILURE\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-lioadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5b93a02603d80b7e7db0339cac1003563d250f4a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-14 03:31:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI http://logs.openstack.netapp.com/logs/82/713082/1/upstream-check/cinder-cDOT-iSCSI/2404d00/ : SUCCESS in 1h 48m 46s\n- cinder-cDOT-NFS http://logs.openstack.netapp.com/logs/82/713082/1/upstream-check/cinder-cDOT-NFS/b3d5c41/ : SUCCESS in 46m 28s\n- cinder-cDOT-FCP http://logs.openstack.netapp.com/logs/82/713082/1/upstream-check/cinder-cDOT-FCP/59a3926/ : FAILURE in 1h 24m 38s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"489e6793c22d2a12833249d67a09fff67ae155e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-14 04:07:26.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-tgtadm : FAILURE\n\n* Cinder-lioadm http://13.74.249.42/refs/changes/82/713082/1/Cinder-lioadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fb87d8a99718b15719da171762176a96b15d387b","author":{"_account_id":21976,"name":"Kaminario K2 CI","email":"cinder.devops.CTR@kaminario.com","username":"kaminario"},"date":"2020-03-16 09:29:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- kaminario-dsvm-tempest-full-iscsi http://52.3.254.104/82/713082/1/check/kaminario-dsvm-tempest-full-iscsi/3e4b088 : SUCCESS in 2h 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cb4d92dc445613d285ef7fb8c330a08f47c1f43f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-23 08:43:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e053ea963f980a20238d1896db390258e0b04b2c","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2020-03-25 16:27:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"295b8b01d3b004796dd361b29d408319bdf39605","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-03-26 12:54:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb51d6d2e15e03a9b4118f7e49997bf939dddc9a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-20 00:31:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"812cc19b10645139a17a2cdc260cfbd2cbf7765b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 00:31:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95421d6899552aa64457d2e2e6aff04249b649f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 02:08:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/09a5c831a0f94811b4c728b5a4326551 : SUCCESS in 1h 28m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db92dbb048c44691bab912e744d24d52 : SUCCESS in 1h 23m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7770625b586443aa1ccd165686ddeb4 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bead5ffdfcc54e08ad20a0287479376b : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1997fe96a9a47e2a991d267dd1720d8 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfbc0642200c41eeafd4640c096d33c0 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9a8686b8f6a42409568546b330486c4 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2e1c75f0154b0ea0b85c5a70508501 : SUCCESS in 5m 33s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/760e7b57f2d64295a31ec233e020ce31 : SUCCESS in 1h 20m 16s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e6fae1ce91954629903269ed9297ab89 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7f3583fe41944dbab731cfef9c59a3e138b3049e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 02:08:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6f6d136a19ac6e6dc45f818d1c2b062c62a17fdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 02:09:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c23751b558543a7a641ddbd6398c823 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1},{"id":"ea5f4142be768a9431d7f3c5d3b1d394e8cd3e68","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-29 18:14:26.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 4e2e1bedb657cdb1ab01e77bc14b976b00df284a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7168bdce5bcdf3f1a27c6fc9f998c1f3f5364f3d","revisions":{"7168bdce5bcdf3f1a27c6fc9f998c1f3f5364f3d":{"kind":"REWORK","_number":1,"created":"2020-03-13 23:57:35.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/82/713082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/713082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/713082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/713082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/713082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/713082/1"}}},"commit":{"parents":[{"commit":"fd6425b2ee725f19aaf9bb4543cceed18ec95cc2","subject":"Merge \"Skip cryptsetup password quality checking\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/fd6425b2ee725f19aaf9bb4543cceed18ec95cc2"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2019-12-18 19:11:33.000000000","tz":-300},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2020-03-13 23:57:29.000000000","tz":-240},"subject":"Read mounts from /proc/mounts instead of running mount","message":"Read mounts from /proc/mounts instead of running mount\n\nWhen creating multiple concurrent volumes, there\u0027s a lock\nthat prevents mount from returning quickly. Reading /proc/mounts\nis way faster.\n\nChange-Id: If31d120955eb216823a55005fdd3d24870aa6b9a\nCloses-Bug: #1856889\n(cherry picked from commit f2409608473c06892b087e44e0041c9ade3f9924)\n(cherry picked from commit a8190cd60b59dedb11a8ed27d236d81adee3255d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7168bdce5bcdf3f1a27c6fc9f998c1f3f5364f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7168bdce5bcdf3f1a27c6fc9f998c1f3f5364f3d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
