)]}'
{"id":"openstack%2Fos-brick~720180","triplet_id":"openstack%2Fos-brick~stable%2Frocky~Id507109df80391699074773f4787f74507c4b882","project":"openstack/os-brick","branch":"stable/rocky","hashtags":[],"change_id":"Id507109df80391699074773f4787f74507c4b882","subject":"rbd: Use showmapped to find the root RBD device during disconnect_volume","status":"MERGED","created":"2020-04-15 10:38:21.000000000","updated":"2020-08-06 09:53:42.000000000","submitted":"2020-08-06 09:53:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720180-1596707622215-afcfe1c4","meta_rev_id":"02b1f55ac6b24ab5a5bae53d99216db95a9ce25f","_number":720180,"virtual_id_number":720180,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-15 12:28:39.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-04-15 14:35:15.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 09:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-15 10:54:59.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-29 18:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-06 08:22:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-06 08:22:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 10:54:59.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-04-15 12:28:39.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-15 14:35:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 18:00:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-06 08:22:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-08-06 09:53:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12d195a0a91219ddac416f01de4b2146362034a1","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-04-15 10:38:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5674f30d70165d18bb68c221ded8f29bf2b83185","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-04-15 10:54:59.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/720180/1/iscsi/ : FAILURE in 7m 34s\n- os-brick-smb http://cloudbase-ci.com/os-brick/720180/1/smb/ : FAILURE in 7m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c9c45d773a9e29f3d2c8d4d49e7f6cb012572ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 12:11:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1f7380e63b1541239d416200c4d6c872 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94d999f8ce94c6b874a7e908bcf3895 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc9b2a2db4c243279825013d2f3260ad : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/94468465985a4b22bef3a61c8125568a : SUCCESS in 6m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/02e9246551c3446cae108e1ec2452ee5 : SUCCESS in 1h 23m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0ad82b3925347a791b70fe06ecee1f9 : SUCCESS in 8m 49s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/a8b2c218b0c04b20836caa7f09774f78 : SUCCESS in 1h 13m 19s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/4817193967624fd88aebcea101cb9c22 : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9c10b8c126a6598d6850c46ff1c4dcf39835d279","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-15 12:28:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/80/720180/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/78a5c4c : FAILURE in 27m 49s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/80/720180/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/152e3b0 : FAILURE in 25m 23s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"42004a4cdd10fd4236af59accdb6747525cb6da0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-15 13:53:44.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds to merge to later branches before being backported.","accounts_in_message":[],"_revision_number":1},{"id":"0eec8a7620530133cdf7235884877424a97a326c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-15 14:35:15.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/80/720180/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/80/720180/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e0f87bd3f3254f94cf52216bf9f4a932e6e934cb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 17:57:18.000000000","message":"Removed Code-Review-1 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"cb40ed933e9709accc6a24ed09eed92b24ce5e9d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 18:00:48.000000000","message":"Patch Set 1: Code-Review+2\n\nFollow on patch cherry-picked to follow this.","accounts_in_message":[],"_revision_number":1},{"id":"35859483c9c0e7fd84bbf3ec7ccf41f9eae397d9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-06 08:22:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nClean cherry pick from Stein, where the patch has merged. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"a170c18256bee4f5296bf7966e7c6f8d76abff5b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-06 08:22:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65f4ad9b523e4a51c2a9cba9b99e8ff86e80e980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 08:26:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"beded55b555469ba6edae55336e17660941b7a62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:53:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f8d38385e8134f03a6f46a3285b5d872 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3f2e24ba854b22b8caa743576cb500 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1937572af62a4ab7a507ef7c207f9031 : SUCCESS in 7m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57f6f0c57a5c4f3187ce2dd0490481e3 : SUCCESS in 6m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7387c6b7f9a948ca8c7eb6c77f845667 : SUCCESS in 1h 24m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca9844a1a36b4802a8b5e073951710d7 : SUCCESS in 8m 05s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/5bd96308cec74e0eb5bacb4cc4bc5796 : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"02b1f55ac6b24ab5a5bae53d99216db95a9ce25f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:53:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcff1e0cb292a461b933bdc70dd21b4057711bca","revisions":{"fcff1e0cb292a461b933bdc70dd21b4057711bca":{"kind":"REWORK","_number":1,"created":"2020-04-15 10:38:21.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/80/720180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/80/720180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/80/720180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/80/720180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/80/720180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/80/720180/1"}}},"commit":{"parents":[{"commit":"d640c73d9b68bdf200c3e71ddec65b166a571f4b","subject":"Skip cryptsetup password quality checking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d640c73d9b68bdf200c3e71ddec65b166a571f4b"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-02-15 17:45:50.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-04-15 10:38:16.000000000","tz":60},"subject":"rbd: Use showmapped to find the root RBD device during disconnect_volume","message":"rbd: Use showmapped to find the root RBD device during disconnect_volume\n\nThe original method of unmapping the higher level RBD device found under\n/dev/rbd/{pool}/{volume} fails when an encryptor has been attached to\nthe volume locally. This is due to a symlink being created that points\nto the the decrypted dm-crypt device, breaking any attempt to unmap the\nRBD volume.\n\nTo avoid this we can simply find and unmap the lower level RBD device of\n/dev/rbd*.\n\nChange-Id: Id507109df80391699074773f4787f74507c4b882\n(cherry picked from commit 9415b3b41f55b66bee589bc8ff5f1c789d8a5d0e)\n(cherry picked from commit 36b207239b2dd4b3cad2b79d7ae32f4084c36403)\n(cherry picked from commit 3ce686a9fa41210fb4c08017220ce063cc869c75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/fcff1e0cb292a461b933bdc70dd21b4057711bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/fcff1e0cb292a461b933bdc70dd21b4057711bca"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
