)]}'
{"id":"openstack%2Fos-brick~721782","triplet_id":"openstack%2Fos-brick~master~I6f1dad8112b816fd23f17c1fb900d982ed0f1a4f","project":"openstack/os-brick","branch":"master","topic":"local/master","hashtags":[],"change_id":"I6f1dad8112b816fd23f17c1fb900d982ed0f1a4f","subject":"Function _connect_single_volume connect volume to virtual machine code logic error","status":"ABANDONED","created":"2020-04-22 07:15:56.000000000","updated":"2020-04-22 12:56:55.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2d8fb646310c3156c8fac13ac78263abc255d039","_number":721782,"virtual_id_number":721782,"owner":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-22 10:28:42.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-04-22 09:01:07.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2020-04-22 10:25:24.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"date":"2020-04-22 10:30:56.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2020-04-22 10:53:47.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-22 09:59:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-22 08:44:10.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 07:16:07.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-04-22 08:44:10.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-04-22 09:01:07.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 09:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 10:25:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 10:28:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-22 10:30:56.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 10:53:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"}],"messages":[{"id":"b86bb1552ad316f67268c7bb25d1c6206ae2a234","author":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"date":"2020-04-22 07:15:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e1430e85a65923310455bd33614abebe964138e","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-04-22 07:16:07.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"f91591c2776b0646d9f14fb9b67cda6417e2e26c","author":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"date":"2020-04-22 07:37:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"138b594269cd5e7b6d4e86659abca6ccd40fd72e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-22 07:38:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/82/721782/1/Cinder-lioadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"a0bcfc28e41a3f4930150b4578c875d6bf11d19c","author":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"date":"2020-04-22 07:50:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fb51b6f3bfb34c4b2ac2809c794a3014ed42e374","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-04-22 08:44:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/721782/3/iscsi/ : SUCCESS in 53m 01s\n- os-brick-smb http://cloudbase-ci.com/os-brick/721782/3/smb/ : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":3},{"id":"78cd62feeb2aa46798721f914de9e744032e196d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-22 09:01:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/82/721782/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b33d4fb : SUCCESS in 55m 15s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/82/721782/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ac323f8 : SUCCESS in 1h 05m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"fb4a884f57772a13aba38090e61b351a2228e7ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 09:59:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c961e7fa9b247378ea2541546fe630c : SUCCESS in 1h 33m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3bdb97cda2040df9dfd3bbe0154d94d : FAILURE in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d73ee5ef8d104564b057c15f2e1879db : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/035e8c93545243f699bfed97e2622d97 : FAILURE in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4c725827db44b01afe65d617ae5f1eb : FAILURE in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01c81ba4c8c24e7487896a0f421161d9 : FAILURE in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfeff00d0c84387945e78c9b883ccdd : SUCCESS in 8m 12s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/e418536ed10c4e3698994dbc70f63c74 : SUCCESS in 1h 22m 26s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e59fcbccbff74d6ba8b2f438151c6c8a : SUCCESS in 2h 01m 09s","accounts_in_message":[],"_revision_number":3},{"id":"bf483bd9815fe01b2d065c7b3a68498b7dd2f966","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-22 10:25:24.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/82/721782/3/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/721782/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6a8661d775120aea3bdcf856f067f380ac0912c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-22 10:28:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3ed5e5371e9da422bdde177ed11bf3ab3ec8b5d7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-22 10:30:56.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/82/721782/3/check/tempest-dsvm-full-bionic/b484aaf/ : SUCCESS in 1h 45m 59s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/82/721782/3/check/devstack-plugin-ceph-dsvm-bionic/74feccb/ : SUCCESS in 1h 56m 27s","accounts_in_message":[],"_revision_number":3},{"id":"3e49013305cbfe435b6593f485989baaf420b351","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-22 10:53:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/82/721782/3/check/DellEMC_Unity_FC_os-brick/fb6a8a8/DellEMC_Unity_FC_os-brick/87 : FAILURE in 27m 16s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/82/721782/3/check/DellEMC_Unity_iSCSI_os-brick/b7278e7/DellEMC_Unity_iSCSI_os-brick/83 : FAILURE in 17m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"31e8d0fe428ce8845bac2757dd8ebff611bd2b0e","author":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"date":"2020-04-22 11:12:27.000000000","message":"Abandoned\n\nssss","accounts_in_message":[],"_revision_number":3},{"id":"e41ab2eedbb4135013a0e5cb4341f4d516051a94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-22 12:56:55.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/82/721782/3/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/82/721782/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3b7842a6161c0440c1e874190cc1c0f4d0432da7","revisions":{"47d02f295ba855e8d96025ec01967b5f28126637":{"kind":"REWORK","_number":1,"created":"2020-04-22 07:15:56.000000000","uploader":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"ref":"refs/changes/82/721782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/721782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/721782/1"}}},"commit":{"parents":[{"commit":"807bd1977722e5283ebe78d27398e6a81f3588f1","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/807bd1977722e5283ebe78d27398e6a81f3588f1"}]}],"author":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2020-04-22 06:14:46.000000000","tz":480},"committer":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2020-04-22 06:28:17.000000000","tz":480},"subject":"Function _connect_single_volume connect volume to virtual machine code logic error","message":"Function _connect_single_volume connect volume to virtual machine code logic error\n\nChange-Id: I6f1dad8112b816fd23f17c1fb900d982ed0f1a4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/47d02f295ba855e8d96025ec01967b5f28126637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/47d02f295ba855e8d96025ec01967b5f28126637"}]},"branch":"refs/heads/master"},"bfbf6ff94a696b7d918e1f67a452563df8d828e4":{"kind":"REWORK","_number":2,"created":"2020-04-22 07:37:12.000000000","uploader":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"ref":"refs/changes/82/721782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/721782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/721782/2"}}},"commit":{"parents":[{"commit":"807bd1977722e5283ebe78d27398e6a81f3588f1","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/807bd1977722e5283ebe78d27398e6a81f3588f1"}]}],"author":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2020-04-22 06:14:46.000000000","tz":480},"committer":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2020-04-22 07:33:57.000000000","tz":480},"subject":"Function _connect_single_volume connect volume to virtual machine code logic error","message":"Function _connect_single_volume connect volume to virtual machine code logic error\n\nCloses-Bug: #1874201\nChange-Id: I6f1dad8112b816fd23f17c1fb900d982ed0f1a4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bfbf6ff94a696b7d918e1f67a452563df8d828e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bfbf6ff94a696b7d918e1f67a452563df8d828e4"}]},"branch":"refs/heads/master"},"3b7842a6161c0440c1e874190cc1c0f4d0432da7":{"kind":"REWORK","_number":3,"created":"2020-04-22 07:50:49.000000000","uploader":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"ref":"refs/changes/82/721782/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/721782/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/721782/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/721782/3"}}},"commit":{"parents":[{"commit":"807bd1977722e5283ebe78d27398e6a81f3588f1","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/807bd1977722e5283ebe78d27398e6a81f3588f1"}]}],"author":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2020-04-22 06:14:46.000000000","tz":480},"committer":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2020-04-22 07:50:17.000000000","tz":480},"subject":"Function _connect_single_volume connect volume to virtual machine code logic error","message":"Function _connect_single_volume connect volume to virtual machine code logic error\n\nCloses-Bug: #1874201\nChange-Id: I6f1dad8112b816fd23f17c1fb900d982ed0f1a4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3b7842a6161c0440c1e874190cc1c0f4d0432da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3b7842a6161c0440c1e874190cc1c0f4d0432da7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
