)]}'
{"id":"openstack%2Fos-brick~725741","triplet_id":"openstack%2Fos-brick~master~I24cab206b16e63859d4886c55d40a03d398ce30d","project":"openstack/os-brick","branch":"master","topic":"tenacity","hashtags":[],"change_id":"I24cab206b16e63859d4886c55d40a03d398ce30d","subject":"Switch from retrying to tenacity","status":"MERGED","created":"2020-05-06 00:54:36.000000000","updated":"2020-06-03 20:28:00.000000000","submitted":"2020-05-18 19:14:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725741-1589829285333-f8705a57","meta_rev_id":"8cf261488e8c872fd9ebd5fa443dabc30abc49ed","_number":725741,"virtual_id_number":725741,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-18 16:17:01.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-05-18 17:47:05.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-05-07 01:11:11.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-18 19:32:21.000000000","post_submit":true,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-18 16:36:42.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 19:14:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-07 01:05:34.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-05-18 19:05:47.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-18 18:10:29.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-18 18:54:23.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-06 19:45:59.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-06 23:20:19.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2020-05-18 18:48:22.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-05-18 17:03:01.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-15 13:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-05-18 14:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-05-18 14:25:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 19:45:59.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-06 23:20:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-05-07 01:05:34.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-05-07 01:11:11.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-15 13:20:07.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-05-18 14:25:51.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-05-18 16:36:42.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 17:03:01.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-05-18 17:47:05.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 18:10:29.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 18:48:22.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-05-18 18:54:23.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 19:05:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 19:14:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 19:32:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f892e0e685a4646f040d9e230fe0cd0e64256dae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 00:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc8a2b6c0b231addb827800aac6548a4735267b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 01:00:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3fc780250aa7e54668865b4142b8970b22f38e5e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 01:09:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/41/725741/1/check/DellEMC_XtremIO_FC_os-brick/15f75a1/DellEMC_XtremIO_FC_os-brick/121 : FAILURE in 10m 24s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/41/725741/1/check/DellEMC_XtremIO_iSCSI_os-brick/e4f897d/DellEMC_XtremIO_iSCSI_os-brick/110 : FAILURE in 14m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"9d074bae3dd93dba453aefeb65e72a124b454469","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 01:16:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"01baccacd46023246624dddb387b7b7f97e9c388","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-06 01:46:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725741/1/iscsi/ : SUCCESS in 51m 23s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725741/1/smb/ : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":1},{"id":"94bad19f45307f67b1f8343bc545a6f9afbcf670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 02:32:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/10c42fe49159436cbd81606d4b737d7a : SUCCESS in 3m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53feefc126244b9589ca248d6fbc6d4d : SUCCESS in 1h 23m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5ba8f2c7612480cb9ee3d07f2963802 : FAILURE in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bcc8f0a728d40739df32c46d1a2bdc8 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed10153bf79496297a97fbadb090ea6 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/444cfa6206294d419807c46e45e364ba : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0664e79127b24a8c807cb2e01bb89009 : SUCCESS in 6m 21s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/e1a4b9eb02f547cebf53b73d72766457 : SUCCESS in 1h 01m 57s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4c256d4e78b2456d91ac789950b6ab17 : SUCCESS in 1h 36m 28s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ff292e1428e94f99a36812415d66f989 : SUCCESS in 1h 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"beed3b7c2c7f47245daefaf2573ae7d34b5c6ed0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-06 02:39:15.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/41/725741/1/check/tempest-dsvm-full-bionic/243bc38/ : SUCCESS in 1h 30m 41s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/41/725741/1/check/devstack-plugin-ceph-dsvm-bionic/9851f04/ : FAILURE in 1h 40m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f0eae161aa9679bf506fd8757e3e8468df10ec20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 02:44:49.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c7173c53862a455411041eae701fabc472e54ad0","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 02:47:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/41/725741/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/75006fe : FAILURE in 1h 52m 57s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/41/725741/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/c85b301 : SUCCESS in 1h 00m 49s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"577efdbc8acf6daab96566280d2965d107cbb80e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 11:55:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44b740b76cef1bdb02c21ab7be95027f0213589c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 12:06:31.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/2/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ba5a4eefc5c5a3d26317a154176b0adf35f5eeed","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-06 12:56:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725741/2/iscsi/ : SUCCESS in 59m 28s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725741/2/smb/ : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9fed13b6f7d3d93227368fc63fdeef6deb834a4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:46:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/87977be818a7425a803102dcea3c8544 : SUCCESS in 3m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ccbbfd9ae06416199a8c5ffb4e75f42 : SUCCESS in 1h 49m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60e5dbc9c544422da32095224a902ac9 : FAILURE in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cf4e8f49c1546b599caad86e0bd5531 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b65b6b3bab04c02810fcae54124d363 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70061a5012b14941b3043fd70a06e07d : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c23b55f8b0248e9b5932278b6e6a33a : SUCCESS in 7m 04s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/6b042667206e4ca9a19365c25adc0c8a : SUCCESS in 1h 12m 34s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bef71c7d69e4930a8e986c0dd720f13 : SUCCESS in 1h 34m 09s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/23b6fe05a2fc4ee8b95f1c00f53e7cfc : SUCCESS in 1h 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf5d4e26eb32327ef7740c116bb2b496afcd0d5d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 14:08:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7a13b66444b4ffba1b3505eaad19c19ddbb30c2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 14:14:41.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"bb871d0f985368073434ff734098a335572515c1","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-06 15:06:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725741/3/iscsi/ : SUCCESS in 56m 29s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725741/3/smb/ : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4580614fe0e9e08611bfec4820a19beb00d2e737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 15:58:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f06019342fa3417d8a58dd21292c2b5a : SUCCESS in 3m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1665f40629447e7a9c0983ed34d0556 : SUCCESS in 1h 18m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c16ef6ca8283459d961c90fe9be6b6a8 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33de72eead24388b042facc761ff8a3 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f475216ede994f20a9a5580655e94656 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afdce059bfcd4ae4a7e2b9372983c46d : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abdf5392a5543e1b699ea0f356df691 : SUCCESS in 7m 26s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/8149ae53bd75432b9b5f37f845132fb3 : SUCCESS in 1h 07m 51s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aac8f325bb814034983a7c670419893c : SUCCESS in 1h 47m 55s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/083d100f6bd7451a956209e417269ad5 : SUCCESS in 1h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa5de6b99b2e277be823650656e01bbf948a1da2","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 16:05:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/41/725741/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/b5bde85 : FAILURE in 1h 00m 06s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/41/725741/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6e468eb : SUCCESS in 1h 03m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"91d3bdea9a6420744b5f5f6d80f62d8c068f96cd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-06 17:03:10.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/41/725741/3/check/tempest-dsvm-full-bionic/02a8370/ : SUCCESS in 2h 15m 27s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/41/725741/3/check/devstack-plugin-ceph-dsvm-bionic/4819179/ : FAILURE in 2h 10m 22s","accounts_in_message":[],"_revision_number":3},{"id":"3cff968e61c0478bd0925ab9055f1ee72475d7ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 17:09:02.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"01721b2644dfd169d98317383d01ad386b2fb4b3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-06 17:18:26.000000000","message":"Patch Set 3:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-41-725741-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"45acfdcba8b9f0ebea5ec0b7bc2f580d37eddec9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 17:24:35.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9a0f138fffa3a621e89447c9719ea903ce858d22","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-05-06 18:08:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/41/725741/3/upstream-solidfire-check/cinder-solidfire-iSCSI/cf14283/ : SUCCESS in 2h 10m 18s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"4b69bc0c50a6440c7165e0a93f88a3406f950b52","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 19:29:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_XtremIO_FC_os-brick/4e79e90/DellEMC_XtremIO_FC_os-brick/123 : FAILURE in 9m 57s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_XtremIO_iSCSI_os-brick/7917756/DellEMC_XtremIO_iSCSI_os-brick/2 : FAILURE in 9m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"286d4088e61f3231ba728b93baf76c1378552301","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 19:34:54.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"5b4c13c0f34af1381d8a2dd79d79cf484b7a3400","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 19:45:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/41/725741/3/upstream-check/cinder-cDOT-iSCSI/9b36cbf/ : SUCCESS in 45m 46s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/41/725741/3/upstream-check/cinder-cDOT-iSCSI-IPv6/0c49c36/ : SUCCESS in 1h 38m 23s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/41/725741/3/upstream-check/cinder-cDOT-NFS/144d610/ : SUCCESS in 44m 22s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/41/725741/3/upstream-check/cinder-cDOT-NFS-IPv6/f7ff5ff/ : SUCCESS in 1h 33m 22s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/41/725741/3/upstream-check/cinder-cDOT-FCP/8649e58/ : FAILURE in 3h 47m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"65738823634c4ed3c40286503789fab3b285637d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-06 23:20:19.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_Unity_FC_os-brick/e23eee9/DellEMC_Unity_FC_os-brick/104 : SUCCESS in 1h 28m 16s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_Unity_iSCSI_os-brick/af1af42/DellEMC_Unity_iSCSI_os-brick/99 : FAILURE in 1h 55m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b6b7053bdeac48cc48c98a148b47a8842bf2ae8d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-06 23:26:07.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"caaed78b9692c85258edc1d914ce0661edfb7398","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-07 00:37:25.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_VNX_FC_os-brick/9b40316/DellEMC_VNX_FC_os-brick/80 : SUCCESS in 1h 05m 43s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_VNX_iSCSI_os-brick/d7ce10a/DellEMC_VNX_iSCSI_os-brick/79 : SUCCESS in 1h 39m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"191f898556259eb619c27a454e9efe41b221c981","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-07 00:54:07.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6bc5881828806731dd0ec5599c61ff74325d5af7","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-05-07 01:05:34.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_SC_FC-os-brick/1ed8f08/DellEMC_SC_FC-os-brick/92 : SUCCESS in 1h 10m 02s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_SC_iSCSI-os-brick/6128ffc/DellEMC_SC_iSCSI-os-brick/89 : FAILURE in 2h 18m 56s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"f264e87106d8746837136f3d0af9fbec2b4bbca9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-07 01:11:11.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/41/725741/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c30897f43198c31afa1f581876260dc434340626","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-15 11:03:54.000000000","message":"Patch Set 3:\n\nWhy don\u0027t we remove retrying from dependencies?","accounts_in_message":[],"_revision_number":3},{"id":"1e856a85ef4ec3d1564613f5ecb0995eb7831c57","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-15 12:31:48.000000000","message":"Patch Set 3:\n\n\u003e Why don\u0027t we remove retrying from dependencies?\n\nIt\u0027s being removed from requirements.txt and lower-constraints.txt. Is there someone else you are referring to?","accounts_in_message":[],"_revision_number":3},{"id":"5c3e74bf3d144b7db24ca017d2831fdecb5d8861","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-15 13:20:07.000000000","message":"Patch Set 3: Code-Review+2\n\nI\u0027m sorry, Sean, I missed that somehow:(","accounts_in_message":[],"_revision_number":3},{"id":"a646c0302839dfed1dd78b7d90656465b7ec005b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-05-18 14:25:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1f9eb7cc599dd770327f92489b2e7f933b875dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 14:26:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fb4ab57a715862850c426ae02f3563c78b7a4126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:15:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/868abf122a6044eaa7e6365d8efd67d7 : SUCCESS in 6m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9295a80c2d8b421f8fe44dadd48d2ca5 : SUCCESS in 1h 18m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e0aef8c5fb04f94b3f1319d413a0af6 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23968a9b23b424c88ed0cc1f2d266bf : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da5241519e56496c9954660679e1e6f2 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1f2656f96bb443891604048099b0610 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb34690d36a45deb39409d5caa377bd : SUCCESS in 8m 25s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/17177ecdb7384f6a9db6a8347a95a5b0 : FAILURE in 1h 47m 34s","accounts_in_message":[],"_revision_number":3},{"id":"75909b03ec281a00eb21bbd15d1ad45175e931a7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-18 16:17:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"621174a4a93dbbc46f78458296648bc36917e5c4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-18 16:36:42.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/41/725741/3/check/tempest-dsvm-full-bionic/73ea434/ : FAILURE in 15m 35s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/41/725741/3/check/devstack-plugin-ceph-dsvm-bionic/5b4958b/ : FAILURE in 14m 49s","accounts_in_message":[],"_revision_number":3},{"id":"529f9fd6c206d99ecc8b36502f9d871ede5cfe6d","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-18 17:03:01.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725741/3/iscsi/ : FAILURE in 45m 31s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725741/3/smb/ : FAILURE in 45m 32s","accounts_in_message":[],"_revision_number":3},{"id":"64b95b35782b3a5d05ccbc4dfe96f90e8fca76e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 17:46:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9c9be1d69944eafacbcc9095ddeec60 : SUCCESS in 5m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d752334dff044dab169480b3c810a0f : SUCCESS in 1h 29m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a3ed3aed7494ea6ac9f2454922cebf6 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0662386f6f6946449e3fdf7dc5d214a7 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa1f65482305435aa9c5178f853510fa : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d241bee1fd14f3ea43e5155d24b4427 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4a4fda03c64537a28ca3c0172df1df : SUCCESS in 7m 20s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/8a929708f9d74a86b3291a63b272e486 : SUCCESS in 1h 07m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/316bd7619ae7402f82a88d0e97ae9287 : SUCCESS in 1h 21m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/4464fc317540444f86d6e16cddfa85d5 : SUCCESS in 1h 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bfb233e903c67b5fc5f032ba90263780cba3618e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 17:47:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8c2665ffc3c7fc185bf3277455608e086c99b755","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 17:47:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_XtremIO_FC_os-brick/7e02fb9/DellEMC_XtremIO_FC_os-brick/None : RETRY_LIMIT\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_XtremIO_iSCSI_os-brick/2d130f5/DellEMC_XtremIO_iSCSI_os-brick/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"c45fab13ca41728c4f13ae24cf921f5a392ea6bf","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-18 18:10:29.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/41/725741/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9458435 : FAILURE in 1h 00m 33s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/41/725741/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/72e1676 : FAILURE in 1h 53m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"702d6ff9eb148c9ef1b547ffcca419b863567f27","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-05-18 18:48:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/41/725741/3/upstream-solidfire-check/cinder-solidfire-iSCSI/8f8482f/ : SUCCESS in 1h 01m 00s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"462fe2f074848e102205466f735e20a149731fa9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 18:54:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_VNX_FC_os-brick/fc2f346/DellEMC_VNX_FC_os-brick/1 : SUCCESS in 1h 07m 17s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_VNX_iSCSI_os-brick/f64aad5/DellEMC_VNX_iSCSI_os-brick/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"ad86dce6fade3c4f6cc1e8074660343c0354db88","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-18 19:05:47.000000000","message":"Patch Set 3:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-41-725741-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"b91625aada326446ef845f0282c0bcf727e99ca1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 19:14:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"40884894fdc117e649c0dc8ae44b1ebcae6c8fa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 19:14:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8e45372f9fd4e6685728fb028f3ea6b : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9122d85769024c588b780dd95fe05576 : SUCCESS in 1h 25m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0caca887c4e4e98802e343de67779c9 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89fb8fa2d43f44dc91b6e582173d8ca9 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8485615ddd1c423b8b11bca6b2abdc36 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3759108f06247768b56a7223f565d74 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d90543e8c54bc4a1a8f043893f3fa7 : SUCCESS in 6m 30s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bf9675cf64c4a8eb5277d61167e330f : SUCCESS in 1h 25m 54s","accounts_in_message":[],"_revision_number":3},{"id":"b49e1c28a60d339ead8d391b4fb3fb2b76162ffe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 19:16:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a18e6d0d4d484aa7be68924e32724e : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3},{"id":"277aa9256b8d997a563f364657dfefbe3ba516a9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 19:32:21.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/41/725741/3/check/DellEMC_VxFlexOS_os-brick/b24e516/DellEMC_VxFlexOS_os-brick/1 : SUCCESS in 1h 45m 13s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"c84fbcf677bad28a0c562cd4afac0977cafc55df","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 20:28:00.000000000","message":"Created a revert of this change as Ieaaf03387685a4be2b709b48660483f71d41d461","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3f1314674d59bca6e9d0e43d4d479bb04f20aca1","revisions":{"f3ca996e9401f13b15bbdf4fefabb496f6d61e82":{"kind":"REWORK","_number":1,"created":"2020-05-06 00:54:36.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/725741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/41/725741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/41/725741/1"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 00:52:12.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 00:52:12.000000000","tz":-300},"subject":"Switch from retrying to tenacity","message":"Switch from retrying to tenacity\n\nThis switching our retry mechanism over from the retrying library to the\ntenacity library. Retrying has been active for a few years now and\nappears to be no longer maintained.\n\nThis has a small behavior change in that before we were applying an\nexponential backoff to the first time a retry was needed. This no longer\nhappens, but retries will exponentially back off with each retry.\n\nAlso cleaned up some minor nits with unit test assert argument order.\n\nCloses-bug: #1635397\n\nChange-Id: I24cab206b16e63859d4886c55d40a03d398ce30d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f3ca996e9401f13b15bbdf4fefabb496f6d61e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f3ca996e9401f13b15bbdf4fefabb496f6d61e82"}]},"branch":"refs/heads/master"},"6bc92e85f265bde4a745f2dad72e0eb82aaf4931":{"kind":"REWORK","_number":2,"created":"2020-05-06 11:55:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/725741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/41/725741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/41/725741/2"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 00:52:12.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 11:55:42.000000000","tz":-300},"subject":"Switch from retrying to tenacity","message":"Switch from retrying to tenacity\n\nThis switching our retry mechanism over from the retrying library to the\ntenacity library. Retrying has been active for a few years now and\nappears to be no longer maintained.\n\nThis has a small behavior change in that before we were applying an\nexponential backoff to the first time a retry was needed. This no longer\nhappens, but retries will exponentially back off with each retry.\n\nAlso cleaned up some minor nits with unit test assert argument order.\n\nCloses-bug: #1635397\n\nChange-Id: I24cab206b16e63859d4886c55d40a03d398ce30d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6bc92e85f265bde4a745f2dad72e0eb82aaf4931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6bc92e85f265bde4a745f2dad72e0eb82aaf4931"}]},"branch":"refs/heads/master"},"3f1314674d59bca6e9d0e43d4d479bb04f20aca1":{"kind":"REWORK","_number":3,"created":"2020-05-06 14:08:07.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/725741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/41/725741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/41/725741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/41/725741/3"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 00:52:12.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 14:08:01.000000000","tz":-300},"subject":"Switch from retrying to tenacity","message":"Switch from retrying to tenacity\n\nThis switching our retry mechanism over from the retrying library to the\ntenacity library. Retrying has been active for a few years now and\nappears to be no longer maintained.\n\nThis has a small behavior change in that before we were applying an\nexponential backoff to the first time a retry was needed. This no longer\nhappens, but retries will exponentially back off with each retry.\n\nAlso cleaned up some minor nits with unit test assert argument order.\n\nCloses-bug: #1635397\n\nChange-Id: I24cab206b16e63859d4886c55d40a03d398ce30d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f1314674d59bca6e9d0e43d4d479bb04f20aca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f1314674d59bca6e9d0e43d4d479bb04f20aca1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
