)]}'
{"id":"openstack%2Fos-brick~725834","triplet_id":"openstack%2Fos-brick~master~Ic9b2b425a4910809a21bd250a4d9730a30fadf22","project":"openstack/os-brick","branch":"master","topic":"lower-constraints","hashtags":[],"change_id":"Ic9b2b425a4910809a21bd250a4d9730a30fadf22","subject":"Update lower-constraints versions","status":"MERGED","created":"2020-05-06 11:54:54.000000000","updated":"2020-06-15 20:12:55.000000000","submitted":"2020-06-15 20:07:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725834-1592251650651-2f414a6f","meta_rev_id":"8000af049864a5a065e797c4adff5927fd4a3a02","_number":725834,"virtual_id_number":725834,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-10 02:45:43.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 06:55:02.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 21:27:07.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 20:07:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 11:55:28.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-06-09 20:57:43.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 08:09:44.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 23:13:38.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 22:57:44.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2020-06-09 21:43:40.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-06-09 19:38:25.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-10 20:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-06-15 18:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-06-15 18:26:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 22:57:46.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 19:38:25.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-06-09 20:57:43.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 21:27:07.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 21:43:40.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-06-09 22:57:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-06-09 23:13:38.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 02:45:43.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 06:55:02.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 08:09:44.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 11:55:28.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-06-10 20:48:02.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-15 18:26:38.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-06-15 20:07:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce499646a74d23e76fdbf3a895ea9f8549ec1bb4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 11:54:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef749921215e0f415f7ba7803d0771b09dc1e696","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-06 12:47:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725834/1/iscsi/ : FAILURE in 52m 09s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725834/1/smb/ : FAILURE in 50m 34s","accounts_in_message":[],"_revision_number":1},{"id":"02ad7933f4bc1017b42713829cfb98841c23b609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:22:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d06dacc19cea48dd9b21cd2b21fa11fb : FAILURE in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d93fd3027d20446bb402af44b1e9108f : SUCCESS in 1h 26m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/734dac94b433450b84f1559b781022b6 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2cf94d72f841bbb6a7e46a9e4d37e5 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/680a8d32074d4d8ea03d27ce0e8d472f : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5acf0b3ea7884270b85a0a5d15faff7a : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6338f5616684454a85f5690be1498e6 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae05fc99b56641e59b276fbfd2784789 : SUCCESS in 5m 50s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/8f80d4e06cf94c67980bc3b51863eb06 : SUCCESS in 59m 07s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7152e3d5e6b9468ba0cc314ea3c67494 : SUCCESS in 1h 17m 55s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5b6743cf9e684346923451ffca4f8931 : SUCCESS in 1h 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"386ecffa1d23f5dff1844f11d1b1a579fe9d7fcc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 13:32:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe215a0fbf84e3f09dc567da4031dbcbfe3f2d0c","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-06 14:21:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725834/2/iscsi/ : FAILURE in 48m 51s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725834/2/smb/ : FAILURE in 48m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3cd9a970d8d85400964608924d4a70753b089d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 15:08:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/58cb44ab072741ddb4e73d34d1de75e4 : SUCCESS in 6m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8154603758de4d1bb2d80c691ed9a99a : SUCCESS in 1h 34m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e75217534e1a4ab79badd2654389d3d8 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12020e85c5c3416e8e84951524f8d828 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad7f684cfd7d418e9af3c881262b4c63 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d17efb8d20d24a9a900ea56e7d6410c6 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984db2cd5ebf439da3237900dbe8ceb1 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4426c857b346444a808a75f53ff76311 : SUCCESS in 6m 49s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/1c3b0e6335734eac8a20dc433e013917 : SUCCESS in 1h 05m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a307bbcbf5a464ebf9c82d220eb4ae5 : SUCCESS in 1h 33m 43s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/42d5ca1eeb904a0e9e787ae10ae33203 : SUCCESS in 1h 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"443080dd42f63fa7d8c8b70baa215d5b7c1f19c5","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 15:20:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/34/725834/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/2dbb7cd : SUCCESS in 55m 26s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/34/725834/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6d2cf4a : SUCCESS in 1h 07m 20s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"56454025f0ac145fbb5b94895fe507ae08d834cb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-06 15:51:48.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/34/725834/2/check/tempest-dsvm-full-bionic/0b60ed5/ : SUCCESS in 1h 51m 59s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/34/725834/2/check/devstack-plugin-ceph-dsvm-bionic/5264022/ : FAILURE in 1h 52m 27s","accounts_in_message":[],"_revision_number":2},{"id":"59ff310be3a4e35c03d078550038107f379a0c16","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-06 16:11:32.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-34-725834-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"2c193f4e7e1f25e7a5f3de35f04fc24f2c911ad3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-05-06 16:17:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/34/725834/2/upstream-solidfire-check/cinder-solidfire-iSCSI/18c5426/ : SUCCESS in 59m 05s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"cd5d33bddd6a2febe24dada6ca9aba57d13d9bff","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 18:26:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/34/725834/2/upstream-check/cinder-cDOT-iSCSI/99a13f0/ : SUCCESS in 48m 20s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/34/725834/2/upstream-check/cinder-cDOT-iSCSI-IPv6/463b5bf/ : SUCCESS in 1h 53m 57s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/34/725834/2/upstream-check/cinder-cDOT-NFS/9f3fb32/ : SUCCESS in 46m 41s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/34/725834/2/upstream-check/cinder-cDOT-NFS-IPv6/27c84a0/ : SUCCESS in 1h 40m 23s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/34/725834/2/upstream-check/cinder-cDOT-FCP/7f981f6/ : FAILURE in 3h 17m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"70eb069fd0adb953005543d989295ca77bdb3edb","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 19:19:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_XtremIO_FC_os-brick/7e21721/DellEMC_XtremIO_FC_os-brick/122 : FAILURE in 11m 05s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_XtremIO_iSCSI_os-brick/aadfdd2/DellEMC_XtremIO_iSCSI_os-brick/1 : FAILURE in 13m 09s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"6647d80f269df9c1fd920b1f3002229a830f1adb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-06 21:24:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_Unity_FC_os-brick/2e52b3b/DellEMC_Unity_FC_os-brick/103 : SUCCESS in 1h 34m 10s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_Unity_iSCSI_os-brick/4673555/DellEMC_Unity_iSCSI_os-brick/98 : SUCCESS in 1h 42m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e9e32824281d7d603a1cf904c477aadfabc37d5d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-05-06 22:46:37.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_SC_FC-os-brick/c287c4d/DellEMC_SC_FC-os-brick/91 : SUCCESS in 1h 14m 15s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_SC_iSCSI-os-brick/10621b0/DellEMC_SC_iSCSI-os-brick/88 : SUCCESS in 2h 22m 40s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"0da7efef1d71774d66825a2012fd02f630b817f2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-06 22:57:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_VNX_FC_os-brick/49b56e2/DellEMC_VNX_FC_os-brick/79 : SUCCESS in 1h 30m 30s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/34/725834/2/check/DellEMC_VNX_iSCSI_os-brick/c6f555d/DellEMC_VNX_iSCSI_os-brick/78 : SUCCESS in 1h 52m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2f2271bfc2526d7361ae28d512aba344a40ab2fc","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-09 14:24:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"12971be52a9339141b05a1226688bf1b4e965787","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-09 14:25:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5c1ad4e58b1ea27c3e548124788963414ade1b0e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 14:34:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6f3759c281ef71c1897ca6a6253057ce004a664f","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-09 15:31:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725834/4/iscsi/ : SUCCESS in 56m 01s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725834/4/smb/ : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":4},{"id":"57352d3286f36bfe57b9c190e621ca524a47013f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-09 15:55:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"281f49f5f35233e7f5f945591669ead1da4dbaaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:19:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/75d25447190f43f0affe7a35a9968a2b : SUCCESS in 5m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03f7bdcd82174d70a5ca0c51c22f1aee : SUCCESS in 1h 08m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4ce821471f44f8597fe4c04e6144891 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517003b1567d487988489469fb05484a : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5adeb76872b946c491c752761b998f0f : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4b3886b670549aaa0275cce3ae8df7b : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59747e1310b148d4927e3cbacf8a9d52 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2075884049d54f358caeacbfbe3447c5 : SUCCESS in 6m 40s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/d0a9aa4df59c43a49e6457ecf1ba894a : SUCCESS in 1h 11m 17s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/da6832fb840347609b67353e40e07dd4 : SUCCESS in 1h 25m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/874ace6f7f074e61874e795b9703f97b : SUCCESS in 1h 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b75ac9d6e99cee87b63e81dc07a2ffdc7b38eb86","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-09 18:07:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/34/725834/4/upstream-solidfire-check/cinder-solidfire-iSCSI/ac6211a/ : SUCCESS in 57m 48s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"847f2ed0e65d49a22326bf4076081c2c3f1a9607","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-09 18:31:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/34/725834/4/check/tempest-dsvm-full-bionic/b47c22e/ : SUCCESS in 1h 30m 38s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/34/725834/4/check/devstack-plugin-ceph-dsvm-bionic/a6ce5e3/ : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":4},{"id":"eb918d77122e467f50103a8ee4434cc9871386d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 18:41:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"570ec999a3650fa1bff26f8dc08871606c8f43d3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-09 19:37:58.000000000","message":"Patch Set 4:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-34-725834-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"8db11a56a9c60dc86deb25dc4685fece80721c22","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-09 19:38:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/725834/5/iscsi/ : SUCCESS in 56m 36s\n- os-brick-smb http://cloudbase-ci.com/os-brick/725834/5/smb/ : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":5},{"id":"a43bd68aa38719bcbf83e4c05ccf44c781c25559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 19:48:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/630049b16034484181e931f18705fe26 : SUCCESS in 3m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bd715b55f0e4dcd915d9c5c32968c6c : SUCCESS in 1h 03m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a77a7ecb227c44bd87c2cf93ea111b96 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0aa01201e194fd7af4f964ff13079e1 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f4034bd465646fa897cd14fd2e18ed6 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a686e9cac4b49e4af75212f8cf55a47 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4425d64a83974306a2992104239fce18 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60f73714603d47e3b9fad11c674d2c04 : SUCCESS in 5m 25s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/3c0d3925abfe4e1c9b42dcaab1cff048 : SUCCESS in 53m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5e2c756b6d764db39419959d751f07fd : SUCCESS in 1h 05m 25s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/69ce6ce007b54893af603e3a132f8939 : SUCCESS in 52m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a1bdfb62b14e60ff64afcd6f542cfa362a0a7df1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-09 20:57:43.000000000","message":"Patch Set 5:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-34-725834-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"fd1ef21d0295680445632def81fb644d43d322ae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-09 21:27:07.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/34/725834/5/check/tempest-dsvm-full-bionic/cb7df6c/ : SUCCESS in 1h 31m 56s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/34/725834/5/check/devstack-plugin-ceph-dsvm-bionic/da42d96/ : SUCCESS in 1h 56m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3347c8aee02a6f2510455df31261d2e74aa94cba","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-09 21:43:40.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/34/725834/5/upstream-solidfire-check/cinder-solidfire-iSCSI/784cfbe/ : FAILURE in 1h 54m 44s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"d6801d0deed63cee136bb03f0fdddded94365199","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-09 22:57:44.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/34/725834/5/check/DellEMC_Unity_FC_os-brick/aeb652b/DellEMC_Unity_FC_os-brick/16 : SUCCESS in 1h 26m 00s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/34/725834/5/check/DellEMC_Unity_iSCSI_os-brick/4406659/DellEMC_Unity_iSCSI_os-brick/14 : SUCCESS in 1h 31m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"27fa7401ae452fcf8dbe8982cfbee4ae1cd53901","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-09 23:13:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/34/725834/5/upstream-check/cinder-cDOT-iSCSI/a632b68/ : SUCCESS in 3h 24m 46s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/34/725834/5/upstream-check/cinder-cDOT-iSCSI-IPv6/26270d3/ : SUCCESS in 1h 29m 05s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/34/725834/5/upstream-check/cinder-cDOT-NFS/bec5f15/ : SUCCESS in 42m 16s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/34/725834/5/upstream-check/cinder-cDOT-NFS-IPv6/3a63336/ : SUCCESS in 1h 34m 05s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/34/725834/5/upstream-check/cinder-cDOT-FCP/53f8458/ : FAILURE in 3h 20m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"e4884c12f639bdaca4bdc417ccd77d682bb65d27","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-10 02:45:43.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/34/725834/5/check/DellEMC_XtremIO_FC_os-brick/aff3e77/DellEMC_XtremIO_FC_os-brick/17 : FAILURE in 7m 43s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/34/725834/5/check/DellEMC_XtremIO_iSCSI_os-brick/13b84f4/DellEMC_XtremIO_iSCSI_os-brick/16 : FAILURE in 7m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"7998e29e3161f95342b3dc735bde824c7bff55a8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-10 06:55:02.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/34/725834/5/check/DellEMC_VxFlexOS_os-brick/7370edf/DellEMC_VxFlexOS_os-brick/14 : SUCCESS in 1h 36m 23s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"8e36acdfe076be3a8f1da89b316bf819c230f897","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-10 08:09:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/34/725834/5/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c329820 : SUCCESS in 1h 00m 01s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/34/725834/5/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/f45cd11 : SUCCESS in 1h 05m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":5},{"id":"2ef0fc902726e62d2f6211692745057263170855","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-06-10 11:55:28.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/34/725834/5/check/DellEMC_SC_FC-os-brick/c59325e/DellEMC_SC_FC-os-brick/12 : FAILURE in 1h 22m 11s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/34/725834/5/check/DellEMC_SC_iSCSI-os-brick/c54f586/DellEMC_SC_iSCSI-os-brick/11 : FAILURE in 2h 44m 01s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"47c59f26fd9f74b3bdfab7daf4673a03dfee1fc8","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-10 20:48:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dc1ce8eb2633a404cf867f56f77ea68b6caca663","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-15 18:26:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b6b5483f920e5d70ad3065229ace86ca0b2d1ecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 18:26:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0608d387e6e23c3a9a35bd4575aa78a521080c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:07:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2b211f9fac5e6989c13f55615b54524104a3622a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:07:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ad5657a92af3475a864774df29502971 : SUCCESS in 6m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ef1e9d5dae54e249cd4938a2a5969bb : SUCCESS in 1h 33m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/802354bc75854e708bfbfec50cd37f97 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d77a81c268844e69cd8e129c97b889d : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0134786c8b5d4981887bc576e362e4bf : SUCCESS in 13m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eb46f9fb8d740b7a40385315411ba45 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64ca4e5a8634a3bab4398c06d38632c : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31f9f378442d4260904d8417ff8be7e4 : SUCCESS in 7m 19s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/97e85146aaea4eee88e62bed26cd0a5c : SUCCESS in 1h 16m 55s","accounts_in_message":[],"_revision_number":5},{"id":"920edddfc5bb1251db61d9b7d93b814b94562cc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:12:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5155ea9cd69342faa41cc407324ce04c : SUCCESS in 4m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5065211d7e7942ef97dca2224d2c8f32 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d95ebc4ec214bd1db368100b4b53cccefc4ab90e","revisions":{"0bd077ca175ee5d3fd27e48fd8c7e5da4192069a":{"kind":"REWORK","_number":1,"created":"2020-05-06 11:54:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/725834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/34/725834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/34/725834/1"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 11:50:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 11:54:45.000000000","tz":-300},"subject":"Raise lower constraints to meet our minimums","message":"Raise lower constraints to meet our minimums\n\nPython 3.6 is now our minimum Python runtime version to support.\nThis raises several lower constraints versions to the minimum that\nsupports 3.6 or later.\n\nThis also raises greenlets and lxml versions to account for updates to\ncpython changes that break installation when installing under 3.7.\n\nChange-Id: Ic9b2b425a4910809a21bd250a4d9730a30fadf22\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0bd077ca175ee5d3fd27e48fd8c7e5da4192069a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0bd077ca175ee5d3fd27e48fd8c7e5da4192069a"}]},"branch":"refs/heads/master"},"87f79d1fc28a3b72c85e0a3ccf9b494b8a6a765e":{"kind":"REWORK","_number":2,"created":"2020-05-06 13:32:40.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/725834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/34/725834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/34/725834/2"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 11:50:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 13:32:34.000000000","tz":-300},"subject":"Raise lower constraints to meet our minimums","message":"Raise lower constraints to meet our minimums\n\nPython 3.6 is now our minimum Python runtime version to support.\nThis raises several lower constraints versions to the minimum that\nsupports 3.6 or later.\n\nThis also raises greenlets and lxml versions to account for updates to\ncpython changes that break installation when installing under 3.7.\n\nChange-Id: Ic9b2b425a4910809a21bd250a4d9730a30fadf22\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/87f79d1fc28a3b72c85e0a3ccf9b494b8a6a765e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/87f79d1fc28a3b72c85e0a3ccf9b494b8a6a765e"}]},"branch":"refs/heads/master"},"4ea7fcdfbb59ad3a01d2ac24161c312b3fc75c19":{"kind":"REWORK","_number":3,"created":"2020-06-09 14:24:07.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/34/725834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/34/725834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/34/725834/3"}}},"commit":{"parents":[{"commit":"9b71c7948fb399071d731d3d877f867f605926ad","subject":"Merge \"Add iscsi-initiator-utils requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9b71c7948fb399071d731d3d877f867f605926ad"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 11:50:47.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-06-09 14:21:54.000000000","tz":-240},"subject":"Raise lower constraints to meet our minimums","message":"Raise lower constraints to meet our minimums\n\nPython 3.6 is now our minimum Python runtime version to support.\nThis raises several lower constraints versions to the minimum that\nsupports 3.6 or later.\n\nThis also raises greenlets and lxml versions to account for updates to\ncpython changes that break installation when installing under 3.7.\n\nChange-Id: Ic9b2b425a4910809a21bd250a4d9730a30fadf22\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4ea7fcdfbb59ad3a01d2ac24161c312b3fc75c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4ea7fcdfbb59ad3a01d2ac24161c312b3fc75c19"}]},"branch":"refs/heads/master"},"f03bd25e2db1df2724d52c2e631fee2f9965df2e":{"kind":"REWORK","_number":4,"created":"2020-06-09 14:34:02.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/725834/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/34/725834/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/34/725834/4"}}},"commit":{"parents":[{"commit":"9b71c7948fb399071d731d3d877f867f605926ad","subject":"Merge \"Add iscsi-initiator-utils requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9b71c7948fb399071d731d3d877f867f605926ad"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 11:50:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-09 14:33:42.000000000","tz":-300},"subject":"Update lower-constraints versions","message":"Update lower-constraints versions\n\nThis updates lower constraints to versions that will work with py38 so\nthat when we move to running on focal nodes, which has py38 as its\ndefault py3 runtime, the lower-constraints job will continue to pass.\n\nIt also cleans out some secondary requirements that are no longer needed\ndue to our direct dependencies being updated.\n\nLinters are removed that are kept in the global requirements blacklist\nas those are not version tracked and are not relevant for our\nlower-constraints unit test runs.\n\nChange-Id: Ic9b2b425a4910809a21bd250a4d9730a30fadf22\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f03bd25e2db1df2724d52c2e631fee2f9965df2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f03bd25e2db1df2724d52c2e631fee2f9965df2e"}]},"branch":"refs/heads/master"},"d95ebc4ec214bd1db368100b4b53cccefc4ab90e":{"kind":"REWORK","_number":5,"created":"2020-06-09 18:41:27.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/725834/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/34/725834/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/34/725834/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/34/725834/5"}}},"commit":{"parents":[{"commit":"70b3925f0d2f1347516e9a1f59e41890f65cfa09","subject":"Merge \"Add oslo.context dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/70b3925f0d2f1347516e9a1f59e41890f65cfa09"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-06 11:50:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-09 18:39:16.000000000","tz":-300},"subject":"Update lower-constraints versions","message":"Update lower-constraints versions\n\nThis updates lower constraints to versions that will work with py38 so\nthat when we move to running on focal nodes, which has py38 as its\ndefault py3 runtime, the lower-constraints job will continue to pass.\n\nIt also cleans out some secondary requirements that are no longer needed\ndue to our direct dependencies being updated.\n\nLinters are removed that are kept in the global requirements blacklist\nas those are not version tracked and are not relevant for our\nlower-constraints unit test runs.\n\nChange-Id: Ic9b2b425a4910809a21bd250a4d9730a30fadf22\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d95ebc4ec214bd1db368100b4b53cccefc4ab90e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d95ebc4ec214bd1db368100b4b53cccefc4ab90e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
