)]}'
{"id":"openstack%2Fos-brick~726467","triplet_id":"openstack%2Fos-brick~master~Ie85b11bc9c9474ba4dbd92e5afa6d996aa19580a","project":"openstack/os-brick","branch":"master","topic":"retype-encrypted","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-09-08 16:09:31.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_4523\u003e using the hovercard menu","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie85b11bc9c9474ba4dbd92e5afa6d996aa19580a","subject":"Retype from non encrypted to encrypted volume is not supported","status":"NEW","created":"2020-05-08 20:11:21.000000000","updated":"2023-09-08 16:09:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d4fcf2cbbcfb0820f6e59a838cfb5e2bcb6d1f3a","_number":726467,"virtual_id_number":726467,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-06-30 20:24:54.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"date":"2020-07-01 01:16:04.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-30 19:57:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"date":"2020-06-30 22:03:16.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"date":"2020-06-30 21:06:27.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"date":"2020-06-30 21:35:09.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"date":"2020-06-30 21:30:33.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"date":"2020-06-30 19:21:23.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":-1,"date":"2020-07-29 19:59:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-07-30 05:25:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-07-30 05:25:02.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 17:29:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-05-11 17:29:17.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-11 17:32:17.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-05-12 00:09:01.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-12 01:46:01.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-12 13:00:07.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-15 23:10:24.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-16 06:31:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-06-08 13:43:30.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-06-30 19:21:23.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-06-30 19:57:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 20:24:54.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 21:06:27.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 21:30:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-06-30 21:35:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 22:03:16.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-07-01 01:16:04.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-29 19:59:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"984e9db2b4cb3f28b13bffc6cbf658f47bace142","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-05-08 20:11:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5952e8c58b7f8eae0ea4d4fb1527e3283a173401","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-08 20:33:24.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/67/726467/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/67/726467/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7c735081e5bd03278072898421ae3ad4044f7f96","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-08 20:53:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/726467/1/iscsi/ : SUCCESS in 41m 38s\n- os-brick-smb http://cloudbase-ci.com/os-brick/726467/1/smb/ : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2a97de052b61222cd22c8ab71cd3feaa00b4b40d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-08 22:28:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/67/726467/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/7d51fb4 : SUCCESS in 1h 06m 37s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/67/726467/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5bff60e : FAILURE in 2h 05m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"91329dbd7e4fbefae5cf14d61c45349caf18764c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-08 22:47:55.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/1/check/tempest-dsvm-full-bionic/ae52d50/ : SUCCESS in 1h 50m 30s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/1/check/devstack-plugin-ceph-dsvm-bionic/3244b87/ : FAILURE in 1h 43m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a7a79a2d1882643926977f6cf5a2095b1918f8aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-08 22:59:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/67/726467/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/67/726467/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5fb9108deece40a0bea6a503766d18a02ba6c2c4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-09 02:41:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/67/726467/1/check/DellEMC_XtremIO_FC_os-brick/0722806/DellEMC_XtremIO_FC_os-brick/127 : FAILURE in 11m 02s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/67/726467/1/check/DellEMC_XtremIO_iSCSI_os-brick/a1cc986/DellEMC_XtremIO_iSCSI_os-brick/8 : FAILURE in 11m 24s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"07b91a6547f76c763e897462d0b69bee84585a29","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-09 02:52:46.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/67/726467/1/Cinder-lioadm : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/67/726467/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"93af06ddc106d25a602dc490c16a9165c1a46121","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-05-11 17:30:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6a3a38432ec90e5e1f466609a041b67e5a23c12e","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-11 17:31:29.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/726467/2/iscsi/ : FAILURE in 46s\n- os-brick-smb http://cloudbase-ci.com/os-brick/726467/2/smb/ : FAILURE in 45s","accounts_in_message":[],"_revision_number":2},{"id":"01e71f756ac7652e3387a91cf1804fefb7c1c846","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-11 18:59:22.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/e0f43291f82148e68d6975beaa4642b9 : POST_FAILURE in 18m 02s\n- devstack-plugin-ceph-dsvm-bionic finger://zuul-v3-executor.novalocal/3087d2eb6f7f42ee8c05349e00106df8 : POST_FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":2},{"id":"18bf80f8acfe100093d23d7ecfb6c00a495c3681","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-11 19:10:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/67/726467/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/770a58e : SUCCESS in 1h 01m 19s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/67/726467/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/414839f : SUCCESS in 1h 14m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"0b90bb043f46b4223631ae6e5f7c65a5a2693bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 19:14:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ef6848f05ac4b4cb1ee39bdb9735d47 : SUCCESS in 1h 23m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d75c17cd6d214617ad4500c8cd23c7af : SUCCESS in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd43c707e0f4615807664ab57c91310 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfa92c9c48374d1987644007a1f471b5 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d5d2bda176447d2b19100cd784c501f : SUCCESS in 14m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ed968b094445d3a0ab2b561d59ed94 : SUCCESS in 9m 05s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/fa8cdd79dd674f45b5dab6e9f961e0c4 : SUCCESS in 1h 12m 18s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6b4ed482602e4e8984475f210ad79746 : SUCCESS in 1h 36m 52s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cf8b94f1fed44b35b2de6e691135872d : SUCCESS in 1h 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e7f89da1714ec85760aec203ddbd65638108cb6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-05-11 20:21:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/67/726467/2/upstream-solidfire-check/cinder-solidfire-iSCSI/8333fa9/ : SUCCESS in 1h 06m 49s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"33b5ba58182a24780e723f9143054d9acf9ab63b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-11 20:29:16.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-67-726467-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"251ff03976627174c748620523c2e01d7cf73167","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-11 21:39:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/67/726467/2/upstream-check/cinder-cDOT-iSCSI/a10b54c/ : SUCCESS in 52m 21s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/67/726467/2/upstream-check/cinder-cDOT-iSCSI-IPv6/2ffdaec/ : SUCCESS in 1h 47m 58s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/67/726467/2/upstream-check/cinder-cDOT-NFS/7cecfc5/ : SUCCESS in 51m 00s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/67/726467/2/upstream-check/cinder-cDOT-NFS-IPv6/1b664b9/ : FAILURE in 2h 24m 15s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/67/726467/2/upstream-check/cinder-cDOT-FCP/b1c3a11/ : SUCCESS in 1h 08m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c9a7b4c15c721378ac723a67bf554bb6da1fbec0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-05-12 00:09:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/67/726467/2/check/DellEMC_VxFlexOS_os-brick/7b29a47/DellEMC_VxFlexOS_os-brick/134 : FAILURE in 1h 40m 22s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"6abcb099175a8f08e1d0958c466b1ec74e2a3308","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-12 00:54:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/67/726467/2/check/DellEMC_Unity_FC_os-brick/a4dbd1f/DellEMC_Unity_FC_os-brick/107 : FAILURE in 1h 15m 54s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/67/726467/2/check/DellEMC_Unity_iSCSI_os-brick/c1ac938/DellEMC_Unity_iSCSI_os-brick/102 : SUCCESS in 1h 43m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7aecb3b8e48e542d794c2f4c38829512788592f6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-12 01:46:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/67/726467/2/check/DellEMC_VNX_FC_os-brick/f430401/DellEMC_VNX_FC_os-brick/83 : SUCCESS in 58m 57s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/67/726467/2/check/DellEMC_VNX_iSCSI_os-brick/e72a751/DellEMC_VNX_iSCSI_os-brick/82 : FAILURE in 2h 51m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"e27cddbe17b4beaa75f6b26c9934ada7f6d5af60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-12 02:15:37.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/67/726467/2/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/67/726467/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b1db17536ebfddc0961c0cb61eb72aed5011d038","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-05-12 02:28:44.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/67/726467/2/check/DellEMC_SC_FC-os-brick/ef17fa9/DellEMC_SC_FC-os-brick/95 : SUCCESS in 1h 09m 24s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/67/726467/2/check/DellEMC_SC_iSCSI-os-brick/6375a58/DellEMC_SC_iSCSI-os-brick/92 : FAILURE in 2h 50m 02s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"5c3e255839d8f64f8175f1a7b3c7d040d838fa62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-12 13:00:07.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-lioadm http://13.74.249.42/refs/changes/67/726467/2/Cinder-lioadm : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/67/726467/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8a07d62a2ac2adfb8c858b8a335fd3b31a6fbdd1","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-05-14 19:24:53.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\nNeed to created a new exception class.","accounts_in_message":[],"_revision_number":2},{"id":"a83b315824c71d3a426f95677edd3d034fec64eb","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-05-15 19:05:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0fc40d172da02ac81e0ee1d94126cea30d70da75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 21:19:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f050c359e6064a15a48510d9d657db26 : TIMED_OUT in 2h 08m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20513014242f409a9efb51c48021602c : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff232c0f90e4649814408a274e645f6 : FAILURE in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee52554890a545e098e384cd8f452442 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aaad394acf54fa08f3cab6ad57c10e1 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dba4f3155a340e0bc2033bf1637f481 : SUCCESS in 8m 25s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/3c4386609a734207a9187af8a92c6d10 : SUCCESS in 2h 07m 16s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5f019d88f9a4e28845a2ba3c120ce0b : SUCCESS in 1h 58m 41s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/eb341207df81418eaa7a56c4ccb164a7 : SUCCESS in 1h 31m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/initiator/connectors/iscsi.py\n  Comments left for invalid file os_brick/remotefs/remotefs.py\n  Comments left for invalid file os_brick/initiator/connectors/hgst.py","accounts_in_message":[],"_revision_number":3},{"id":"1a155b3092122b02a9bf64388611ec73cc6a9171","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-05-15 21:22:02.000000000","message":"Patch Set 3:\n\nrecheck openstack-tox-pep8","accounts_in_message":[],"_revision_number":3},{"id":"86221d8438990784baa0eab7d69975f025eb3c6d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-15 23:10:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/67/726467/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/26425cc : SUCCESS in 57m 08s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/67/726467/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ba8b977 : SUCCESS in 1h 10m 51s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"918a0df9f2c248e514f57688fde389b26c9e0fee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:18:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/751c94eaa66f444fa3f98a2082576015 : SUCCESS in 1h 53m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b0de29655b0486c8a83b6cdf82f9e5f : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51f88c896db4c42b067ed587f3baa7f : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a717daa61644bf2913cf23321236f07 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5277738b9d784b6fa07a2948d3b131f9 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac94727579b444fa73109538c9d9e01 : SUCCESS in 7m 34s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/a70c2cce30ed45259b7157cb5a6dd404 : SUCCESS in 1h 07m 17s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/54d6d2120f674e35b3cf1d14a8119ae5 : SUCCESS in 1h 45m 58s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/06096a6fd74947c897798ddf4420b036 : SUCCESS in 1h 12m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/initiator/connectors/hgst.py\n  Comments left for invalid file os_brick/remotefs/remotefs.py\n  Comments left for invalid file os_brick/initiator/connectors/iscsi.py","accounts_in_message":[],"_revision_number":3},{"id":"8241f5ef05302d60fd0211a19e244ed008d23d3e","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-16 03:37:33.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/726467/3/iscsi/ : FAILURE in 14m 32s\n- os-brick-smb http://cloudbase-ci.com/os-brick/726467/3/smb/ : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":3},{"id":"77a7c235ede8e9457c2c00a0615fe83e9615b30b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-16 06:31:05.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/67/726467/3/check/DellEMC_Unity_FC_os-brick/f96354f/DellEMC_Unity_FC_os-brick/113 : SUCCESS in 1h 16m 20s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/67/726467/3/check/DellEMC_Unity_iSCSI_os-brick/3d88ed2/DellEMC_Unity_iSCSI_os-brick/108 : SUCCESS in 1h 49m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3b0259c4874075376fd9235634985de72ffb03dd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-16 06:33:29.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/3/check/tempest-dsvm-full-bionic/a6d168c/ : SUCCESS in 1h 39m 18s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/3/check/devstack-plugin-ceph-dsvm-bionic/60d7ea4/ : FAILURE in 1h 32m 21s","accounts_in_message":[],"_revision_number":3},{"id":"46de6f1e396d41824eb4d4c7b9532412adb5d4a9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-05-16 06:42:51.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/67/726467/3/check/DellEMC_XtremIO_FC_os-brick/c8ce442/DellEMC_XtremIO_FC_os-brick/136 : FAILURE in 10m 40s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/67/726467/3/check/DellEMC_XtremIO_iSCSI_os-brick/fba87e2/DellEMC_XtremIO_iSCSI_os-brick/18 : FAILURE in 12m 37s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"25fb917599eaa0d7ee73a4f293ee49c201d61d2d","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-05-27 13:39:03.000000000","message":"Patch Set 3:\n\nrecheck openstack-tox-pep8","accounts_in_message":[],"_revision_number":3},{"id":"d462a21484762c08934064071296e5d09e0ae72a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-27 14:00:50.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/3/check/tempest-dsvm-full-bionic/7b8f84f/ : FAILURE in 15m 03s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/3/check/devstack-plugin-ceph-dsvm-bionic/a825b02/ : FAILURE in 16m 45s","accounts_in_message":[],"_revision_number":3},{"id":"840cfe61d788e197d41730412bc02cfde741bef8","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-05-27 14:34:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/726467/3/iscsi/ : SUCCESS in 55m 21s\n- os-brick-smb http://cloudbase-ci.com/os-brick/726467/3/smb/ : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":3},{"id":"5974f47e4f6b4668d02201c1a620da66ae9c3d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:00:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87ecf3d5c02142fcb11ffda485ecb674 : SUCCESS in 1h 15m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38309583d22447a385b3b8f74920b838 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdc4cd085f441acac00cb01d2703042 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e280f0ed13174889b81cebaac9721ba4 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f935c6b0aec3478ebedcb875da6171ec : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f4481ac6e94a97954d8a2fef40f9b4 : SUCCESS in 8m 42s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/cb06ba569f0946f7a59a9fb122b2e312 : SUCCESS in 1h 03m 27s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59ed7ce188a747d0a7517941e5fb3989 : FAILURE in 1h 15m 46s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/23ba2e4282ec44099b6e5bec1ee3c3ba : SUCCESS in 1h 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e748ce0f2d63110be4f1cb9b1a62bc863b61d95","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-08 13:43:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7d316ec6046f8864ffaedeb1c0d7e812a8005bdd","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-06-30 18:04:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9365b242c45e88d2a30a26f963ee8d0d7d9e1f92","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-06-30 18:26:07.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"8e02b97efdb8ca3a2880e81cbba1d47ccf33679d","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-30 19:15:00.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/67/726467/5/check/DellEMC_XtremIO_FC_os-brick/6c6b9bf/DellEMC_XtremIO_FC_os-brick/38 : SUCCESS in 48m 39s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/67/726467/5/check/DellEMC_XtremIO_iSCSI_os-brick/e8cbce2/DellEMC_XtremIO_iSCSI_os-brick/36 : FAILURE in 8m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"3da22076c39e18f4f74f8b70873ae48255fca13e","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-30 19:21:23.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/726467/5/iscsi/ : SUCCESS in 54m 55s\n- os-brick-smb http://cloudbase-ci.com/os-brick/726467/5/smb/ : FAILURE in 51m 54s","accounts_in_message":[],"_revision_number":5},{"id":"62928ac4f7c08d21ea2525b480cbc4dcec0ace99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 19:57:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ada3b95d7f5e4bfd829bb1de01087944 : SUCCESS in 1h 29m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ce7ef5dbe8d468aa419a4c158a1e6cf : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2977030e28be4fe6ba0eb66b7172ca4f : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b8a8822afe450c9fd7b17e2f1ff86c : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/995f0f830e1e4df989fdc5b50ee33320 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be95cc1a1e8848d8a0e2cec996822616 : SUCCESS in 9m 31s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/8741134778064b5caf83165978832f0f : FAILURE in 26m 54s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/541b983c69994b80884b5a0ec879fbcf : SUCCESS in 1h 20m 23s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c96b54dbe04a4f37b6f99a401c9a1cac : SUCCESS in 1h 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"299e5ab6ced93d10a194d2616ff6c30896913039","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-30 20:24:54.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/67/726467/5/check/DellEMC_XtremIO_FC_os-brick/7445a6e/DellEMC_XtremIO_FC_os-brick/None : EXCEPTION\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/67/726467/5/check/DellEMC_XtremIO_iSCSI_os-brick/85de8d3/DellEMC_XtremIO_iSCSI_os-brick/37 : FAILURE in 8m 13s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"cbfd2b12bda804b1b37f5e8e675e9f1425f3d78e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-30 21:06:27.000000000","message":"Patch Set 5:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-67-726467-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"2f0ad1a496a3ec093165db58168b46d72723e13d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-30 21:30:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/67/726467/5/upstream-solidfire-check/cinder-solidfire-iSCSI/81a4944/ : SUCCESS in 1h 32m 53s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"25fabb40b1ac12e0662346d99cd621581028ccd1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-30 21:35:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/67/726467/5/upstream-check/cinder-cDOT-iSCSI/c012227/ : SUCCESS in 35m 20s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/67/726467/5/upstream-check/cinder-cDOT-iSCSI-IPv6/a586e3e/ : SUCCESS in 1h 33m 16s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/67/726467/5/upstream-check/cinder-cDOT-NFS/9636cd9/ : SUCCESS in 34m 32s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/67/726467/5/upstream-check/cinder-cDOT-NFS-IPv6/b0a199c/ : SUCCESS in 1h 37m 32s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/67/726467/5/upstream-check/cinder-cDOT-FCP/2bfa533/ : SUCCESS in 58m 02s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"1b515c86bb502f09f1ee2966a0f03365316ddacb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-06-30 22:03:16.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/67/726467/5/check/DellEMC_SC_FC-os-brick/dff4ee0/DellEMC_SC_FC-os-brick/24 : SUCCESS in 1h 08m 13s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/67/726467/5/check/DellEMC_SC_iSCSI-os-brick/a501ff8/DellEMC_SC_iSCSI-os-brick/21 : SUCCESS in 2h 05m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"0a1e48ca443ba6965105b81ff96ff0e03cd9a72e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-01 01:16:04.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/5/check/tempest-dsvm-full-bionic/7d5a3e6/ : SUCCESS in 1h 30m 17s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/67/726467/5/check/devstack-plugin-ceph-dsvm-bionic/228f832/ : FAILURE in 59m 57s","accounts_in_message":[],"_revision_number":5},{"id":"21e3b33ce4e0995b162263023ac87882832a1230","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 19:59:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"66a5d46f9b45f114f762d689c5c4658e9744c97b","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2020-07-30 05:25:02.000000000","message":"Patch Set 5: Code-Review-1 Workflow-1\n\nI agree. After debugging a bit more I\u0027m almost sure that this problem could be solved only with these two patches[1][2], because rbd works fine if a proper vol size is passed from the manager to the driver.  \n[1] https://review.opendev.org/#/c/743869/ \n[2] https://review.opendev.org/#/c/742522/","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aa97f3c082fc7f3b653e623334eef755d5c21e4f","revisions":{"279f3c05130f24abb4e788f4ec058643ad4ffe41":{"kind":"REWORK","_number":1,"created":"2020-05-08 20:11:21.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/67/726467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/67/726467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/67/726467/1"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 19:54:43.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 20:11:12.000000000","tz":0},"subject":"Retype from non encrypted to encrypted volume in ceph backend fails","message":"Retype from non encrypted to encrypted volume in ceph backend fails\n\nCurrently RBD has a device_path set and that to an object rather than a string.\n\n\u0027device_path\u0027: \u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object at 0x7fd3e710da90\u003e\n\nRBDVolumeIOWrapper object being passed to os.path.basename is causing the next two errors:\n- TypeError: expected str, bytes or os.PathLike object, not RBDVolumeIOWrapper\n- AttributeError: \u0027RBDVolumeIOWrapper\u0027 object has no attribute \u0027rfind\u0027\n\nThis patch fixes the type confusion problem that makes this break in a messy way now.\n\nChange-Id: Ie85b11bc9c9474ba4dbd92e5afa6d996aa19580a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/279f3c05130f24abb4e788f4ec058643ad4ffe41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/279f3c05130f24abb4e788f4ec058643ad4ffe41"}]},"branch":"refs/heads/master"},"16988c8183e4a033ae6af91a7e6b192f42d1814a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-11 17:30:20.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/67/726467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/67/726467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/67/726467/2"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 19:54:43.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-11 17:30:16.000000000","tz":0},"subject":"Retype from non encrypted to encrypted volume in ceph backend fails","message":"Retype from non encrypted to encrypted volume in ceph backend fails\n\nCurrently RBD has a device_path set and that to an object rather than a\nstring.\n\n\u0027device_path\u0027: \u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object\u003e\n\nRBDVolumeIOWrapper object being passed to os.path.basename is causing\nthe next two errors:\n\n- TypeError: expected str, bytes or os.PathLike object, not\nRBDVolumeIOWrapper.\n\n- AttributeError: \u0027RBDVolumeIOWrapper\u0027 object has no attribute \u0027rfind\u0027.\n\nThis patch fixes the type confusion problem that makes this break in a\nmessy way now.\n\nChange-Id: Ie85b11bc9c9474ba4dbd92e5afa6d996aa19580a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/16988c8183e4a033ae6af91a7e6b192f42d1814a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/16988c8183e4a033ae6af91a7e6b192f42d1814a"}]},"branch":"refs/heads/master"},"dbcdb8424432dcbaab1d6229a0fbab52ccf21f95":{"kind":"REWORK","_number":3,"created":"2020-05-15 19:05:19.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/67/726467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/67/726467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/67/726467/3"}}},"commit":{"parents":[{"commit":"73434e0152ba949f21396d6de0d1ff10544de7ae","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73434e0152ba949f21396d6de0d1ff10544de7ae"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 19:54:43.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-15 19:05:07.000000000","tz":0},"subject":"Retype from non encrypted to encrypted volume in ceph backend fails","message":"Retype from non encrypted to encrypted volume in ceph backend fails\n\nCurrently RBD has a device_path set and that to an object rather than a\nstring.\n\n\u0027device_path\u0027: \u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object\u003e\n\nRBDVolumeIOWrapper object being passed to os.path.basename is causing\nthe next two errors:\n\n- TypeError: expected str, bytes or os.PathLike object, not\nRBDVolumeIOWrapper.\n\n- AttributeError: \u0027RBDVolumeIOWrapper\u0027 object has no attribute \u0027rfind\u0027.\n\nThis patch fixes the type confusion problem that makes this break in a\nmessy way now.\n\nChange-Id: Ie85b11bc9c9474ba4dbd92e5afa6d996aa19580a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/dbcdb8424432dcbaab1d6229a0fbab52ccf21f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/dbcdb8424432dcbaab1d6229a0fbab52ccf21f95"}]},"branch":"refs/heads/master"},"7193600227638481d192c8bd04e951f44dcc7fb4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-30 18:04:29.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/67/726467/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/67/726467/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/67/726467/4"}}},"commit":{"parents":[{"commit":"39b201160bbff39117fca46b2709ff3f74c53d80","subject":"Merge \"rbd: Support \u0027rbd showmapped\u0027 output from ceph 13.2.0+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/39b201160bbff39117fca46b2709ff3f74c53d80"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 19:54:43.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-06-30 18:04:29.000000000","tz":0},"subject":"Retype from non encrypted to encrypted volume in ceph backend fails","message":"Retype from non encrypted to encrypted volume in ceph backend fails\n\nCurrently RBD has a device_path set and that to an object rather than a\nstring.\n\n\u0027device_path\u0027: \u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object\u003e\n\nRBDVolumeIOWrapper object being passed to os.path.basename is causing\nthe next two errors:\n\n- TypeError: expected str, bytes or os.PathLike object, not\nRBDVolumeIOWrapper.\n\n- AttributeError: \u0027RBDVolumeIOWrapper\u0027 object has no attribute \u0027rfind\u0027.\n\nThis patch fixes the type confusion problem that makes this break in a\nmessy way now.\n\nChange-Id: Ie85b11bc9c9474ba4dbd92e5afa6d996aa19580a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7193600227638481d192c8bd04e951f44dcc7fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7193600227638481d192c8bd04e951f44dcc7fb4"}]},"branch":"refs/heads/master"},"aa97f3c082fc7f3b653e623334eef755d5c21e4f":{"kind":"REWORK","_number":5,"created":"2020-06-30 18:26:07.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/67/726467/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/67/726467/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/67/726467/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/67/726467/5"}}},"commit":{"parents":[{"commit":"39b201160bbff39117fca46b2709ff3f74c53d80","subject":"Merge \"rbd: Support \u0027rbd showmapped\u0027 output from ceph 13.2.0+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/39b201160bbff39117fca46b2709ff3f74c53d80"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-05-08 19:54:43.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2020-06-30 18:25:55.000000000","tz":0},"subject":"Retype from non encrypted to encrypted volume is not supported","message":"Retype from non encrypted to encrypted volume is not supported\n\nCurrently RBD\u0027s device_path is an RBDVolumeIOWrapper object rather than\na string. Because of that, the os.path.basename fails and shows the next\ntwo errors:\n\n- TypeError: expected str, bytes or os.PathLike object, not\nRBDVolumeIOWrapper.\n- AttributeError: \u0027RBDVolumeIOWrapper\u0027 object has no attribute \u0027rfind\u0027.\n\nThis patch adds a clearer  error for the type confusion problem that\nmakes this break in a messy way now.\n\nChange-Id: Ie85b11bc9c9474ba4dbd92e5afa6d996aa19580a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/aa97f3c082fc7f3b653e623334eef755d5c21e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/aa97f3c082fc7f3b653e623334eef755d5c21e4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
