)]}'
{"id":"openstack%2Fos-brick~734616","triplet_id":"openstack%2Fos-brick~master~I04ad07cd4ea1055c0244e7444d25da6a920c530e","project":"openstack/os-brick","branch":"master","topic":"pycodestyle","hashtags":[],"change_id":"I04ad07cd4ea1055c0244e7444d25da6a920c530e","subject":"Bump pycodestyle to 2.6.0","status":"MERGED","created":"2020-06-09 14:28:04.000000000","updated":"2020-06-19 17:15:14.000000000","submitted":"2020-06-19 17:12:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734616-1592586771407-3a0d45b0","meta_rev_id":"8842c87a6bdec4fa336791f55fece74bfb5530b0","_number":734616,"virtual_id_number":734616,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-10 02:53:27.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 08:30:49.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 22:20:31.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 17:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 12:08:51.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-06-09 22:16:21.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 09:18:23.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 23:59:34.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 00:32:38.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2020-06-10 00:27:33.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-06-09 19:37:05.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-09 18:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 15:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 15:44:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 16:20:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-09 19:37:05.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-06-09 22:16:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 22:20:31.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 23:59:34.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 00:27:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-06-10 00:32:38.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-06-10 02:53:27.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 08:30:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 09:18:23.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 12:08:51.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-06-19 15:44:22.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-19 17:12:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8752f7184b5c2baf763c19fcc63400439f1daa5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-09 14:28:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29b9cd531d167ad47ca67f1e9b346414eac990f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 14:35:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da80d644a2ae178695952aeb7576d1e3eda50242","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-09 16:12:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/734616/2/iscsi/ : SUCCESS in 1h 00m 51s\n- os-brick-smb http://cloudbase-ci.com/os-brick/734616/2/smb/ : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":2},{"id":"13c91d4e7cce1f107c7b8679d135fcc28aeaf6ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:15:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/63968514e1734b8396d398fbd5bc4fa0 : SUCCESS in 2m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/262ed69da20e4393bfe75e7066638b0c : SUCCESS in 1h 16m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ac1e5da11d744c2a341f16e49abb81e : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3539378232a0418095309cb6da2e905b : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1c961da6648405c98e4d0e415dc34c7 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/848b68a323f146c1ab775937d54908b8 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998432d7206d4a0893b7f7c1ec194d7e : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b8be41a554649af9cfd5ef63fbee61a : SUCCESS in 5m 00s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/17e5e41d0ce84657a025e0208ce3760a : SUCCESS in 1h 11m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cac4baf47c4440659cf46c351228dc13 : SUCCESS in 1h 26m 01s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0c00f5aa962b430292d65cc05a0083a7 : FAILURE in 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"970c6c09e5165034d3ba9c735f30b3c60da0c9b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 16:20:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e24d37078256a72558857558a177ffe55577f1fa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-09 18:34:40.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-734616-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"a31de344d91f47f9c627b6a1adb94dd25ec18a83","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 18:41:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"181cb55d05bc045cd43e84384afdf8d35464a05f","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-09 19:37:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/734616/3/iscsi/ : SUCCESS in 54m 22s\n- os-brick-smb http://cloudbase-ci.com/os-brick/734616/3/smb/ : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0660122e7c9fbad05ac6347b0e95cb292ba6efd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:33:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/138593a712b94c34bbb0657e87184e88 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0e38a06c6234935ae5ccdaa637f67d9 : SUCCESS in 1h 33m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7ec31223ae2412789b0ec72a0f092d3 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c80ec0a13746c799c21ac023fd203c : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b21a1dff07418b8a3bea2b900c3ec3 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32a9a570597d470693d674b7c3c26538 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e113524025b4c649670a557a2f437ba : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeddf0b0156a4f038592d96bc6e4f4e1 : SUCCESS in 4m 43s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/b36f4799943f4a9bb4887f603292f3d2 : SUCCESS in 1h 23m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b9b5e8b1cf5d4b91a2b69c46143957a9 : SUCCESS in 1h 48m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/398dcb32b66f49b2bc86e1489d4f10ce : SUCCESS in 1h 41m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7b33fea01d63df13fca3dff1cb045fe8470b86bd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-09 22:16:21.000000000","message":"Patch Set 3:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-734616-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"825907072d8636f9be1b657f49b6c4517dade6f6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-09 22:20:31.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/734616/3/check/tempest-dsvm-full-bionic/acd936e/ : SUCCESS in 1h 23m 35s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/16/734616/3/check/devstack-plugin-ceph-dsvm-bionic/e68f292/ : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":3},{"id":"fd8b21ce62afa8ccb1b242e5e8f4bcf193ebb597","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-09 23:59:34.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/16/734616/3/upstream-check/cinder-cDOT-iSCSI/4d10233/ : SUCCESS in 3h 25m 28s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/16/734616/3/upstream-check/cinder-cDOT-iSCSI-IPv6/340dfbf/ : SUCCESS in 1h 42m 36s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/16/734616/3/upstream-check/cinder-cDOT-NFS/9fb6e0a/ : SUCCESS in 43m 58s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/16/734616/3/upstream-check/cinder-cDOT-NFS-IPv6/1cb61d4/ : SUCCESS in 1h 43m 30s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/16/734616/3/upstream-check/cinder-cDOT-FCP/514f56d/ : FAILURE in 3h 20m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b060af91bc9e11532ba76ab26fdad663d3cb5438","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-10 00:27:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/16/734616/3/upstream-solidfire-check/cinder-solidfire-iSCSI/fe16604/ : FAILURE in 3h 53m 21s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"f9f3ef095871cab52a1724e8537f3bc0c923371d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-10 00:32:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/16/734616/3/check/DellEMC_Unity_FC_os-brick/5b59528/DellEMC_Unity_FC_os-brick/17 : SUCCESS in 1h 31m 53s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/16/734616/3/check/DellEMC_Unity_iSCSI_os-brick/8b7fdf3/DellEMC_Unity_iSCSI_os-brick/15 : SUCCESS in 1h 34m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"1d003b9932bc6a0668569169c14747a583d48168","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-06-10 02:53:27.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/16/734616/3/check/DellEMC_XtremIO_FC_os-brick/1dbc11c/DellEMC_XtremIO_FC_os-brick/18 : FAILURE in 7m 38s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/16/734616/3/check/DellEMC_XtremIO_iSCSI_os-brick/d36e7ca/DellEMC_XtremIO_iSCSI_os-brick/17 : FAILURE in 7m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"f5bc65e620980ca7e68b28aca9f069e907a4ef3e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-10 08:30:49.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VxFlexOS_os-brick http://publiclogs.emc.com/16/734616/3/check/DellEMC_VxFlexOS_os-brick/334e555/DellEMC_VxFlexOS_os-brick/15 : SUCCESS in 1h 35m 42s\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"5758fef3421407246072f20b8d5d08864b2fbc6f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-10 09:18:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/16/734616/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/ae1bad9 : SUCCESS in 56m 30s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/16/734616/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/0a413c9 : SUCCESS in 1h 08m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"61884c821853c4fd9c897dc4d31e164eee24c0ae","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-06-10 12:08:51.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/16/734616/3/check/DellEMC_SC_FC-os-brick/6b38f53/DellEMC_SC_FC-os-brick/13 : SUCCESS in 1h 08m 42s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/16/734616/3/check/DellEMC_SC_iSCSI-os-brick/9bbd756/DellEMC_SC_iSCSI-os-brick/12 : FAILURE in 13m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"cb0e44731bca77694d3f1c0bce12467819d1eab1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-15 18:26:13.000000000","message":"Topic changed from lower-constraints to pycodestyle","accounts_in_message":[],"_revision_number":3},{"id":"128fc718368470696350d0832e60382d76b290e3","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-19 15:44:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9ddf97e5c52a730f90363207b0bf1c58ba5abc55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 15:44:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"124670642fcbd0c9f71e27e220abcd44bacb8b74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:12:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d11ac7a72fa60c0e61f404acc41ef8881c9e4d5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:12:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/55b3e7e0879e4c3a9ab49ef876fd18d2 : SUCCESS in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7de841fb255f433092e7c9fa1dfb2945 : SUCCESS in 59m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dbe40f5b3d74fa5960ba7b2e08407fb : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fecdbd26d18747b2b795580a6d8d77d8 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/481d0bb611f34f6297d77e9dbe131931 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62de5ef8f2704b2d996e9acedeb52272 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e431a0a8d6f34433bc40053048de8205 : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/798a62ad44874345bf9106edb5e6cf19 : SUCCESS in 8m 50s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/44fdff56895a4652b980a0f0ce33c30c : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":3},{"id":"8842c87a6bdec4fa336791f55fece74bfb5530b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:15:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99dc19207834fa5834b3cffc404f201 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf5444b503af4e6c9dc55ebc48bcfc7c : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0804e817de7b3f0cb6a65178793b815086686690","revisions":{"72cc170c6e79c26ea6fdc5a35716402fae6afdbc":{"kind":"REWORK","_number":1,"created":"2020-06-09 14:28:04.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/16/734616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/16/734616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/16/734616/1"}}},"commit":{"parents":[{"commit":"4ea7fcdfbb59ad3a01d2ac24161c312b3fc75c19","subject":"Raise lower constraints to meet our minimums","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4ea7fcdfbb59ad3a01d2ac24161c312b3fc75c19"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-06-09 14:10:00.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-06-09 14:27:29.000000000","tz":-240},"subject":"Bump pycodestyle to 2.6.0","message":"Bump pycodestyle to 2.6.0\n\nThis brings in a couple of new checks, and\nfixes E741 errors.\n\nChange-Id: I04ad07cd4ea1055c0244e7444d25da6a920c530e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/72cc170c6e79c26ea6fdc5a35716402fae6afdbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/72cc170c6e79c26ea6fdc5a35716402fae6afdbc"}]},"branch":"refs/heads/master"},"60377a29de7c27e1a741c45a57f9efab114faf14":{"kind":"REWORK","_number":2,"created":"2020-06-09 14:35:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/16/734616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/16/734616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/16/734616/2"}}},"commit":{"parents":[{"commit":"f03bd25e2db1df2724d52c2e631fee2f9965df2e","subject":"Update lower-constraints versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f03bd25e2db1df2724d52c2e631fee2f9965df2e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-06-09 14:10:00.000000000","tz":-240},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-09 14:35:19.000000000","tz":-300},"subject":"Bump pycodestyle to 2.6.0","message":"Bump pycodestyle to 2.6.0\n\nThis brings in a couple of new checks, and\nfixes E741 errors.\n\nChange-Id: I04ad07cd4ea1055c0244e7444d25da6a920c530e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/60377a29de7c27e1a741c45a57f9efab114faf14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/60377a29de7c27e1a741c45a57f9efab114faf14"}]},"branch":"refs/heads/master"},"0804e817de7b3f0cb6a65178793b815086686690":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-09 18:41:27.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/16/734616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/16/734616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/16/734616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/16/734616/3"}}},"commit":{"parents":[{"commit":"d95ebc4ec214bd1db368100b4b53cccefc4ab90e","subject":"Update lower-constraints versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d95ebc4ec214bd1db368100b4b53cccefc4ab90e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-06-09 14:10:00.000000000","tz":-240},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-09 18:40:48.000000000","tz":-300},"subject":"Bump pycodestyle to 2.6.0","message":"Bump pycodestyle to 2.6.0\n\nThis brings in a couple of new checks, and\nfixes E741 errors.\n\nChange-Id: I04ad07cd4ea1055c0244e7444d25da6a920c530e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0804e817de7b3f0cb6a65178793b815086686690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0804e817de7b3f0cb6a65178793b815086686690"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
