)]}'
{"id":"openstack%2Fos-brick~735650","triplet_id":"openstack%2Fos-brick~stable%2Fstein~I7e761828b3799cef720e15ca7896e8e8d6f98182","project":"openstack/os-brick","branch":"stable/stein","hashtags":[],"change_id":"I7e761828b3799cef720e15ca7896e8e8d6f98182","subject":"rbd: Correct local_attach disconnect test and showmapped arguments","status":"MERGED","created":"2020-06-15 18:29:07.000000000","updated":"2020-07-29 16:39:16.000000000","submitted":"2020-07-29 16:37:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735650-1596040637831-4e4c0980","meta_rev_id":"55768a7fe058471a6a5c3de766a496a315fee8bf","_number":735650,"virtual_id_number":735650,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-28 17:56:19.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-07-28 19:16:39.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-28 19:12:34.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-28 15:41:46.000000000","_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"date":"2020-07-28 18:02:30.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 16:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-28 19:11:49.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-07-28 15:54:52.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-29 14:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-28 22:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-29 14:56:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 15:54:52.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-07-28 17:56:19.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 18:02:30.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 19:11:49.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-28 19:12:34.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 19:16:39.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 22:03:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-29 14:56:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-29 16:37:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f45c25e6811bdda2a9efab518e0db7f0e7d8bd81","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-15 18:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbb04adeb1629f8226f974199374cff9fc305d08","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-15 18:34:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/50/735650/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"dad5e552a0c45b0cf9b871f2e3fde6657cceb230","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-15 18:39:26.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735650/1/iscsi/ : FAILURE in 9m 49s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735650/1/smb/ : FAILURE in 8m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7591c1c20785be5af8922e2785d92640756fa886","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-15 18:44:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/50/735650/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1589e3aa0cb67bba3a314ca2fcc1126641c30eda","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-15 18:51:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/50/735650/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3b093ba : FAILURE in 16m 58s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/50/735650/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/376ec1a : FAILURE in 22m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"cbe42422b13da59186e1a46df072c387c56a4a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:11:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/dbd09a66a135448483651541821dbe3f : FAILURE in 18m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1e40496131f4727b41a6113d1f29d18 : FAILURE in 20m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f8cf249ed2d46f992159901320a93b9 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61fef4bbfd2d4b15bd3f073a8fa4592f : SUCCESS in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36369e9fb44a4fbc82470ed935ba31cf : SUCCESS in 9m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42ffdd02acdc4ccfb2327b3aeab0ebd1 : SUCCESS in 9m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa83d908ea8c415f91c0525484312e7b : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf0366e57244849a0d7d795525ce30d : SUCCESS in 9m 16s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/41f9b42691964947945a0284b4ab3cad : FAILURE in 24m 54s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/6ec25171a90f45138417d6153fcf9d7f : FAILURE in 21m 08s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/596fc0e33c914d31a5576f86fb426686 : FAILURE in 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b53d2d09793b3a2be02830d41324ac908068ba2b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-15 23:19:14.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/tempest-dsvm-full-bionic/f2aac69/ : POST_FAILURE in 32m 07s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/devstack-plugin-ceph-dsvm-bionic/8fd518b/ : POST_FAILURE in 41m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f9dbd392468e4dbceb6c7706e16970b2fa93d2a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-15 23:24:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/50/735650/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"19defdd57be8b93b17a48e11519bc6cd5ddcedb7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-16 16:11:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"563017dac7d4de334e5f16f36f5479cb2d2c3c1a","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 16:12:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735650/1/iscsi/ : FAILURE in 45s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735650/1/smb/ : FAILURE in 46s","accounts_in_message":[],"_revision_number":1},{"id":"c10966451d532850d8fc021be09fd740a06c3bd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 16:16:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/50/735650/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d69c279326f072e15aad64fea17d578ed1e39397","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 18:14:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/50/735650/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/fd6860e : SUCCESS in 59m 32s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/50/735650/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/bf867c8 : FAILURE in 1h 53m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"52bc2c7d67f2703a257d040ee1537f9079feed61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:23:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4ba8dc514b804dc0909c5d3b77a4db9e : RETRY_LIMIT in 4m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32bfe45a694a4cccacbbd8467b3f628d : RETRY_LIMIT in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2948416b3dc24b7baac7cee8f0a44812 : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a70d0e306c346dbaeeebcce31812007 : SUCCESS in 12m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/762649eacbe44e59b9b9b2ba3de12e54 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdefa4f504f34b59b6313b22d718199a : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/932b13387c92499daa77db88dad98951 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2962bef158472aa780877536f5d404 : SUCCESS in 5m 05s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/10c9a1ffb19f4caa97f2bfddb7665743 : SUCCESS in 1h 24m 10s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/85b3093141c4430b9fe7cf884e146c63 : SUCCESS in 1h 38m 35s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/625450b2316f49fe8b7139d5e4db28f5 : SUCCESS in 1h 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f8ccb56ac4e3d30f468051969fbf30d1baa810b4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 20:24:40.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/tempest-dsvm-full-bionic/6ffeea7/ : TIMED_OUT in 2h 37m 37s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/devstack-plugin-ceph-dsvm-bionic/512c0c5/ : POST_FAILURE in 57m 02s","accounts_in_message":[],"_revision_number":1},{"id":"10460555ab5ea91eda6793cd02b85853ef94c9b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 20:28:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/50/735650/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"07f2189013188862076d4305c4fd3b9714fcebde","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-06 13:58:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c482b1d3e1c8db4dcca136794362162d90022f41","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-07-06 14:07:18.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735650/1/iscsi/ : FAILURE in 7m 49s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735650/1/smb/ : FAILURE in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e0392bb2a958128fae938a4a24498857204ef729","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-06 15:14:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/50/735650/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/8ead03a : SUCCESS in 1h 01m 00s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/50/735650/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/35e9a3c : SUCCESS in 1h 14m 46s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"a560b112852fbc8b9d4e9062c04cd238512e1f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:40:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/303d8818e6c74b21bd16715bd5d09567 : SUCCESS in 1h 39m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8573a2a6f99942029a8803aff215fc43 : SUCCESS in 1h 00m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/859ae9781379400cb6cd43ce4bfb5e5d : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb5618bc1d24b0fbba4808eaee5e0d6 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66b81dd35dd24f07876422582879bfc5 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23a6dbbbb76944b3b44d4164c20601ba : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/996c1e0629fe4ba58f89927149dc8f8b : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495ce08ebca847a79002b91ade92813d : SUCCESS in 5m 08s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/9a6f297319234b35ade2929108aacc91 : FAILURE in 1h 24m 06s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/8476ab20e29245c090d0c510aabf7f39 : SUCCESS in 53m 14s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fecb6d32aa7e46cb96ad8eba83f309e8 : FAILURE in 1h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c316fac53f9f5928845afd144e36bdeb05b38322","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-06 18:48:44.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/tempest-dsvm-full-bionic/cfba274/ : TIMED_OUT in 2h 37m 57s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/devstack-plugin-ceph-dsvm-bionic/e9ed749/ : FAILURE in 1h 25m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2913451b06bc557c444faeb9104726be65429e9c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-28 15:41:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15123690df62c6264cac0faf4aad323ef8e680db","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-07-28 15:54:52.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735650/1/iscsi/ : FAILURE in 12m 39s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735650/1/smb/ : FAILURE in 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2e4f9d5a863a97ef6f03a67b4f9e79aeac34ed6a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-28 17:06:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/50/735650/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f07ff96352dd33202c67f0b07223d857b58e0d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 17:19:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1fedb374b9204b9993252bfdf567ed12 : SUCCESS in 1h 24m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70ede9166a1648efbb44430a5728302d : SUCCESS in 1h 30m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efa42349f1c44e77b6bf5631e038cadc : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82093af65df4df2869d443419e8646e : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c9da7f756a640219ba68c9f17768464 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bdaa64b74c747798014b803993e9d38 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/436be6f48b3b4f3db31f0ebd7c8cb070 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f082c405aa14ef790a450f577e4018f : SUCCESS in 6m 49s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/f39b1b4c69ec4330b9d06f742d805333 : FAILURE in 52m 50s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/b95875e74f0a4168aa6eee34a8ca3401 : SUCCESS in 1h 15m 46s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/631367b2e9194450ac4a8ef09cfbc98b : SUCCESS in 1h 19m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4e63a88b2b0fb94635bf0d087712299c5db18013","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-28 17:56:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/50/735650/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/a7c2d9d : FAILURE in 1h 53m 37s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/50/735650/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/8d36e65 : SUCCESS in 1h 21m 02s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"804212c00897e692f782b6e9370f25b14fcf5f51","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-28 18:02:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/50/735650/1/upstream-check/cinder-cDOT-iSCSI/761ffeb/ : FAILURE in 1m 44s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/50/735650/1/upstream-check/cinder-cDOT-iSCSI-IPv6/d1abb38/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/50/735650/1/upstream-check/cinder-cDOT-NFS/4f52cb0/ : SUCCESS in 42m 32s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/50/735650/1/upstream-check/cinder-cDOT-NFS-IPv6/55339b7/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/50/735650/1/upstream-check/cinder-cDOT-FCP/c77532e/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fc7620287d89e709851addb24a7b4343e5e67c83","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-28 19:11:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/50/735650/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f677a12/ : FAILURE in 1h 51m 45s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"e4871002cfdaefe779eef55a0527d16230317069","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-28 19:12:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/tempest-dsvm-full-bionic/e4ded8a/ : TIMED_OUT in 2h 42m 57s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/50/735650/1/check/devstack-plugin-ceph-dsvm-bionic/6076a76/ : FAILURE in 1h 18m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fd1183500832732f6720ed0209b62ef5d6bb6531","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-28 19:16:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/50/735650/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9205e25b84c968d3c76ceca80363832eef064e25","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-28 22:03:07.000000000","message":"Patch Set 1: Code-Review+2\n\nMeets backport criteria.","accounts_in_message":[],"_revision_number":1},{"id":"2dfe82b85bff73aaa5dd954d2b2528451b86b86c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 14:56:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52e80796dde6c83df549384b05205e2471fb51b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:56:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19472521cd1e79c1a640e4348524128b29d7ff34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:37:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"575b0dc77b4269a3469b2d8ceb7ab897ec68fd8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:37:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8b6df908182c4473b11254ac8fe469fe : SUCCESS in 1h 30m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e795949404d248c48331b3048d792eaf : SUCCESS in 1h 22m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0734aaaab5a347df9b7b934e92ca3286 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da1a4d9c1c949e6ab974e2867bc572a : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bee403d79087405cabe066fcecda2549 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/225a5aaa039142a38673eb0466922f21 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6f000981fa54c8ba212c089038387dc : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0a59cd67db4c1686555127096b3528 : SUCCESS in 4m 09s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/b3e94b1b2a304152b95ecd1707decbea : SUCCESS in 1h 22m 47s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1276adc8c162456687be7abcde0c82ec : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"55768a7fe058471a6a5c3de766a496a315fee8bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:39:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87596c9d2d164a86bf64b8004027af8c : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f25b9c92a97410ea83ed89d49018ec3915942ec","revisions":{"4f25b9c92a97410ea83ed89d49018ec3915942ec":{"kind":"REWORK","_number":1,"created":"2020-06-15 18:29:07.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/50/735650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/50/735650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/50/735650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/50/735650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/50/735650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/50/735650/1"}}},"commit":{"parents":[{"commit":"3ce686a9fa41210fb4c08017220ce063cc869c75","subject":"rbd: Use showmapped to find the root RBD device during disconnect_volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3ce686a9fa41210fb4c08017220ce063cc869c75"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-05-07 13:37:48.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-06-15 18:28:46.000000000","tz":-240},"subject":"rbd: Correct local_attach disconnect test and showmapped arguments","message":"rbd: Correct local_attach disconnect test and showmapped arguments\n\nId507109df80391699074773f4787f74507c4b882 introduced the showmapped\ncommand when attempting to disconnect locally attached rbd volumes.\nUnfortunately at the time the test changes incorrectly attempted to\nassert the commands made by using has_calls instead of the valid\nassert_has_calls method.\n\nThis change now corrects this, ensures _get_rbd_args is called to\npopulate all of the required arguments for the showmapped command and\nfinally corrects the name stored within the fake showmapped output in\nthe test.\n\nChange-Id: I7e761828b3799cef720e15ca7896e8e8d6f98182\n(cherry picked from commit 71331b0e06a713a0883f0c1199923e0943adcf09)\n(cherry picked from commit 55cfc975811584bb533211253b8506028dbbb958)\n(cherry picked from commit 0ae0afbb0facbecd8768c6052c9ac1e85d7b6329)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4f25b9c92a97410ea83ed89d49018ec3915942ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4f25b9c92a97410ea83ed89d49018ec3915942ec"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
