)]}'
{"id":"openstack%2Fos-brick~735989","triplet_id":"openstack%2Fos-brick~stable%2Ftrain~Ie2db587c3bc379acd53cfd449788d171ae58dec5","project":"openstack/os-brick","branch":"stable/train","topic":"bug/1883654","hashtags":[],"change_id":"Ie2db587c3bc379acd53cfd449788d171ae58dec5","subject":"Fix \"Remove VxFlex OS credentials\" regression","status":"MERGED","created":"2020-06-16 18:02:15.000000000","updated":"2020-06-18 01:28:41.000000000","submitted":"2020-06-17 22:40:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735989-1592433651769-f763604f","meta_rev_id":"24d8adc72b4756832137d1ae7e658cdb224d89bf","_number":735989,"virtual_id_number":735989,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-17 18:11:34.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-06-17 19:19:56.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 22:40:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-17 19:00:48.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"date":"2020-06-17 19:16:03.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-17 16:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":2,"date":"2020-06-17 21:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":1,"date":"2020-06-17 21:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 23:50:39.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 14:35:42.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-06-17 15:38:01.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-06-17 15:38:10.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2020-06-17 15:38:16.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-06-17 16:05:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-17 18:11:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-17 19:00:48.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 19:16:03.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-06-17 19:19:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 21:01:11.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-06-17 22:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"798d1495ad3c96fb8bcdc8f82fe0eb8329d1e8c6","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 18:02:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6e23c5dba8660e613377e1657099db6d62c1f56","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 18:04:06.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/1/iscsi/ : FAILURE in 1m 31s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/1/smb/ : FAILURE in 1m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ebb35fc79c225e51e3ff4f55ead3c45e6b330331","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 18:04:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"cf3053f3af944464f8842220fdc837184ccc457b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 18:07:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"06dd806ae3332dc6943d7832673e88484b30feb1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-16 18:25:40.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nWould be great to have a unit test that covers this path. We must not have coverage today if the code was able to be merged this way and have the py27 unit tests pass.","accounts_in_message":[],"_revision_number":1},{"id":"705039fff3fc23d3f9ba822dad1d519032a59321","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 18:27:01.000000000","message":"Patch Set 1:\n\nsure i will update that","accounts_in_message":[],"_revision_number":1},{"id":"494bcce6c726ec1581f4a07b6d5f36559116336d","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 18:49:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e0a5636e3b3348c3e1292a0f64c3e77ef16f8907","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-16 18:50:00.000000000","message":"Patch Set 2:\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"ab959a7451e53516a185122217353b39af4dd057","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 18:53:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f78b5b29cdaf3d811743423df1765a92cd1d48eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 19:20:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/82285621711c4b96856c39069688ef6d : RETRY_LIMIT in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd549f30048d46cd88623b9cbef666e0 : RETRY_LIMIT in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a59bca1739ba4d41876c9516d41ec8ec : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a13dd09f49a441bba825c2145fc13e67 : SUCCESS in 10m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de89d68388e34202a5eec4b0211dc4a3 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/668f051dc4814ca79750b492686a2de8 : SUCCESS in 11m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/769df14051a64093856b64fb7e3b37c6 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bead11aba48f493680e3d246d4579ac7 : SUCCESS in 8m 25s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/f0a521f0cad740548961a8b3c6fcf099 : FAILURE in 15m 48s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/098eccaffaad4c02af5a570b206a2c60 : FAILURE in 28m 23s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/afe5199db6524bbe8d39f7e21683e37a : FAILURE in 18m 31s","accounts_in_message":[],"_revision_number":2},{"id":"996cf99b24dc0c3a92e900aae9cf5283d85e01fc","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 19:29:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/2/iscsi/ : FAILURE in 40m 07s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/2/smb/ : FAILURE in 39m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7335b5f873bf087f552dc021be0696c71a809943","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 19:33:37.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"686f02e80363c4eb778fb780fafc8a885760f73c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 19:57:23.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/2/check/tempest-dsvm-full-bionic/57360ba/ : POST_FAILURE in 31m 48s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/2/check/devstack-plugin-ceph-dsvm-bionic/b77fd67/ : POST_FAILURE in 31m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f255a54b512181622f629aa89024df9362276bc6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 20:01:05.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"5704ce396a71d15f07fd7f66a88f90971ee8eb06","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 20:24:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/89/735989/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9640fb3 : FAILURE in 16m 38s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/89/735989/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/64256f1 : FAILURE in 16m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"ef1fe56954bc99bb52bd00afb5fa278fbaf7ef06","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-16 20:59:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"11023fc5b6b8fba24e37853c808f2bd0f45c2ef3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-16 21:00:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d360dec93c8dd7fc1cef2d61aa1f54e819fbd0b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 21:01:09.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"5241c4b43339ab600ea7d43e63818868d1c161dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 21:04:39.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"714265a4d8a4e9dd249dc2980721a96a9866b8d8","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 21:14:57.000000000","message":"Patch Set 4: Code-Review+1\n\ni run the unit test locally it works and tried the change on devstack i can attach volume","accounts_in_message":[],"_revision_number":4},{"id":"cbf86f9d509c23e6320c0fa508ab47a52576dc49","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 21:23:38.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/89/735989/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/93a9811 : FAILURE in 16m 15s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/89/735989/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/96e420c : FAILURE in 16m 05s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"30bb50ef7e95cf60066b7c59a0dccdf6bd22f24e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 21:39:48.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/94ab609fbd73459584fcef127da7ea33 : POST_FAILURE in 32m 52s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/4/check/devstack-plugin-ceph-dsvm-bionic/2c2f868/ : POST_FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":4},{"id":"832cdf52f03fdf32efec24e2b3a04f40621f0b3f","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 21:42:32.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/4/iscsi/ : FAILURE in 40m 15s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/4/smb/ : FAILURE in 41m 14s","accounts_in_message":[],"_revision_number":4},{"id":"14eb548f8285a1a26b653316794670a3481e9cb4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 21:45:06.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"fa7d78332ad63e708d3edb4b1d2b5555e3f837ed","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-16 21:55:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"24b8316624b7d20eb708b6c8c98f261a030a6d25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 22:00:18.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/5/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"da6f99af8fd95ab987c8538f7aaa2bff14c73304","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-16 22:00:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a889f1e59c24924786f0dd62cc6b21730697dffa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 22:04:27.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"003761da2517ddd71f4f91d989be29439e877bf9","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 22:22:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/89/735989/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/dbdfc1c : FAILURE in 16m 16s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/89/735989/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/63831c1 : FAILURE in 19m 45s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"9cb033b506179df836bfd09f3501a6b09e5009b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 22:26:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/37f4ab9d23b5458b996b1d99cfe35edd : FAILURE in 12m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f868461fd83b4d6d9a84e85b2324b261 : FAILURE in 12m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59058472abd14e91a20ed2c792c8cfe4 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b12134c16a024f38b14c16c0fbe1befb : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71ca6e7d225b4173935a4f61b82be9b1 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a1a7a78021f46d890614c2f800be10f : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a409eb7f1484c20bfb394348dd6a8f7 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59af3f14f13541c989f1411c97351523 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2978a7bea15f4638a866c21e5ace6201 : SUCCESS in 8m 15s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/02f36a46cba947d594f66506744c5429 : FAILURE in 15m 01s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/d22e46a3e8d7458f9f009f408d47fe2f : FAILURE in 15m 40s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/330774ab3a664a8a93a7da05813c7aad : FAILURE in 13m 07s","accounts_in_message":[],"_revision_number":6},{"id":"50ab3378224e28c97703326cc762f099cb207e74","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 22:38:19.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/6/check/tempest-dsvm-full-bionic/06bfe8d/ : POST_FAILURE in 28m 59s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/6/check/devstack-plugin-ceph-dsvm-bionic/7f72cc4/ : POST_FAILURE in 30m 51s","accounts_in_message":[],"_revision_number":6},{"id":"077f26812a45583607d8bb9a76905e25bfdf5108","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 22:41:02.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/6/iscsi/ : FAILURE in 39m 42s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/6/smb/ : FAILURE in 39m 58s","accounts_in_message":[],"_revision_number":6},{"id":"b38cf000c7d77822de8dba73d1dae6b56a1c9d44","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 22:42:04.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"3a50f146eb501fba3742696cfbf773a160b75d24","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 23:11:42.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6c2ddfe98281227714ef2cb50817d31ef7955b38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:15:26.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"ab977bf833e9d6db4fabcccaec9ec71afb90e801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:29:37.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4c71a14bce744053a7c5aee9e7fb46e8 : FAILURE in 12m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d87275407ff34e978e9154e4ab1b5189 : FAILURE in 11m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18a961a2e2504c458ac846fbf692626c : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f6d32b03704b49ac54dd6dc05d3d37 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72ce3899a0174c6d9b925885d39727e2 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c7c58d6fff34e779140f823a4f2013b : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b53d67914d1e4899b34a38cc9f24e18a : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c8a05b3e2d4d5a80cc8bcdf6d4dc9e : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee6696b780924bc1a5130451d6c1435b : SUCCESS in 7m 50s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/2e098316fa474bd68736809bdce0acb0 : FAILURE in 14m 39s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/cc4d3feb2e3547d9a988f8d176c35bbb : FAILURE in 15m 44s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c34ace02c455487f80f9f4d0d6939bd1 : FAILURE in 15m 43s","accounts_in_message":[],"_revision_number":6},{"id":"9c3b892be71d77af3a7cdcc0f7054f84b4f01591","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 23:46:40.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/89/735989/6/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/039f5bb : FAILURE in 16m 00s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/89/735989/6/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6d7e4b0 : FAILURE in 16m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":6},{"id":"f303f65e7aa00f23a09d9639c14bafe9345bbeed","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-16 23:50:39.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/6/check/tempest-dsvm-full-bionic/1a413c0/ : POST_FAILURE in 27m 15s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/6/check/devstack-plugin-ceph-dsvm-bionic/0f8af07/ : POST_FAILURE in 33m 14s","accounts_in_message":[],"_revision_number":6},{"id":"52ee4feedc5a40a10b83b35c89367e38523585ab","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 23:53:11.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/6/iscsi/ : FAILURE in 40m 50s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/6/smb/ : FAILURE in 41m 08s","accounts_in_message":[],"_revision_number":6},{"id":"23880f8822d7367897bc2d2dc6499b7c47b6617b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:54:23.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"147803d50fd1e5ca6d9ff90c17ef03e714e07054","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 13:27:50.000000000","message":"Patch Set 6: Code-Review-1\n\nWe need to update the commit message; should mention that the problem is a regression caused by change Ia1d2b2151e5676037d40bfaf388b54023fc37093","accounts_in_message":[],"_revision_number":6},{"id":"8cf73133adfaf7445a18d7df5bb675a0a2982e2e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 13:42:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f5ce769478e30bceb3e8c3bec5c5969dc3945917","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 14:13:36.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b93f10e319e73400dc12f436bb57ecc1247ca9fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 14:26:27.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"53c2add526130faaebd32a1493d4ebac874b638f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-17 14:35:42.000000000","message":"Patch Set 6: Code-Review-1\n\nI agree with Brian\u0027s comment.  Otherwise the patch looks good.","accounts_in_message":[],"_revision_number":6},{"id":"c9a8e9ff1d421a495e107457788a359d2e0ea4e3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 14:36:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"18a61fc469069181eee423905cd8dc06fa6d4bf5","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-17 15:12:08.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/6/iscsi/ : SUCCESS in 58m 11s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/6/smb/ : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":6},{"id":"01a017c67b1ddc9189b01c35d962932e1ccbe55b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 15:36:57.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"5fabb3b26079bc0cf74dc01d33e6350e18676e1c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 15:40:42.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/7/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"afc4b04e4f6ef81c7753462471af5f8b355a1b4b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 15:58:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"59258df7a3302cd6088fada2d581a56283e3c587","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 16:00:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bad9ba59beb6d5a7be14f01184ce23e9d4720a91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 16:04:23.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"c9a74a7b003343d21c5a9d920cdd199b3e815939","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 16:05:23.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":8},{"id":"c0819951b592d1a0f6fce12dc5453fdd17b355bb","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-17 16:58:19.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/8/iscsi/ : FAILURE in 30m 15s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/8/smb/ : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":8},{"id":"82f141da38e184cdbddeec82f46ecfa7a4c88af5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 17:02:29.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"0155cbc4da74a176a88e9d4ffd291fee4dc85c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 17:42:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6505610d02ef4332a0442ea7d339f333 : SUCCESS in 1h 08m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79bed14dd19c4dbdb0f1f4ab98765435 : SUCCESS in 1h 28m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bbfa2f95695449c9363ca7680d01ac6 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ada4da7cc24d5683f526e27dd8ea39 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/29b608a1f69c4e849a7a86be7849645d : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc3a12b9e103495881506cb3fdc15129 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edae4fe61b834839a66a6f8c6c60e669 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83768b191f90474d97c997651c363e9f : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af65ccda1ed8408c96591566b0229ad7 : SUCCESS in 8m 24s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/3a9253724fc149b3819d59e59a29a7a9 : SUCCESS in 1h 27m 47s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/d3e5a6d1e89044f49d6310f44313e418 : SUCCESS in 1h 24m 11s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00c4c64552b7403890f34ac673b2a7b9 : FAILURE in 1h 30m 47s","accounts_in_message":[],"_revision_number":8},{"id":"715d0c2e77f3d9f0b0d37987c52d33004f821d7f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 18:11:23.000000000","message":"Patch Set 8:\n\nos-brick-cinder-tempest-dsvm-lvm-lio-barbican failure was tempest.scenario.test_volume_boot_pattern.TestVolumeBootPattern.test_volume_boot_pattern couldn\u0027t ssh into the server","accounts_in_message":[],"_revision_number":8},{"id":"22b1b096462f92a3367e3ad17d6f6a5f35afbf2c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 18:11:34.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c2932a7bef6d3a19078538817ec3a6e8555fc881","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 18:15:20.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"d5631100f98a45bb6c84a896e370cf0dc844dfdc","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-17 19:00:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/89/735989/8/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/62bb43b : SUCCESS in 1h 05m 45s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/89/735989/8/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/5346cba : SUCCESS in 1h 18m 37s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":8},{"id":"694ee16ce1de7051aeafb20d097a41dec29e1733","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-17 19:16:03.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/735989/8/iscsi/ : FAILURE in 1h 04m 08s\n- os-brick-smb http://cloudbase-ci.com/os-brick/735989/8/smb/ : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":8},{"id":"7e842ff11149e9aeb2e79bff64da5a56c377b34c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 19:19:56.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"ee4e16f445bcd20935b7706fbf7585edde849c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 20:06:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/93ea477c93e24cd68fbd270fb0b06241 : SUCCESS in 1h 25m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08097df5413c4cb1a34e2b232ea21749 : SUCCESS in 1h 35m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7e3871c936a45b6be3a7b8c781fe0d3 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e511a27385047f4b2f849361c37663e : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc346468a1924c558e68089c33375f08 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd41f5ccb3e4ca78f05a8a2fc5ee2c9 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d395747ddbc344ebb4fda3f0424a66de : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d888948a1e034609a542067cc66823db : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28ebf89c1d5c45a081d4b4cab5ade597 : SUCCESS in 8m 14s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/874de65b1c374ce3a016047df9a2ab0d : SUCCESS in 1h 43m 22s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/5809a61349e64089b90e319f31dcbd8d : SUCCESS in 1h 28m 15s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ac4bb46afae4663bd13bb3bfa13f659 : SUCCESS in 1h 43m 38s","accounts_in_message":[],"_revision_number":8},{"id":"27e93f180d23c51d13693f270c3714962ea6c9a5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-06-17 21:01:11.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"712bca6480e18a3fc2cb8349fd4a6a67c8d9e5b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:01:23.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"239b1601764a085b4b70b03c3c23e053d9b561ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:40:51.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/71713e3a3faf4694a83853176a37fd6f : SUCCESS in 1h 33m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/875f7a97ae5642d59213c56d8f263777 : SUCCESS in 1h 29m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcc7e04fa9394d069b69b6e4af412dd8 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db2a5762f3584bae99a08138d999845b : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a30fb6489794ac9ad581d510d89f5b7 : SUCCESS in 10m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc10998af3e046e390465b59a3d4005c : SUCCESS in 8m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48281cda072d47b38330ead0dbf25cf6 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1919fe3a7da84f13976377ac5634e210 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a9140b32e774df2a3b4d97ca6cda26a : SUCCESS in 6m 17s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/d7d14f85506a417aa0c7ed11f0f00e14 : SUCCESS in 1h 21m 53s\n- os-brick-cinder-tempest-dsvm-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89ca4e7962a14fcdba04f9a56b12a2c9 : SUCCESS in 1h 30m 05s","accounts_in_message":[],"_revision_number":8},{"id":"10e8396dcd2bcb75884030b2052ef00dee48e6b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:40:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"e0884a68ef35d147cfe73599cd6bc5c6e0f3a25d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:43:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4ed8fa720d4382a0b9e595577d8c21 : SUCCESS in 2m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b3a84264fb44d79aa173e6971c0b14 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":8},{"id":"ae668b7c5e2a42eda940fc8697922084e90b867a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-18 01:24:58.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/8/check/tempest-dsvm-full-bionic/0be62dd/ : TIMED_OUT in 2h 38m 06s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/735989/8/check/devstack-plugin-ceph-dsvm-bionic/9cdd159/ : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":8},{"id":"58c654787241f7fda2812edea904640a5d391949","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-18 01:28:41.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/89/735989/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"12d252db9cb9deffea3c87b86ea71b3013d93892","revisions":{"1f1038dd9ff56bff453918b9af20ae6c7c141d59":{"kind":"REWORK","_number":1,"created":"2020-06-16 18:02:15.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/89/735989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/1"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 18:02:06.000000000","tz":0},"subject":"Fix the get configparser get password for py27, using configparser get function","message":"Fix the get configparser get password for py27, using configparser get function\n\nfor python 2.7 the error ConfigParser instance has no attribute \u0027__getitem__\u0027 is\nraised when the code try to get the password from config\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1f1038dd9ff56bff453918b9af20ae6c7c141d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1f1038dd9ff56bff453918b9af20ae6c7c141d59"}]},"branch":"refs/heads/stable/train"},"e48de324dcfc7619da08040feed19b29ed8538c9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-16 18:49:25.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/89/735989/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/2"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 18:36:06.000000000","tz":0},"subject":"Fix the get configparser get password for py27","message":"Fix the get configparser get password for py27\n\nin python 2.7 the error below will be raised\nConfigParser instance has no attribute \u0027__getitem__\u0027\ntrying to get the password using configparser with new\nbehaviour introduced in configparser in py3\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e48de324dcfc7619da08040feed19b29ed8538c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e48de324dcfc7619da08040feed19b29ed8538c9"}]},"branch":"refs/heads/stable/train"},"872af40fb28082c0508a44e7e62591fc77ffbacd":{"kind":"REWORK","_number":3,"created":"2020-06-16 20:59:40.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/89/735989/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/3"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-16 20:58:31.000000000","tz":-240},"subject":"Fix the get configparser get password for py27","message":"Fix the get configparser get password for py27\n\nin python 2.7 the error below will be raised\nConfigParser instance has no attribute \u0027__getitem__\u0027\ntrying to get the password using configparser with new\nbehaviour introduced in configparser in py3\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/872af40fb28082c0508a44e7e62591fc77ffbacd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/872af40fb28082c0508a44e7e62591fc77ffbacd"}]},"branch":"refs/heads/stable/train"},"2748dab08358f57e5c46a3d3961701e35321c65d":{"kind":"REWORK","_number":4,"created":"2020-06-16 21:00:57.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/89/735989/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/4"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-16 21:00:50.000000000","tz":-240},"subject":"Fix the get configparser get password for py27","message":"Fix the get configparser get password for py27\n\nin python 2.7 the error below will be raised\nConfigParser instance has no attribute \u0027__getitem__\u0027\ntrying to get the password using configparser with new\nbehaviour introduced in configparser in py3\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2748dab08358f57e5c46a3d3961701e35321c65d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2748dab08358f57e5c46a3d3961701e35321c65d"}]},"branch":"refs/heads/stable/train"},"64dcef5632c0efe19fc5b332491f69c3da88b369":{"kind":"REWORK","_number":5,"created":"2020-06-16 21:55:20.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/89/735989/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/5"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-16 21:51:47.000000000","tz":-240},"subject":"Fix the get configparser get password for py27","message":"Fix the get configparser get password for py27\n\nin python 2.7 the error below will be raised\nConfigParser instance has no attribute \u0027__getitem__\u0027\ntrying to get the password using configparser with new\nbehaviour introduced in configparser in py3\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/64dcef5632c0efe19fc5b332491f69c3da88b369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/64dcef5632c0efe19fc5b332491f69c3da88b369"}]},"branch":"refs/heads/stable/train"},"8f7b8c82297ab20f46fd148f551f2360fa8abfb6":{"kind":"REWORK","_number":6,"created":"2020-06-16 22:00:42.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/89/735989/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/6"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-16 21:59:20.000000000","tz":-240},"subject":"Fix the get configparser get password for py27","message":"Fix the get configparser get password for py27\n\nin python 2.7 the error below will be raised\nConfigParser instance has no attribute \u0027__getitem__\u0027\ntrying to get the password using configparser with new\nbehaviour introduced in configparser in py3\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8f7b8c82297ab20f46fd148f551f2360fa8abfb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8f7b8c82297ab20f46fd148f551f2360fa8abfb6"}]},"branch":"refs/heads/stable/train"},"44aeac2db9f623a0e920b91dec781cc93dce35ac":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-06-17 15:36:57.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/89/735989/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/7"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-17 15:26:38.000000000","tz":-240},"subject":"Fix \"Remove VxFlex OS credentials\" regression","message":"Fix \"Remove VxFlex OS credentials\" regression\n\nChange Ia1d2b2151e5676037d40bfaf388b54023fc37093 uses a ConfigParser\ninstance in a way that is incompatible with python 2.7, resulting in\na \"ConfigParser instance has no attribute \u0027__getitem__\u0027\" error.  This\npatch accesses the ConfigParser using the \"old\" API that works in both\nPython 2.7 and Python 3 and adds some tests.\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/44aeac2db9f623a0e920b91dec781cc93dce35ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/44aeac2db9f623a0e920b91dec781cc93dce35ac"}]},"branch":"refs/heads/stable/train"},"12d252db9cb9deffea3c87b86ea71b3013d93892":{"kind":"REWORK","_number":8,"created":"2020-06-17 16:00:27.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/89/735989/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/735989/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/735989/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/735989/8"}}},"commit":{"parents":[{"commit":"55fc99852166f72b95d85dc917197f5544861e7c","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55fc99852166f72b95d85dc917197f5544861e7c"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-17 16:00:19.000000000","tz":-240},"subject":"Fix \"Remove VxFlex OS credentials\" regression","message":"Fix \"Remove VxFlex OS credentials\" regression\n\nChange Ia1d2b2151e5676037d40bfaf388b54023fc37093 uses a ConfigParser\ninstance in a way that is incompatible with python 2.7, resulting in\na \"ConfigParser instance has no attribute \u0027__getitem__\u0027\" error.  This\npatch accesses the ConfigParser using the \"old\" API that works in both\nPython 2.7 and Python 3 and adds some tests.\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/12d252db9cb9deffea3c87b86ea71b3013d93892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/12d252db9cb9deffea3c87b86ea71b3013d93892"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
