)]}'
{"id":"openstack%2Fos-brick~736059","triplet_id":"openstack%2Fos-brick~stable%2Frocky~I6df6429f5b98ef75aef83da7faf49845b30236eb","project":"openstack/os-brick","branch":"stable/rocky","topic":"bug/1883654","hashtags":[],"change_id":"I6df6429f5b98ef75aef83da7faf49845b30236eb","subject":"Fix get password from config error in py27","status":"ABANDONED","created":"2020-06-16 22:53:52.000000000","updated":"2020-06-17 14:32:52.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c05fb3b0fec62d22661d61f6115382d34fd258e","_number":736059,"virtual_id_number":736059,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-06-17 00:40:34.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2020-06-17 00:21:16.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2020-06-17 00:17:32.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-16 23:51:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-16 23:39:21.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-2,"date":"2020-06-16 23:30:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":-1,"date":"2020-06-17 13:40:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2020-06-17 14:20:39.000000000","_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"date":"2020-06-17 14:20:21.000000000","_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 23:11:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-16 23:39:21.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-06-16 23:51:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 00:17:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 00:21:16.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 00:40:34.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 13:40:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-17 14:20:21.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-06-17 14:20:39.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"}],"messages":[{"id":"04b8255238ac46739c05f3301576e9a7d92414d1","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 22:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1ad268c6e0096374657d2189b85927b61db22c2","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 22:57:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97c450ffc9ef0219ecc08e393be2365b6dffa575","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 22:57:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3ca345489486dc50462546653cf129303795e16d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:01:12.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a2d96c9922b83236810bbb74f2503ec8d852cc21","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 23:06:31.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/736059/2/iscsi/ : FAILURE in 8m 43s\n- os-brick-smb http://cloudbase-ci.com/os-brick/736059/2/smb/ : FAILURE in 7m 46s","accounts_in_message":[],"_revision_number":2},{"id":"84cbaa2e80b19c1ed7b7f3bc8ae738ef02e79ed7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:10:12.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"5e793210dda6c4ad04ac3a3b6c7e71932b407e5a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-16 23:11:00.000000000","message":"Patch Set 2: Code-Review-2\n\nPlease don\u0027t backported until merged to more recent branches.","accounts_in_message":[],"_revision_number":2},{"id":"645e558f6d9d026d4275a55562c179b01c37e114","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 23:13:35.000000000","message":"Patch Set 2:\n\nwe already have the patch for train\nhttps://review.opendev.org/#/c/735989/","accounts_in_message":[],"_revision_number":2},{"id":"d8313502c8dd64d93b0327143b9fb7ccbc47e2fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:18:08.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c857d4d3ee3a488a8cff0838f9ad7a16 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27460ad0c15f431696137985d0193378 : FAILURE in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/feba5ef0833e461ca04fcf8c1a5b11f7 : SUCCESS in 7m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90ab49566e314ead9acdf5ac291ddc12 : SUCCESS in 5m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7f963a44521147599b902c043cf11bb7 : FAILURE in 14m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9f7e0bc40f2499b99f899f8085a897f : SUCCESS in 8m 06s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/9db757ff0b7a402fa3eb820525f55300 : FAILURE in 18m 35s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/30e8de47721d43be83dfd564eb8b4c44 : FAILURE in 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a23ca9c9e4ef50eb67d0e9c85f2e81ce6429544e","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-16 23:18:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/59/736059/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9c6f6b5 : FAILURE in 17m 24s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/59/736059/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/c349223 : FAILURE in 17m 06s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"db3be212050befe3b67a07fd5f941074789b31d7","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 23:20:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2d7ae80210f1410992f7569c830359eaf9bc70a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:24:32.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"967a7ab0504026b0b4d99711efe6b6d956239c76","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 23:29:09.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/736059/3/iscsi/ : FAILURE in 7m 56s\n- os-brick-smb http://cloudbase-ci.com/os-brick/736059/3/smb/ : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":3},{"id":"93d1e006ec7220141a3ae52962edfb51fd175a6a","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-06-16 23:30:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9d9120993c0cbc706930c510700f446d4fb47899","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:30:53.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"b3e2f1ef733c01942210e4a4c03db88c17bd0b76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:34:17.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"d2b2c39067ee38408a834a772475c8f27df10e92","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-16 23:39:21.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/736059/4/iscsi/ : FAILURE in 7m 51s\n- os-brick-smb http://cloudbase-ci.com/os-brick/736059/4/smb/ : FAILURE in 8m 17s","accounts_in_message":[],"_revision_number":4},{"id":"946ee638b57e6f01bd50b80f2bebc5acc47c3835","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-16 23:43:03.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"57138cc637bca61dfaedc35ee6d04c266907ceeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:51:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8e8243bb39954ba9b7cdcaf9b18cba44 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c50ba6eebc54072acde36e6c57c194a : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d516167b7cde4dd68ae13e58d51fe11e : SUCCESS in 6m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b4e3ee64465449f6966c0c2b1d12fbfc : SUCCESS in 6m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9036a6a9d0824194b414f01b4d2d4590 : FAILURE in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d1f2d1ff2046dfa3391ebb7bba9b3a : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2ce3ace4b22496da9b10826f836777b : SUCCESS in 8m 40s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/e4ae8c64229a4a9c95feb7f7f58411e4 : FAILURE in 18m 57s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/353c96da4b0b4cd3b9e35768efb910d9 : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":4},{"id":"7951869d741d1e61d9779bf1d8a9de9ec1236d98","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-17 00:17:32.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/59/736059/4/check/tempest-dsvm-full-bionic/ce4643e/ : POST_FAILURE in 22m 28s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/59/736059/4/check/devstack-plugin-ceph-dsvm-bionic/0675a9b/ : POST_FAILURE in 24m 50s","accounts_in_message":[],"_revision_number":4},{"id":"641d20e413b67496b4dff66c20f5ab5f9e230691","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-17 00:21:16.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/736059/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"75316b11f2875fbd816e48c3ab0045cc12c92e40","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-17 00:40:34.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/59/736059/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e40d606 : FAILURE in 17m 50s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/59/736059/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ab5fb7e : FAILURE in 17m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"a8c6982badbc74f842c9b2ff9fb212a69c125eed","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-17 13:40:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nSee comment inline about the commit message","accounts_in_message":[],"_revision_number":4},{"id":"240d07f7719e9bd3aacc34d518c04d1d23310f3c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 14:32:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f477e5551dc3edc47e083ab7e146a9a554011cb4","revisions":{"43162fe61d8b608ab52a4dd921ef1ba6a976ada3":{"kind":"REWORK","_number":1,"created":"2020-06-16 22:53:52.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/59/736059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/736059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/736059/1"}}},"commit":{"parents":[{"commit":"88fbec6b1d8ac9a0f8f259a825e6139781021e3f","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88fbec6b1d8ac9a0f8f259a825e6139781021e3f"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 22:52:04.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 22:53:42.000000000","tz":0},"subject":"Fix get password from config error in py27","message":"Fix get password from config error in py27\n\ngetting password from config file is using a configparser\nbehaviour introduced in py3 and fails with py2.7 with below error\nConfigParser instance has no attribute \u0027__getitem__\u0027\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: I6df6429f5b98ef75aef83da7faf49845b30236eb\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/43162fe61d8b608ab52a4dd921ef1ba6a976ada3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/43162fe61d8b608ab52a4dd921ef1ba6a976ada3"}]},"branch":"refs/heads/stable/rocky"},"ae026dfeb75cc829576bbd801901535e0f93a8f3":{"kind":"REWORK","_number":2,"created":"2020-06-16 22:57:27.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/59/736059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/736059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/736059/2"}}},"commit":{"parents":[{"commit":"88fbec6b1d8ac9a0f8f259a825e6139781021e3f","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88fbec6b1d8ac9a0f8f259a825e6139781021e3f"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 22:52:04.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 22:56:51.000000000","tz":0},"subject":"Fix get password from config error in py27","message":"Fix get password from config error in py27\n\ngetting password from config file is using a configparser\nbehaviour introduced in py3 and fails with py2.7 with below error\nConfigParser instance has no attribute \u0027__getitem__\u0027\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: I6df6429f5b98ef75aef83da7faf49845b30236eb\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ae026dfeb75cc829576bbd801901535e0f93a8f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ae026dfeb75cc829576bbd801901535e0f93a8f3"}]},"branch":"refs/heads/stable/rocky"},"c6725c65b985a491d0ce08b6a51a0ae228c45883":{"kind":"REWORK","_number":3,"created":"2020-06-16 23:20:52.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/59/736059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/736059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/736059/3"}}},"commit":{"parents":[{"commit":"88fbec6b1d8ac9a0f8f259a825e6139781021e3f","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88fbec6b1d8ac9a0f8f259a825e6139781021e3f"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 22:52:04.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 23:20:25.000000000","tz":0},"subject":"Fix get password from config error in py27","message":"Fix get password from config error in py27\n\ngetting password from config file is using a configparser\nbehaviour introduced in py3 and fails with py2.7 with below error\nConfigParser instance has no attribute \u0027__getitem__\u0027\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: I6df6429f5b98ef75aef83da7faf49845b30236eb\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c6725c65b985a491d0ce08b6a51a0ae228c45883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c6725c65b985a491d0ce08b6a51a0ae228c45883"}]},"branch":"refs/heads/stable/rocky"},"f477e5551dc3edc47e083ab7e146a9a554011cb4":{"kind":"REWORK","_number":4,"created":"2020-06-16 23:30:43.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/59/736059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/736059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/736059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/736059/4"}}},"commit":{"parents":[{"commit":"88fbec6b1d8ac9a0f8f259a825e6139781021e3f","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88fbec6b1d8ac9a0f8f259a825e6139781021e3f"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 22:52:04.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 23:30:09.000000000","tz":0},"subject":"Fix get password from config error in py27","message":"Fix get password from config error in py27\n\ngetting password from config file is using a configparser\nbehaviour introduced in py3 and fails with py2.7 with below error\nConfigParser instance has no attribute \u0027__getitem__\u0027\nsee https://docs.python.org/3/whatsnew/3.2.html#configparser\nusing the configparser old way should work for py 3/2.7.\n\nChange-Id: I6df6429f5b98ef75aef83da7faf49845b30236eb\nCloses-Bug: 1883654\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f477e5551dc3edc47e083ab7e146a9a554011cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f477e5551dc3edc47e083ab7e146a9a554011cb4"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
