)]}'
{"id":"openstack%2Fos-brick~736788","triplet_id":"openstack%2Fos-brick~master~I0c783313207b3f402b274360ee791c343030f781","project":"openstack/os-brick","branch":"master","topic":"trivial","hashtags":[],"change_id":"I0c783313207b3f402b274360ee791c343030f781","subject":"Add TODO to switch to new style rbd commands","status":"MERGED","created":"2020-06-18 17:45:19.000000000","updated":"2020-07-30 20:51:18.000000000","submitted":"2020-07-30 19:27:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736788-1596137265107-566d28bb","meta_rev_id":"907e5414d58048b2a5cfab2837c4a254c785e52a","_number":736788,"virtual_id_number":736788,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-30 18:57:23.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-30 13:18:27.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-30 18:17:18.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 19:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-30 14:18:26.000000000","_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"date":"2020-07-30 18:13:06.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-30 16:32:08.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-30 18:04:28.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-07-30 14:16:36.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-30 14:21:54.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2020-07-30 13:50:28.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-07-30 16:37:16.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"date":"2020-07-30 13:14:14.000000000","_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-30 13:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-30 13:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 09:25:04.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-06-20 09:57:11.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-07 20:35:22.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-07-30 13:14:14.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-07-30 13:18:27.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 13:50:28.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-30 13:57:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-30 14:16:36.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 14:18:26.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"REVIEWER"},{"updated":"2020-07-30 14:21:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-30 16:32:08.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 16:37:16.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-07-30 18:04:28.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 18:13:06.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 18:17:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 18:57:23.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 19:27:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b85c730c2f426351ba7a36e15b4d0a80e10d4d2e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-18 17:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48f8aad7785292ea20c0f0a22d3bd5c003ad217c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-18 17:53:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c2cc64ae6642cd22ab6e2c1208f8b3b7e5ac4e93","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-18 17:55:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"20eacb797a3622dd79d12f8c74b254e7f07f8d60","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-18 18:00:48.000000000","message":"Topic changed from bug/1884052 to trivial","accounts_in_message":[],"_revision_number":2},{"id":"2ab847073418870b74a36373f80018ce40962747","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-18 18:05:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"09fc411aae2038adb7459a5fc3c62f7ae21d6721","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-18 19:02:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/736788/2/iscsi/ : SUCCESS in 1h 05m 34s\n- os-brick-smb http://cloudbase-ci.com/os-brick/736788/2/smb/ : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"57a9482da51a8a9cb66d60b6adf734049e0ac004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:47:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/600ac1a98b0c462c9d3a2d07c45e4a8f : SUCCESS in 1h 47m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41007a53e49245ceb715c6de1fd16c56 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41690cec39594a5291171a3762a36fa7 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b4184bdbb36462ba4df150a17e4ef82 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68a2c352eb23454a94c3f49da4f07852 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de57ea1164b54527b44e2fd2d335f31b : SUCCESS in 11m 02s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/7ec036c716064dc68c5553914e9b15c8 : SUCCESS in 1h 05m 30s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c3c2ae223fd34467b64d840bfa76125e : SUCCESS in 1h 22m 56s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/622e14dfe5cd490cab1129d8a44e1c86 : SUCCESS in 55m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f6fb4c47449255ccf1ad81e39d866a54f1965cb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-18 20:15:46.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-88-736788-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"8f6b6bf7da0117783bd59271ddae7e6e4a35ad25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-18 20:25:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"53d6d101dd7aaa5242fcc7c5208f96c51d4c8315","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-18 20:26:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/88/736788/2/upstream-check/cinder-cDOT-iSCSI/e7bd275/ : FAILURE in 37m 45s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/88/736788/2/upstream-check/cinder-cDOT-iSCSI-IPv6/16f291b/ : FAILURE in 30m 04s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/88/736788/2/upstream-check/cinder-cDOT-NFS/ac27fb1/ : FAILURE in 37m 26s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/88/736788/2/upstream-check/cinder-cDOT-NFS-IPv6/cb34ff7/ : FAILURE in 30m 14s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/88/736788/2/upstream-check/cinder-cDOT-FCP/b136ab8/ : FAILURE in 21m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"75b4769cf5b141f81b775b6a6dfdc3d0b7a67ab4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-18 20:26:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/88/736788/2/upstream-solidfire-check/cinder-solidfire-iSCSI/488139e/ : FAILURE in 37m 47s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"2f5989d6567e6e863601ed5191c2773cec0f6a5b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-18 20:34:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/88/736788/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/2b581c8 : FAILURE in 26m 10s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/88/736788/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/3956341 : FAILURE in 19m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"f163859f227c5f8787b8a5147be17cdb0d177619","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-18 22:31:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/2/check/tempest-dsvm-full-bionic/b15e397/ : SUCCESS in 1h 39m 42s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/2/check/devstack-plugin-ceph-dsvm-bionic/845d06e/ : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":2},{"id":"78bbfc2f4d9b7dcc46f229f59382ca21c01b4286","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-18 23:14:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0c76a67c8d70a4ead4f24c57429570f2818fddf8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-06-19 04:57:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/88/736788/2/check/DellEMC_SC_FC-os-brick/1b377be/DellEMC_SC_FC-os-brick/19 : FAILURE in 16m 57s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/88/736788/2/check/DellEMC_SC_iSCSI-os-brick/365960b/DellEMC_SC_iSCSI-os-brick/16 : SUCCESS in 2h 29m 47s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"fcb0c23c76c49f4150115b92c222f082a250a448","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-19 05:07:37.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d29ae90d1029dff314e4859e9a02123aab4ca1e5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-19 17:44:07.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2c24bc6b9dc6be982bd99ec493e58be1bdf50411","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-19 17:53:42.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"dedd676871ad5501d0a1c049eb1a5ea66251a576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 19:25:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7766c4c24910408489ffb60eeb3165da : SUCCESS in 1h 24m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d52c7ed70ca640f49a8b989e8792f3e8 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f09758afc724cc9b7ebd18423f781f3 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed1428c1c20a49dbaec29bf0bc72286c : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dad4f79e48346c1953a077c79d94cf7 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b223f098ad643988dae31de8afb8d72 : SUCCESS in 11m 22s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/be182fd8d9b845fab459abc604fcb428 : SUCCESS in 1h 19m 15s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a7b8b4d297943dc8d45de063ba4d4b5 : SUCCESS in 1h 17m 35s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/acd96059249e4942908a93c01128ed7b : SUCCESS in 1h 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"53c4f7f3fb22576b9d30f8617a93e43d2f66aef1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-19 19:52:40.000000000","message":"Patch Set 3:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-88-736788-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"a4a9cfac47817bc6d80434df7f77b0f43951a379","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-06-19 19:53:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/88/736788/3/upstream-solidfire-check/cinder-solidfire-iSCSI/d54cdd7/ : FAILURE in 27m 20s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"72a612b4048cf94f75c30a76f9f9b1aad151b37d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-19 20:02:21.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"977622a057d905ef0a844f2063a65aebce7646b6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-19 20:03:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/88/736788/3/upstream-check/cinder-cDOT-iSCSI/1f4e520/ : FAILURE in 16m 11s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/88/736788/3/upstream-check/cinder-cDOT-iSCSI-IPv6/a6ef2b9/ : FAILURE in 37m 44s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/88/736788/3/upstream-check/cinder-cDOT-NFS/fcc160e/ : FAILURE in 16m 03s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/88/736788/3/upstream-check/cinder-cDOT-NFS-IPv6/4798b40/ : FAILURE in 34m 30s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/88/736788/3/upstream-check/cinder-cDOT-FCP/5a02022/ : FAILURE in 20m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"75cb3b89c90dc3e5d7f05013d283246f0fcd622b","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-19 22:02:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/88/736788/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/2cb2e01 : FAILURE in 26m 11s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/88/736788/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/55de728 : FAILURE in 20m 15s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"b4c773f9e5c463df235b33c61ea56ed7c3904201","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-06-20 06:15:46.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/736788/3/iscsi/ : FAILURE in 15m 04s\n- os-brick-smb http://cloudbase-ci.com/os-brick/736788/3/smb/ : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":3},{"id":"57389d94b0bad252ff2dcfd96e0af5e704874d2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-20 06:25:26.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c0d506b4f451acd04ab4f48c6301a0d9ff50ff5e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-20 09:47:37.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/3/check/tempest-dsvm-full-bionic/f473fe4/ : SUCCESS in 1h 27m 58s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/3/check/devstack-plugin-ceph-dsvm-bionic/ec0e929/ : FAILURE in 1h 51m 19s","accounts_in_message":[],"_revision_number":3},{"id":"5cb6d28948011738a9903b7fb68f764c146aac4a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-20 09:57:11.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"d82e319145bf908235d898325103cd52d7c49c94","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-07-07 20:35:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"03300bc0b274fc952335b2fc4275e88e847fe5af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 14:49:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"75c72ce9f69a93e05a28c8c7c8308aafaf3c2458","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-30 10:32:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cb1ef4f11b7391a76a719a17420e98f647fc1f91","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-30 11:00:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/88/736788/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/1d61c3e : FAILURE in 27m 08s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/88/736788/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/8c9e3f6 : FAILURE in 20m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"05504df3b1f385c15416b77b0fb42c82c3e8b105","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-07-30 11:21:50.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/736788/4/iscsi/ : SUCCESS in 48m 34s\n- os-brick-smb http://cloudbase-ci.com/os-brick/736788/4/smb/ : FAILURE in 38m 43s","accounts_in_message":[],"_revision_number":4},{"id":"467c4f873bf2183844b1565940c122f2400ed83c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 12:06:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e52d5c11372d4d94b337014186fb04cc : SUCCESS in 1h 21m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f12d487724d40bdb6136546aacc25bc : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288db03794054d3398193e6f529774ed : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0e4ac3148c7432eb4f1989abd2d2ce1 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e97ccbe1d604f62a6cb99b408a7cf6f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f754120ba7a4e429885a77991bbf42b : SUCCESS in 7m 58s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/df824da66b0741739b98bc97711a5862 : SUCCESS in 1h 22m 26s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5f84e2ffc974ac098b0957854ae1dba : SUCCESS in 1h 27m 43s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f14543b7f47546cc8660c954be6b8a1e : SUCCESS in 1h 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"db61f3f9aa7c8f59cf4c736aba5d75d00f1e317c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 12:24:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_VNX_FC_os-brick/11d9a9e/DellEMC_VNX_FC_os-brick/28 : FAILURE in 15m 34s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_VNX_iSCSI_os-brick/07e5806/DellEMC_VNX_iSCSI_os-brick/22 : FAILURE in 17m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"52350415c103a2022f2b754f242ee6c58177cf34","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-30 12:29:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/4/check/tempest-dsvm-full-bionic/ea1971d/ : SUCCESS in 1h 31m 20s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/4/check/devstack-plugin-ceph-dsvm-bionic/312cf6b/ : SUCCESS in 1h 46m 03s","accounts_in_message":[],"_revision_number":4},{"id":"8de8abce4049ab81ec4207cf20a9ca7efc11e1ed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-30 12:58:35.000000000","message":"Patch Set 4:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-88-736788-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"4264d20c285c21daa529af8b250b165d0a993524","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-07-30 13:14:14.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/88/736788/4/check/DellEMC_PowerStore_FC-ob/b464c4c/DellEMC_PowerStore_FC-ob/9 : SUCCESS in 1h 07m 43s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/88/736788/4/check/DellEMC_PowerStore_iSCSI-ob/1095b7f/DellEMC_PowerStore_iSCSI-ob/9 : SUCCESS in 1h 04m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"4f9117e5e739e70b31158c590aec6d9c6daf5abf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 13:18:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_PowerFlex_os-brick/e072c54/DellEMC_PowerFlex_os-brick/34 : SUCCESS in 1h 11m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"9fac50f2ad9390c90d981d796c8a099994c44a08","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-30 13:50:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/88/736788/4/upstream-solidfire-check/cinder-solidfire-iSCSI/181932f/ : SUCCESS in 1h 43m 43s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"4df67212976c19c8ece6eadd92f17f3f3ef3f753","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-30 13:57:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nMinor change that was previously approved and just rebased, so restoring approval.","accounts_in_message":[],"_revision_number":4},{"id":"d85e5cbe89b5e5a85672189b4e72e8c3c4d8c422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 13:58:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"767c95807008776a7aa3e735254d0e770e5734c8","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 14:10:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_XtremIO_FC_os-brick/452428d/DellEMC_XtremIO_FC_os-brick/56 : FAILURE in 52m 02s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_XtremIO_iSCSI_os-brick/d105d10/DellEMC_XtremIO_iSCSI_os-brick/56 : SUCCESS in 1h 20m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"4aca110ab0ef8fc54da96c339fe11180ab6d262a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-30 14:16:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/88/736788/4/upstream-check/cinder-cDOT-iSCSI/ea70924/ : SUCCESS in 35m 17s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/88/736788/4/upstream-check/cinder-cDOT-iSCSI-IPv6/4d6d01d/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/88/736788/4/upstream-check/cinder-cDOT-NFS/bd0178b/ : SUCCESS in 35m 24s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/88/736788/4/upstream-check/cinder-cDOT-NFS-IPv6/f551396/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/88/736788/4/upstream-check/cinder-cDOT-FCP/d184259/ : FAILURE in 2h 09m 55s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"fe48c602aae10931c2f33dbeb021a8208151cf89","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-07-30 14:18:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_SC_FC-os-brick/25889fd/DellEMC_SC_FC-os-brick/35 : SUCCESS in 1h 16m 40s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_SC_iSCSI-os-brick/d0e07bb/DellEMC_SC_iSCSI-os-brick/31 : SUCCESS in 2h 11m 50s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"d7e5f3622bb135e0c14ce608b5ce5737fec93d3f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-30 14:21:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_Unity_FC_os-brick/9d52604/DellEMC_Unity_FC_os-brick/40 : FAILURE in 1h 16m 57s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_Unity_iSCSI_os-brick/3a52379/DellEMC_Unity_iSCSI_os-brick/38 : SUCCESS in 1h 38m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"f1e59b924390debc968176698dd4b4c3faf57976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 15:53:57.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17e607120f3340b6a029be4f9cc8daf9 : SUCCESS in 1h 17m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66382fc045d7488fbda696a9ca0f5f19 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/736a2b5dc3414d21bed003210447f31c : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64d9919eea344022b377f1e620579889 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/006d96f407844347b253cdbf3cb9ec49 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e16886b9d145a5ab5c3aaaabb61ff6 : SUCCESS in 5m 55s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/856560f7646640c28dc74b710981f213 : FAILURE in 1h 52m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7643fee5034a57dca914fa9b0b279218259ced8c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-30 15:54:27.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b521dfb2ca42d43ad4cee9af26866aeac50f530d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-30 16:32:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/88/736788/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e75eed2 : FAILURE in 20m 39s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/88/736788/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/ba57c00 : FAILURE in 27m 17s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"3d31246fd5fa2009750c73a49c8b8d8507dc0f67","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-07-30 16:37:16.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/736788/4/iscsi/ : SUCCESS in 42m 29s\n- os-brick-smb http://cloudbase-ci.com/os-brick/736788/4/smb/ : FAILURE in 40m 18s","accounts_in_message":[],"_revision_number":4},{"id":"1aa0c722436deede6c6588a749a58f18846ab01d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 17:46:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5e9deebf20a46d0a0d67413e140add7 : SUCCESS in 1h 25m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a56619554c104614b7b81e7cf747ebf9 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20788ef79fe450195d967fd8daeaa57 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a5fc0b70087479c9b5248eb2dd3ce76 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffdfefe22e4e491caf3c2120ced29f37 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f31cec6d58a4d398ed4e4ad08fcb2ac : SUCCESS in 5m 41s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/b4ade59f2add40aa8e7fae50209fbae0 : SUCCESS in 1h 09m 52s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3052ed43d5f24c6db86231889bd5b4ef : SUCCESS in 1h 41m 45s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/92aa456a43144e5daa4de53a854b6f83 : SUCCESS in 1h 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4fd2c5f7c68fe2319d4afb1b60dcad6495b5a658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 17:47:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2db1fdee0ad642cf02595b3db0ddf4f64ec9835e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 18:04:28.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_VNX_FC_os-brick/a87c9f9/DellEMC_VNX_FC_os-brick/33 : FAILURE in 15m 24s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_VNX_iSCSI_os-brick/9422a15/DellEMC_VNX_iSCSI_os-brick/27 : FAILURE in 17m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"5e5a11a6bf0b37c91b8d76d5d0be2304f0b927bb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-07-30 18:13:06.000000000","message":"Patch Set 4:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-88-736788-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"4e98d28df55a0764fa9b14360f38dad6b82743f0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-30 18:17:18.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/4/check/tempest-dsvm-full-bionic/b76a241/ : SUCCESS in 1h 39m 12s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/88/736788/4/check/devstack-plugin-ceph-dsvm-bionic/ff6979b/ : SUCCESS in 1h 49m 06s","accounts_in_message":[],"_revision_number":4},{"id":"9cf5d48d7f128858523945e8f757b284cc03d033","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 18:57:23.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_XtremIO_FC_os-brick/b439b10/DellEMC_XtremIO_FC_os-brick/None : EXCEPTION\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/88/736788/4/check/DellEMC_XtremIO_iSCSI_os-brick/c483a12/DellEMC_XtremIO_iSCSI_os-brick/None : EXCEPTION\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"98151205ce7fa7e2bbca9b7768aaf1cb3cd87a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 19:27:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e28bb2f74ed744c6b31e7e096ec2a95d : SUCCESS in 1h 18m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c3b528da4db4e8095eef6a1639dea64 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e4c4042e6e8407682ce437b3f364b05 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67016dbe1b9045d38ffea67b67abfeac : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91b25bb4bc304ba9a375e35c4ac1ff05 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26835087b5a4af9a4b580d03399b0fd : SUCCESS in 7m 15s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e51ef8310c394715a836aca5ab70ee9d : SUCCESS in 1h 38m 05s","accounts_in_message":[],"_revision_number":4},{"id":"6abb7bcb9c72b270811c0d8bfbbe09ff3aac05c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 19:27:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"72868f195689d77093d3953a574c50619a03de6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 19:29:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218a929f620945a29d0add0bb6a2c4ff : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4},{"id":"907e5414d58048b2a5cfab2837c4a254c785e52a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-30 20:51:18.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/88/736788/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4bd4066e76e54acfcb8ce08f6c9dc0ef0788aed6","revisions":{"53dc92824722395c3d5889fb1cf64f7da1f3290f":{"kind":"REWORK","_number":1,"created":"2020-06-18 17:45:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/736788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/88/736788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/88/736788/1"}}},"commit":{"parents":[{"commit":"1f57ac7fa2b5f4d46105fc4a8c4a2c7f6b0c186d","subject":"rbd: Create symlink when udev is not present or configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1f57ac7fa2b5f4d46105fc4a8c4a2c7f6b0c186d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-06-18 17:44:22.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-06-18 17:44:22.000000000","tz":60},"subject":"Add TODO to switch to new style rbd commands","message":"Add TODO to switch to new style rbd commands\n\nThe old variants will likely go away at some point. Leave a breadcrumb\nhere for future people trying to figure out why things broke.\n\nChange-Id: I0c783313207b3f402b274360ee791c343030f781\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/53dc92824722395c3d5889fb1cf64f7da1f3290f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/53dc92824722395c3d5889fb1cf64f7da1f3290f"}]},"branch":"refs/heads/master"},"436f3084d77d6f5f12c19f89b9d0714813a76509":{"kind":"NO_CHANGE","_number":2,"created":"2020-06-18 17:55:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/736788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/88/736788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/88/736788/2"}}},"commit":{"parents":[{"commit":"c946fc397fbd80a41b526b4cf315e97d20c51234","subject":"rbd: Create symlink when udev is not present or configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c946fc397fbd80a41b526b4cf315e97d20c51234"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-06-18 17:44:22.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-06-18 17:55:02.000000000","tz":60},"subject":"Add TODO to switch to new style rbd commands","message":"Add TODO to switch to new style rbd commands\n\nThe old variants will likely go away at some point. Leave a breadcrumb\nhere for future people trying to figure out why things broke.\n\nChange-Id: I0c783313207b3f402b274360ee791c343030f781\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/436f3084d77d6f5f12c19f89b9d0714813a76509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/436f3084d77d6f5f12c19f89b9d0714813a76509"}]},"branch":"refs/heads/master"},"192ea87870d7962ea7de9bc75a5f71757dc769e1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-19 17:44:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/736788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/88/736788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/88/736788/3"}}},"commit":{"parents":[{"commit":"c4cd8794aa340f917cb3e6854ab40ee660d881a1","subject":"rbd: Create symlink when udev is not present or configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c4cd8794aa340f917cb3e6854ab40ee660d881a1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-06-18 17:44:22.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-06-19 17:43:58.000000000","tz":60},"subject":"Add TODO to switch to new style rbd commands","message":"Add TODO to switch to new style rbd commands\n\nThe old variants will likely go away at some point. Leave a breadcrumb\nhere for future people trying to figure out why things broke.\n\nChange-Id: I0c783313207b3f402b274360ee791c343030f781\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/192ea87870d7962ea7de9bc75a5f71757dc769e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/192ea87870d7962ea7de9bc75a5f71757dc769e1"}]},"branch":"refs/heads/master"},"4bd4066e76e54acfcb8ce08f6c9dc0ef0788aed6":{"kind":"REWORK","_number":4,"created":"2020-07-30 10:32:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/736788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/88/736788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/88/736788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/88/736788/4"}}},"commit":{"parents":[{"commit":"460e92c4ce57f8e053c173fbb39e9f2eb381f307","subject":"Merge \"rbd: Warn if ceph udev rules are not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/460e92c4ce57f8e053c173fbb39e9f2eb381f307"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-06-18 17:44:22.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-30 10:32:32.000000000","tz":60},"subject":"Add TODO to switch to new style rbd commands","message":"Add TODO to switch to new style rbd commands\n\nThe old variants will likely go away at some point. Leave a breadcrumb\nhere for future people trying to figure out why things broke.\n\nChange-Id: I0c783313207b3f402b274360ee791c343030f781\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4bd4066e76e54acfcb8ce08f6c9dc0ef0788aed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4bd4066e76e54acfcb8ce08f6c9dc0ef0788aed6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
