)]}'
{"id":"openstack%2Fos-brick~740318","triplet_id":"openstack%2Fos-brick~stable%2Fpike~Ie2db587c3bc379acd53cfd449788d171ae58dec5","project":"openstack/os-brick","branch":"stable/pike","topic":"bug/1823200","hashtags":[],"change_id":"Ie2db587c3bc379acd53cfd449788d171ae58dec5","subject":"Fix \"Remove VxFlex OS credentials\" regression","status":"MERGED","created":"2020-07-09 14:54:49.000000000","updated":"2020-07-25 02:33:26.000000000","submitted":"2020-07-24 23:36:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740318-1595633818484-46e369dc","meta_rev_id":"7d42dbf1d652702835b8232b2f53dbca97d799c6","_number":740318,"virtual_id_number":740318,"owner":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-24 21:03:21.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-07-24 22:20:07.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 20:27:35.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 23:03:48.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 23:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 22:28:14.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-07-24 20:57:38.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-24 18:07:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":2,"date":"2020-07-24 18:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":1,"date":"2020-07-24 18:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 20:37:35.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 14:54:49.000000000","updated_by":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"reviewer":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"state":"REVIEWER"},{"updated":"2020-07-24 18:07:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 18:31:43.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-07-24 20:27:35.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 20:37:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-24 20:57:38.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-07-24 21:03:21.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 22:20:07.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 22:28:14.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-07-24 23:03:48.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 23:36:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"338579847cbf8ed318f76888ea900e44bfceabeb","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-07-09 14:54:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f311abb6b019f389ade9876863cc4053bb82ddbb","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-07-09 15:14:21.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/740318/1/iscsi/ : FAILURE in 19m 09s\n- os-brick-smb http://cloudbase-ci.com/os-brick/740318/1/smb/ : FAILURE in 18m 55s","accounts_in_message":[],"_revision_number":1},{"id":"731be8f74a272677f328152d4752098f497a27cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 16:36:03.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/776f84a01c9347d4b20b7a4b57cdaf5f : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6def63af21743a3959b5cca3441abf4 : FAILURE in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/861612d0ffae4935b3d3d3bacb11504c : FAILURE in 5m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/86a8716d6c2b4dfa873a8a7c4a51a49b : FAILURE in 7m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf401f0d359e4d599db9143b8e2539a9 : FAILURE in 1h 35m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cfffd2d9fce40e6982de3f435b247b4 : SUCCESS in 7m 40s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/e8fab4c2eb87444f99c2df2d956377f3 : FAILURE in 26m 24s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/c7e519af421e4a5c9d9ab8926b5b67b6 : SUCCESS in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fae19ebbff5a865ee7a3b847d215741b890bbed6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-09 16:58:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/18/740318/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c21d802 : FAILURE in 10m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/18/740318/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6a4123b : FAILURE in 10m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"b6d7661eaf1cdc144169b515f4beae6c99ee1161","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-09 17:43:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/18/740318/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"01fbe08bdaa43c8420ac798aea8d963734ea314b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-10 04:19:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- legacy-tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/18/740318/1/check/legacy-tempest-dsvm-full-xenial/719c78c/ : FAILURE in 2h 27m 55s\n- legacy-devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/18/740318/1/check/legacy-devstack-plugin-ceph-dsvm-xenial/0e0badd/ : FAILURE in 56m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53e7ebe8e292e40ed1e5d0c49cf97f4ae0b715e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-10 04:56:35.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/18/740318/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"88bfde365fa354d83ee268b9059055cefe732c25","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-24 16:05:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d0a64ae78488e4ae50d10c7b129db8ede1551fb","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-24 16:17:56.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/18/740318/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/4367765 : FAILURE in 11m 09s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/18/740318/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/adbc1b0 : FAILURE in 11m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc23c4c899539319e8692c9e6f7f655dd195b00c","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-07-24 16:25:37.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/740318/2/iscsi/ : FAILURE in 19m 12s\n- os-brick-smb http://cloudbase-ci.com/os-brick/740318/2/smb/ : FAILURE in 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"37ad483aba9d7567cf687ef797fe689935f5463b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-24 16:40:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/18/740318/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"017ac4805383ebcda7f51ae7c70b0f727d09fc9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:54:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8f0dcaf3725b413fad92aae2da9759d5 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae39e909ff374744acd2420206281064 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/443882eabca6409680e5daa3278a7e08 : SUCCESS in 6m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6521c37adb834ec0b677426468cdad06 : SUCCESS in 6m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca73e9b14bd9491fb101208df3147695 : SUCCESS in 1h 44m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04b267be28f542f9b47e8a3dc7df4a9e : SUCCESS in 5m 43s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/204f3ceb753b4fb4baab94400f7d4bb4 : FAILURE in 27m 49s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/3288f219f6ab4b4288b4ebb160e91b49 : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d3fb2818d5a08b390ec7ce69bc85190e7cb12e82","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 18:07:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2f1fbdf17111f7d346783f5fc31ea7a0638984f1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-07-24 18:31:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis looks like an appropriate backport.","accounts_in_message":[],"_revision_number":2},{"id":"a3d8b9c120073f3bdd9f3ee629737d54b603cde7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:31:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ac7a8b8e7deb7c9a4a26431168c15ea96ca0505","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-24 18:46:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/18/740318/2/upstream-solidfire-check/cinder-solidfire-iSCSI/63c644d/ : FAILURE in 51m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"57efcfcc475b780162cad61a7d742d1e16b78019","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-24 19:37:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-iSCSI/e36464f/ : FAILURE in 26m 54s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-iSCSI-IPv6/d00613b/ : FAILURE in 1h 26m 01s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-NFS/76518d1/ : FAILURE in 1h 03m 24s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-NFS-IPv6/311ed4b/ : FAILURE in 1h 42m 57s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-FCP/224b7db/ : NOT_REGISTERED\n- cinder-7mode-iSCSI https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-7mode-iSCSI/fdd3c10/ : NOT_REGISTERED (non-voting)\n- cinder-7mode-NFS https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-7mode-NFS/f07b13d/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c5ebbac6bb4144b3167fa86d414ca00f750638fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 20:15:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/af2471bfc8344693ba58e6f8f6deccd6 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b8ed1548624e629b301dde60d68874 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40807d1def4a44e5b5c5d15b954f629b : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/92aaea225d8a435091ecaae70a1498f1 : SUCCESS in 5m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba554fdee8dc4ebc90cabca70e041a42 : FAILURE in 1h 38m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d90b4d31884a45adadb0658cf310536c : SUCCESS in 9m 36s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/d4e6014170bf4328b4c8fa82be21999c : SUCCESS in 1h 26m 43s","accounts_in_message":[],"_revision_number":2},{"id":"16f9469a58e3d969525990dd5792e4dd44c54fe9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-24 20:27:35.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- legacy-tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/18/740318/2/check/legacy-tempest-dsvm-full-xenial/84258a8/ : TIMED_OUT in 2h 42m 16s\n- legacy-devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/18/740318/2/check/legacy-devstack-plugin-ceph-dsvm-xenial/4e941f7/ : FAILURE in 57m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20aead58e1997052ffc8add5ade80eeeb61ee60c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-24 20:37:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7efeaf34e07da294b0b1932be3532854ded7d8f3","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-07-24 20:57:38.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/740318/2/iscsi/ : FAILURE in 19m 24s\n- os-brick-smb http://cloudbase-ci.com/os-brick/740318/2/smb/ : FAILURE in 19m 41s","accounts_in_message":[],"_revision_number":2},{"id":"cdf642fee80e51039102423299a63d9a0678cabe","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-24 21:03:21.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/18/740318/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/3b54a4e : FAILURE in 10m 40s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/18/740318/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/2179cee : FAILURE in 10m 38s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"2a23dd6f3c8cbcd923e982972471b2f97893744e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 22:00:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/254bdaef82724dbcb3265d8617d6302f : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b4015171354a888aa1786e536b0704 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6dbd41b7ae4146f288d31304073ed79d : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/463bc4d07c81437385cf7142904ce37c : SUCCESS in 4m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/736b671faa8942b1ac7ab832bc684561 : SUCCESS in 1h 19m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/656804030f604d9e8305a7cc961811f8 : SUCCESS in 8m 13s\n- legacy-tempest-dsvm-full-ceph-plugin-src-os-brick https://zuul.opendev.org/t/openstack/build/3c8412e5dfc146c2bfcd8630c5d21e70 : FAILURE in 26m 50s (non-voting)\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/1fe2da2c8d694cff8e5746f4cb94e019 : SUCCESS in 1h 20m 29s","accounts_in_message":[],"_revision_number":2},{"id":"126f9a7646c9e43cfb6a284405e312e0b87c495e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 22:00:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7571532740436d424b88c91cdda6e388dd12b8a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-24 22:20:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/18/740318/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d0197051e3add1dbd8d75cd6b3dde5efa2997345","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-07-24 22:28:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/18/740318/2/upstream-solidfire-check/cinder-solidfire-iSCSI/309c460/ : FAILURE in 27m 18s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"2d5708248ba26e715fc355fe5f8e83b3acea81c5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-24 23:03:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-iSCSI/0cc5fbc/ : FAILURE in 27m 47s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-iSCSI-IPv6/564a091/ : FAILURE in 52m 27s (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-NFS/4d05929/ : FAILURE in 27m 37s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-NFS-IPv6/8a418ca/ : FAILURE in 1h 03m 00s (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-cDOT-FCP/dcb3811/ : NOT_REGISTERED\n- cinder-7mode-iSCSI https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-7mode-iSCSI/f128380/ : NOT_REGISTERED (non-voting)\n- cinder-7mode-NFS https://logs.openstack.netapp.com/logs/18/740318/2/upstream-check/cinder-7mode-NFS/25055dd/ : NOT_REGISTERED (non-voting)\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7a23f08a86a2cf24d564704b02ca8b2ac6226be6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:36:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9b337e8a30bc4b3287d5ae057ae10032 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c50d3a7fe04a8bb806e31b07f013d4 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54d7e20560b94bcb874063d6379b7f5d : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cfb653d2ef94445aa865190553a7f197 : SUCCESS in 4m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/37c5c30398054d2f8f21bbfd1a9f1202 : SUCCESS in 1h 30m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42743682318e4a64a8dc59f7757f6b7b : SUCCESS in 5m 04s\n- legacy-tempest-dsvm-full-lio-src-os-brick https://zuul.opendev.org/t/openstack/build/55c8bce3cfce4aa19ca332fae4826290 : SUCCESS in 1h 25m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f3d6b70a7d1c0486f0f2ee5887198e74f53012a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:36:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"765534eb7cec56063b28919ef763a20c1a9d7453","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:38:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab0ebb91982c40a18da8f25d3732cf0d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2},{"id":"fc24afab0cb4f4a412292a8e5eca399e26285ab4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-24 23:49:24.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- legacy-tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/18/740318/2/check/legacy-tempest-dsvm-full-xenial/4986ebf/ : FAILURE in 2h 26m 53s\n- legacy-devstack-plugin-ceph-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/18/740318/2/check/legacy-devstack-plugin-ceph-dsvm-xenial/9896342/ : FAILURE in 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e32802f3acc1c21cf9a5524a6fad4e0f9308d02e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-07-25 02:33:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/18/740318/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e56b950e2984a6829f5e7cced28ab5e82d28d4b","revisions":{"39c0f9d8bd168bac912a388ba1a7df80662f1307":{"kind":"REWORK","_number":1,"created":"2020-07-09 14:54:49.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/18/740318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/18/740318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/18/740318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/18/740318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/18/740318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/18/740318/1"}}},"commit":{"parents":[{"commit":"099c8be3bf4cc362e58441d4c2aba03a1a37f61f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/099c8be3bf4cc362e58441d4c2aba03a1a37f61f"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"KeithMnemonic","email":"kberger@suse.com","date":"2020-07-09 14:54:48.000000000","tz":-240},"subject":"Fix \"Remove VxFlex OS credentials\" regression","message":"Fix \"Remove VxFlex OS credentials\" regression\n\nChange Ib7778ba9d38a68d8b56ca632c5f1c353d55830b0 uses a ConfigParser\ninstance in a way that is incompatible with python 2.7, resulting in\na \"ConfigParser instance has no attribute \u0027__getitem__\u0027\" error.  This\npatch accesses the ConfigParser using the \"old\" API that works in both\nPython 2.7 and Python 3 and adds some tests.\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n(cherry picked from commit 12d252db9cb9deffea3c87b86ea71b3013d93892)\n(cherry picked from commit 750999db0de0fcfc923cd453add808ca42a4e2b4)\nConflicts:\n- remove os_brick/privileged/scaleio.py (function with bug is in the\n  connector in this branch)\n- move the test file to tests/initiator/connectors/\n- modify the unit tests because the config file is accessed differently\n  in this branch\n(cherry picked from commit 867767ffdae124fcdb9b04a9089e226689c476d7)\n(cherry picked from commit f4019f7831e5366f7e5e40e9715f3bd981f2b7df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/39c0f9d8bd168bac912a388ba1a7df80662f1307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/39c0f9d8bd168bac912a388ba1a7df80662f1307"}]},"branch":"refs/heads/stable/pike"},"1e56b950e2984a6829f5e7cced28ab5e82d28d4b":{"kind":"REWORK","_number":2,"created":"2020-07-24 16:05:59.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/18/740318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/18/740318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/18/740318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/18/740318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/18/740318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/18/740318/2"}}},"commit":{"parents":[{"commit":"4047948f1ac8055a025972ad73ec3ec421450775","subject":"Remove VxFlex OS credentials from connection_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4047948f1ac8055a025972ad73ec3ec421450775"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-06-16 17:59:07.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-07-24 16:00:35.000000000","tz":-240},"subject":"Fix \"Remove VxFlex OS credentials\" regression","message":"Fix \"Remove VxFlex OS credentials\" regression\n\nChange Ib7778ba9d38a68d8b56ca632c5f1c353d55830b0 uses a ConfigParser\ninstance in a way that is incompatible with python 2.7, resulting in\na \"ConfigParser instance has no attribute \u0027__getitem__\u0027\" error.  This\npatch accesses the ConfigParser using the \"old\" API that works in both\nPython 2.7 and Python 3 and adds some tests.\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ie2db587c3bc379acd53cfd449788d171ae58dec5\nCloses-Bug: 1883654\n(cherry picked from commit 12d252db9cb9deffea3c87b86ea71b3013d93892)\n(cherry picked from commit 750999db0de0fcfc923cd453add808ca42a4e2b4)\nConflicts:\n- remove os_brick/privileged/scaleio.py (function with bug is in the\n  connector in this branch)\n- move the test file to tests/initiator/connectors/\n- modify the unit tests because the config file is accessed differently\n  in this branch\n(cherry picked from commit 867767ffdae124fcdb9b04a9089e226689c476d7)\n(cherry picked from commit f4019f7831e5366f7e5e40e9715f3bd981f2b7df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1e56b950e2984a6829f5e7cced28ab5e82d28d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1e56b950e2984a6829f5e7cced28ab5e82d28d4b"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
