)]}'
{"id":"openstack%2Fos-brick~747417","triplet_id":"openstack%2Fos-brick~master~I35c54556ea58768d078ac732c9361e32d141a5dd","project":"openstack/os-brick","branch":"master","topic":"bug/1892500","hashtags":[],"change_id":"I35c54556ea58768d078ac732c9361e32d141a5dd","subject":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi","status":"ABANDONED","created":"2020-08-21 13:46:01.000000000","updated":"2020-08-25 04:59:51.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f78c11997b7cea96f2e29d7a7124a77a4091f02b","_number":747417,"virtual_id_number":747417,"owner":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-25 04:58:57.000000000","_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2020-08-25 03:35:04.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2020-08-24 17:45:27.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"date":"2020-08-24 17:14:09.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-24 16:33:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-08-24 16:59:08.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2020-08-24 15:00:08.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"date":"2020-08-24 17:43:41.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-08-25 02:15:29.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"date":"2020-08-24 17:42:02.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"date":"2020-08-24 15:53:07.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"date":"2020-08-24 17:40:27.000000000","_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-25 02:25:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"date":"2020-08-24 14:54:50.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 07:17:14.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2020-08-24 09:42:18.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 13:08:38.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-08-24 13:18:59.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 13:51:48.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-08-24 13:52:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-24 14:08:25.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-08-24 15:00:08.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 15:53:07.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-08-24 16:33:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 16:59:08.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 17:14:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 17:40:27.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-08-24 17:42:02.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-08-24 17:43:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 17:45:27.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-25 02:15:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-08-25 02:25:01.000000000","updated_by":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"reviewer":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"state":"REVIEWER"},{"updated":"2020-08-25 03:35:04.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1843c9f88bb64e3a882828d1171532b09d9e6324","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-21 13:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c20d95dce1a53d7e26ae42159f42e6ccdc74b310","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-21 13:54:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/17/747417/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9a4d41c : FAILURE in 4m 25s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/17/747417/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/3158c92 : FAILURE in 4m 32s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b52aca60c3b9007f4edb66e235844eb367be7ea","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-08-21 14:36:48.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/747417/1/iscsi/ : FAILURE in 20m 41s\n- os-brick-smb http://cloudbase-ci.com/os-brick/747417/1/smb/ : FAILURE in 50m 26s","accounts_in_message":[],"_revision_number":1},{"id":"82b586598560b9c3ec8770c07b6e2601e3ff8415","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-08-21 15:16:08.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease add a unit test for this scenario.","accounts_in_message":[],"_revision_number":1},{"id":"0a5eeacf817bb66345d48d8b1ee3a2a28b74b029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:37:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0641ff75fb624782b0abb8aec6060656 : FAILURE in 1h 08m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8de24793848c4f589451ea77056db72a : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532a5819007f4e16bdf1c5384a06bd31 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/079d83e6856e42e59a0a00e02c22c02f : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f90d30b6bda419d9c9c38683257ce6a : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6994ca9c18644eae8e71974620f3588b : SUCCESS in 6m 17s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/6e8c8eb17a934d45bcd6a237fd5b91d9 : SUCCESS in 1h 02m 05s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9eb18989949c4f69bda0397197c12113 : SUCCESS in 1h 29m 41s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/66acb7acbd1245b3898f87d24a10f8b5 : SUCCESS in 1h 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3495cf1a2e11e0472bee6f71a1c519d1ffff725","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-21 16:36:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/1/check/tempest-dsvm-full-bionic/d5b7ffb/ : SUCCESS in 1h 38m 32s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/1/check/devstack-plugin-ceph-dsvm-bionic/4d3fb01/ : SUCCESS in 2h 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"97f20bfc81ff5f0a7eb6221da2c5e9a0b5085fdd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-21 16:55:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"be5ae59717bcb54868e5ef094061747e947196ac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-21 18:22:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/17/747417/1/check/DellEMC_Unity_FC_os-brick/94f8c3b/DellEMC_Unity_FC_os-brick/62 : FAILURE in 1h 42m 29s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/17/747417/1/check/DellEMC_Unity_iSCSI_os-brick/082dade/DellEMC_Unity_iSCSI_os-brick/59 : SUCCESS in 1h 49m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b8c2b809a3b0992cb1a7bea3a78f6186221251e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-21 22:12:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"99857aa3a9c2659d9d53436a79240e3da5fee58e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-22 10:22:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/17/747417/1/check/DellEMC_XtremIO_FC_os-brick/76f32f5/DellEMC_XtremIO_FC_os-brick/85 : SUCCESS in 55m 29s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/17/747417/1/check/DellEMC_XtremIO_iSCSI_os-brick/d232ff3/DellEMC_XtremIO_iSCSI_os-brick/84 : SUCCESS in 1h 18m 31s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"3fd0a31dc069d7e29689cccc9536dce72f60636d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-22 13:47:54.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9858e5255e259adf74f5e055aa3f7eae4e1408f9","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 02:17:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"27a1d9759a5cda5f95dbb2bd5261c9d67c672d24","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 02:19:08.000000000","message":"Patch Set 2:\n\n\u003e Please add a unit test for this scenario.\n\nHi Eric, \nchanged unit test for this scenario.\nthanks and regards\n\nSam","accounts_in_message":[],"_revision_number":2},{"id":"3d022e25551dad400180659c60b866878e90f724","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 02:22:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/17/747417/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9eaa3db : FAILURE in 4m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/17/747417/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/9f8d6ce : FAILURE in 4m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"eb5b0fd1397b8a8eb40d8812b339c307924e1b78","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-08-24 03:15:08.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/747417/2/iscsi/ : FAILURE in 48m 01s\n- os-brick-smb http://cloudbase-ci.com/os-brick/747417/2/smb/ : FAILURE in 56m 57s","accounts_in_message":[],"_revision_number":2},{"id":"031a7859e3839f7d4ecded97f49ec28866f537f3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 03:15:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"f2af14ca95edaaa365bcaf0f8a5d66b73a0560a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 03:49:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88900b9d94a74baba2aeb0afcc508aff : SUCCESS in 53m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93037a21c927485487b4c73f4c034734 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5000a6119545619c77a928c60e82f4 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3788bda66a3842bdb084274d46d8f3dc : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e7655eaa844734a0d56ac094c2453a : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fcb8a967b2e4c5981adf8c466ff40c1 : SUCCESS in 6m 43s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/206c47a252ff41028fb853fa77e059d3 : SUCCESS in 1h 14m 45s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c04198c1af854cfea2adb97ff8f6a92d : SUCCESS in 1h 19m 42s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/260cfc12473e40208cf715cb2dd0ec3f : SUCCESS in 1h 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"639acb059aac4e8bd3c78a1c984b6b6efc68f65e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-24 03:54:02.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/2/check/tempest-dsvm-full-bionic/00ef436/ : SUCCESS in 1h 12m 28s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/2/check/devstack-plugin-ceph-dsvm-bionic/4247388/ : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":2},{"id":"55797d4007480ea448301498c1448d1a8d8f3eee","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 04:01:21.000000000","message":"Patch Set 2:\n\nrun-DellEMC PowerStore CI","accounts_in_message":[],"_revision_number":2},{"id":"1e9f98769a641221c10c45a5eb15061c9c7f6e63","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-24 04:09:15.000000000","message":"Patch Set 2:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-17-747417-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"fa7c2a867deed3a3cf6dcc635c38b83324eeb390","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 04:18:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/17/747417/2/check/DellEMC_XtremIO_FC_os-brick/1743f82/DellEMC_XtremIO_FC_os-brick/87 : SUCCESS in 53m 29s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/17/747417/2/check/DellEMC_XtremIO_iSCSI_os-brick/0def988/DellEMC_XtremIO_iSCSI_os-brick/86 : SUCCESS in 1h 19m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"952119bb85ac36811d02d0d4f5ed79d88728f353","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 04:24:45.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"50dd085050ed8935691b972688be30ac84201a12","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 04:52:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/17/747417/2/upstream-check/cinder-cDOT-iSCSI/4a2d175/ : SUCCESS in 36m 33s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/17/747417/2/upstream-check/cinder-cDOT-iSCSI-IPv6/bf82724/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/17/747417/2/upstream-check/cinder-cDOT-NFS/7a08ea7/ : SUCCESS in 35m 16s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/17/747417/2/upstream-check/cinder-cDOT-NFS-IPv6/70d4956/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/17/747417/2/upstream-check/cinder-cDOT-FCP/a40ee26/ : SUCCESS in 1h 02m 51s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2f4495be63d685a8ede155c66ebf78864fb6a67a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-08-24 05:16:37.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/17/747417/2/check/DellEMC_PowerStore_FC-ob/a445ad5/DellEMC_PowerStore_FC-ob/30 : SUCCESS in 1h 06m 02s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/17/747417/2/check/DellEMC_PowerStore_iSCSI-ob/0523006/DellEMC_PowerStore_iSCSI-ob/30 : SUCCESS in 1h 15m 06s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"11c1b8f3baa2eaebe2b7e5ce879f2e9f5b1ee0b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-24 05:20:31.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/17/747417/2/check/DellEMC_Unity_FC_os-brick/4663e38/DellEMC_Unity_FC_os-brick/65 : SUCCESS in 1h 32m 47s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/17/747417/2/check/DellEMC_Unity_iSCSI_os-brick/ff58861/DellEMC_Unity_iSCSI_os-brick/62 : SUCCESS in 1h 54m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b61e8e7aa6dc5eaab79d0ca05c9f3f513464e1a8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-24 05:29:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/747417/2/upstream-solidfire-check/cinder-solidfire-iSCSI/e0dcbbc/ : SUCCESS in 1h 40m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"4d95f2b0ee2a636c6e56b4d6c0fbd5fa094447a3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 06:53:31.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/17/747417/2/check/DellEMC_PowerFlex_os-brick/f2add1c/DellEMC_PowerFlex_os-brick/58 : SUCCESS in 1h 15m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"81a07f7634f3f31a53df2262dd3897ecdda0c25f","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2020-08-24 07:17:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"494cfd35fe0f445c445121801833ca3798de5e46","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-08-24 08:14:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24ec27918d975214e90f83a19ffef00e6013cfc7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 08:33:34.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c1ff35ab3554830d486f01c762d9f92a1ab949c5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 09:42:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/17/747417/2/check/DellEMC_VNX_FC_os-brick/ec1937b/DellEMC_VNX_FC_os-brick/55 : FAILURE in 1h 15m 07s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/17/747417/2/check/DellEMC_VNX_iSCSI_os-brick/a70705b/DellEMC_VNX_iSCSI_os-brick/47 : FAILURE in 2h 47m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"60075c44c5c4749c70032ca206e6182f708019a4","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 10:41:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"91d96f40d8ebc634a5cb1f545f7e8869181725a5","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 10:42:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eecf069deb666060bf88a18fef14a944a60bdfd3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 10:43:18.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"4209fd974e2acb4fa983006080535b750152c8c6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 10:46:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/17/747417/3/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/fccac2c : FAILURE in 4m 21s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/17/747417/3/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/381ac9f : FAILURE in 4m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":3},{"id":"fdabcbbee10a9ee43936c14879e4867998e587bf","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-08-24 11:31:09.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/747417/3/iscsi/ : FAILURE in 49m 04s\n- os-brick-smb http://cloudbase-ci.com/os-brick/747417/3/smb/ : FAILURE in 47m 34s","accounts_in_message":[],"_revision_number":3},{"id":"6e03ef1ac67268fc0cc08f6489bfe550b3538ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:12:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46aa5a55469045429ba13ecce22f99fb : SUCCESS in 1h 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/041bf8206d2d4bf9b1f6ca1b551e9b71 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85e4f16b9e774f3bbc41eefe2dfef1ab : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37beb523f21848f7a94a69e171c06e8c : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1af8d8280791446c84d5d6cf02e4f9a7 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ce0d3e723e4550af5cccb66fc0fa31 : SUCCESS in 7m 04s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/c9f302c17c66405490b5ad5d2eb00f0c : SUCCESS in 1h 01m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0f86c875fe024f539b67bbe0f7e5621f : SUCCESS in 1h 25m 58s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c580beba90d04ebaa38ba003ea4f822c : SUCCESS in 1h 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ae1004eb28d57378ce6fb5ddc5d8e28380e5aa49","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-24 12:33:21.000000000","message":"Patch Set 3:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-17-747417-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"09af6b2a1d03fbbf2a31d213bba7a8c49f3af3f6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-24 12:52:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/3/check/tempest-dsvm-full-bionic/2dc1010/ : SUCCESS in 1h 32m 53s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/3/check/devstack-plugin-ceph-dsvm-bionic/6ef082d/ : SUCCESS in 1h 53m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ad6b3b140fd9a4354c55e92ac6f1359b7bc59e09","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-08-24 13:08:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f7ede99ea34a26b932b60e0bdef7f5d2fc61bf3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 13:10:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/17/747417/3/upstream-check/cinder-cDOT-iSCSI/451b322/ : SUCCESS in 38m 34s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/17/747417/3/upstream-check/cinder-cDOT-iSCSI-IPv6/0b59a72/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/17/747417/3/upstream-check/cinder-cDOT-NFS/6c8a67e/ : SUCCESS in 36m 42s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/17/747417/3/upstream-check/cinder-cDOT-NFS-IPv6/352952a/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/17/747417/3/upstream-check/cinder-cDOT-FCP/159c47c/ : SUCCESS in 57m 18s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"75268c167f89db7d7ddecd42820dc9f6d48771f2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-24 13:17:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/747417/3/upstream-solidfire-check/cinder-solidfire-iSCSI/2a0d4b4/ : SUCCESS in 1h 04m 28s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"85cc79df3a21f1a7f815e8b53aac95bf70a77661","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 13:18:59.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/17/747417/3/check/DellEMC_XtremIO_FC_os-brick/f420801/DellEMC_XtremIO_FC_os-brick/88 : SUCCESS in 53m 19s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/17/747417/3/check/DellEMC_XtremIO_iSCSI_os-brick/45e6ccc/DellEMC_XtremIO_iSCSI_os-brick/87 : SUCCESS in 1h 20m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"25286a5280fe61b3b5ed975baa0f3be68054017d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-08-24 13:22:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/17/747417/3/check/DellEMC_PowerStore_FC-ob/fa9e85a/DellEMC_PowerStore_FC-ob/31 : SUCCESS in 1h 07m 31s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/17/747417/3/check/DellEMC_PowerStore_iSCSI-ob/bc6a8f7/DellEMC_PowerStore_iSCSI-ob/31 : SUCCESS in 1h 09m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"7eb54e601f8a655dc61fd5fb5bd54bc200a1a4b3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 13:23:21.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/17/747417/3/check/DellEMC_PowerFlex_os-brick/e295702/DellEMC_PowerFlex_os-brick/59 : SUCCESS in 1h 10m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"e088472c66b1baceebfea8e9fd5be2b16440738d","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-08-24 14:08:25.000000000","message":"Patch Set 3:\n\nexcellent observation Eric.\nDoes that mean that the fix should check for the version lsscsi and only apply it if version is below 0.29?","accounts_in_message":[],"_revision_number":3},{"id":"303f8f6970275d90ab0d4fe396baeaa41651222d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 14:25:04.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"aa38ad29d0b069714727f390ce5e148b35b42c3b","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 14:44:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cbfdc4ccfcbd18701029474666688c488bc2f5c","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 14:45:10.000000000","message":"Patch Set 3:\n\n\u003e excellent observation Eric.\n \u003e Does that mean that the fix should check for the version lsscsi and\n \u003e only apply it if version is below 0.29?\n\nArkady,\nThere\u0027s no need to check for versions because latest patchset works for all versions. \nplease check the new ut test for examples.\nthanks and regards\nSam","accounts_in_message":[],"_revision_number":3},{"id":"5ea22688a465222b7dbac155fdc6c8e4a7d06595","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 14:54:50.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b99a87c841f6c5c1117779c9667f75cf9c231365","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-24 14:56:16.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nchanged the description.","accounts_in_message":[],"_revision_number":4},{"id":"aa70d58026eb0a7f3fe7b95eb561d10b3b93b7ad","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 15:00:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/17/747417/4/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/edec1d8 : FAILURE in 4m 22s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/17/747417/4/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/df73d72 : FAILURE in 5m 04s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":4},{"id":"057ec45b2ed06100cdecdf44fededaa6398e1b98","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-08-24 15:53:07.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/747417/4/iscsi/ : FAILURE in 51m 03s\n- os-brick-smb http://cloudbase-ci.com/os-brick/747417/4/smb/ : FAILURE in 57m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c7a2109a3382d4ec22a5d638c9321216333d00bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 16:33:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3c27afdc54e4376ad863f6af8412b84 : SUCCESS in 1h 16m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c60bf0d7cb64458b0717206f3734607 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d5e603d5574461b1a801e606699ebc : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fd20952f43e49379b713a8aef4d2ba7 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f572764864c84b968e865d481ef1f50c : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7d2e7a3a9140cdaad495a6eb692781 : SUCCESS in 5m 48s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/0876d94a25bf44afb6473a528fb5d72c : SUCCESS in 1h 15m 00s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b9197746de754f37b58356c4fe90be54 : SUCCESS in 1h 34m 24s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d3dfb8f1c850472d860837e655a2dfe1 : SUCCESS in 53m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cbbaf4bbdb076e19bd078a4180fe0cf92afd178f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-24 16:59:08.000000000","message":"Patch Set 4:\n\n* brick-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-17-747417-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"3049e651318843f17c4542f4d0d6134038e5333b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-24 17:14:09.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/4/check/tempest-dsvm-full-bionic/900e8e6/ : SUCCESS in 1h 41m 16s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/17/747417/4/check/devstack-plugin-ceph-dsvm-bionic/9713dac/ : SUCCESS in 1h 46m 26s","accounts_in_message":[],"_revision_number":4},{"id":"13c157ba8cf78114fa28c4e877a2c14dec48f81a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-08-24 17:40:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/17/747417/4/check/DellEMC_PowerStore_FC-ob/0086383/DellEMC_PowerStore_FC-ob/32 : SUCCESS in 1h 06m 20s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/17/747417/4/check/DellEMC_PowerStore_iSCSI-ob/ec6fe77/DellEMC_PowerStore_iSCSI-ob/32 : SUCCESS in 1h 03m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"e46d08c12ee1cfd1d757c70986171d5876c20858","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-24 17:42:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/747417/4/upstream-solidfire-check/cinder-solidfire-iSCSI/e370df1/ : SUCCESS in 1h 03m 19s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"36e9684d87762e5f84e3963b3bc6bf0e4494bb10","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-24 17:43:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/17/747417/4/upstream-check/cinder-cDOT-iSCSI/d5b2ac6/ : SUCCESS in 36m 37s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/17/747417/4/upstream-check/cinder-cDOT-iSCSI-IPv6/293fc26/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/17/747417/4/upstream-check/cinder-cDOT-NFS/cf3d6f6/ : SUCCESS in 36m 18s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/17/747417/4/upstream-check/cinder-cDOT-NFS-IPv6/90c8726/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/17/747417/4/upstream-check/cinder-cDOT-FCP/f650e75/ : SUCCESS in 1h 06m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"34c156bc23809d528894f42972bf8515d891fc65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 17:43:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"641f0671131864a2b792ca05db325277706dc403","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 17:45:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/17/747417/4/check/DellEMC_PowerFlex_os-brick/d328f76/DellEMC_PowerFlex_os-brick/60 : SUCCESS in 1h 11m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"6cdc73427deb289813de08b39127bf53828c2414","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-24 20:24:01.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"e702fe283dbbce7b1c9443de9ef80ab082de7f62","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-25 02:15:29.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/17/747417/4/check/DellEMC_Unity_FC_os-brick/9bec333/DellEMC_Unity_FC_os-brick/67 : FAILURE in 2h 50m 41s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/17/747417/4/check/DellEMC_Unity_iSCSI_os-brick/9cd5de7/DellEMC_Unity_iSCSI_os-brick/64 : SUCCESS in 1h 46m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"592bf9fa3dc7b6a604248226f66c46e6ee340e8b","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2020-08-25 02:25:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3401b02fb8bddb9ce3219541d7f78526a1594006","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-25 03:35:04.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/747417/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"918f4c713ff4a3d6960d8a34b656a536f4a0e8c0","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-25 04:58:57.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nHi Eric,\nYou\u0027re right, this bug should have been fixed in 0.29.\nI downloaded lsscsi-0.30 and installed it on our ubuntu CI servers and the issue has gone.\nSo there\u0027s no need for this patch now unless some CI OS doesn\u0027t have lsscsi-0.29 or above, then we maybe consider to reopen this.\nThis change is just more fail-safe compared to the original one because it doesn\u0027t care about lsscsi versions.\nI\u0027ll close the bug with suggestion of upgrading lsscsi to 0.29 or above.\nAppreciate for your suggestion.\nthanks and regards\nSam","accounts_in_message":[],"_revision_number":4},{"id":"8468abfabba7de5585a2426b3e2c8c49e1a8db78","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-08-25 04:59:51.000000000","message":"Abandoned\n\nabondon this change because the bug has been fixed in lsscsi-0.29.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bc7342deba4a52f23af196acb14499871542998b","revisions":{"787f4c588f03c437cfefd1ce18262ac41a723492":{"kind":"REWORK","_number":1,"created":"2020-08-21 13:46:01.000000000","uploader":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"ref":"refs/changes/17/747417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/747417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/747417/1"}}},"commit":{"parents":[{"commit":"6672ea00233512e3aa9a397c02265f847024cd7c","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6672ea00233512e3aa9a397c02265f847024cd7c"}]}],"author":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-21 13:29:00.000000000","tz":-240},"committer":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-21 13:29:00.000000000","tz":-240},"subject":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi","message":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi\n\nCurrent get_device_info uses space as delimiter to get the 1st colume of\nlsscsi output as \u0027H:C:T:L\u0027 info. But if the hlu# \u003e\u003d 10000, the 1st and\n2nd columns are concated and then 1st column becomes \u0027[H:C:T:L]disk\u0027.\nWhen this happens, wrong hlu# is returned in dev_info.\n\nChange-Id: I35c54556ea58768d078ac732c9361e32d141a5dd\nCloses-bug: #1892500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/787f4c588f03c437cfefd1ce18262ac41a723492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/787f4c588f03c437cfefd1ce18262ac41a723492"}]},"branch":"refs/heads/master"},"7aa53d19b6972794daddfaa67c491af17ef950e0":{"kind":"REWORK","_number":2,"created":"2020-08-24 02:17:49.000000000","uploader":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"ref":"refs/changes/17/747417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/747417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/747417/2"}}},"commit":{"parents":[{"commit":"6672ea00233512e3aa9a397c02265f847024cd7c","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6672ea00233512e3aa9a397c02265f847024cd7c"}]}],"author":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-21 13:29:00.000000000","tz":-240},"committer":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-24 02:03:40.000000000","tz":-240},"subject":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi","message":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi\n\nCurrent get_device_info uses space as delimiter to get the 1st colume of\nlsscsi output as \u0027H:C:T:L\u0027 info. But if the hlu# \u003e\u003d 10000, the 1st and\n2nd columns are concated and then 1st column becomes \u0027[H:C:T:L]disk\u0027.\nWhen this happens, wrong hlu# is returned in dev_info.\n\nChange-Id: I35c54556ea58768d078ac732c9361e32d141a5dd\nCloses-bug: #1892500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7aa53d19b6972794daddfaa67c491af17ef950e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7aa53d19b6972794daddfaa67c491af17ef950e0"}]},"branch":"refs/heads/master"},"08f73c78012edf775b3c8d5d4f1dc42834c325b0":{"kind":"REWORK","_number":3,"created":"2020-08-24 10:41:45.000000000","uploader":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"ref":"refs/changes/17/747417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/747417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/747417/3"}}},"commit":{"parents":[{"commit":"6672ea00233512e3aa9a397c02265f847024cd7c","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6672ea00233512e3aa9a397c02265f847024cd7c"}]}],"author":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-21 13:29:00.000000000","tz":-240},"committer":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-24 10:27:54.000000000","tz":-240},"subject":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi","message":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi\n\nCurrent get_device_info uses space as delimiter to get the 1st colume of\nlsscsi output as \u0027H:C:T:L\u0027 info. But if the hlu# \u003e\u003d 10000, the 1st and\n2nd columns are concated and then 1st column becomes \u0027[H:C:T:L]disk\u0027.\nWhen this happens, wrong hlu# is returned in dev_info.\n\nChange-Id: I35c54556ea58768d078ac732c9361e32d141a5dd\nCloses-bug: #1892500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/08f73c78012edf775b3c8d5d4f1dc42834c325b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/08f73c78012edf775b3c8d5d4f1dc42834c325b0"}]},"branch":"refs/heads/master"},"bc7342deba4a52f23af196acb14499871542998b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-08-24 14:54:50.000000000","uploader":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"ref":"refs/changes/17/747417/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/747417/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/747417/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/747417/4"}}},"commit":{"parents":[{"commit":"6672ea00233512e3aa9a397c02265f847024cd7c","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6672ea00233512e3aa9a397c02265f847024cd7c"}]}],"author":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-21 13:29:00.000000000","tz":-240},"committer":{"name":"Sam Wan","email":"sam.wan@emc.com","date":"2020-08-24 14:40:23.000000000","tz":-240},"subject":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi","message":"Change delimiter to get \u0027H:C:T:L\u0027 from lsscsi\n\nChange delimiter from space to \u0027]\u0027 to get \u0027H:C:T:L\u0027 from lsscsi output.\n\nChange-Id: I35c54556ea58768d078ac732c9361e32d141a5dd\nCloses-bug: #1892500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bc7342deba4a52f23af196acb14499871542998b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bc7342deba4a52f23af196acb14499871542998b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
