)]}'
{"id":"openstack%2Fos-brick~748660","triplet_id":"openstack%2Fos-brick~stable%2Fussuri~I2775f55039695c7ec029106c0dafe4d46255b336","project":"openstack/os-brick","branch":"stable/ussuri","topic":"bug/1884114","hashtags":[],"change_id":"I2775f55039695c7ec029106c0dafe4d46255b336","subject":"rbd: Warn if ceph udev rules are not configured","status":"MERGED","created":"2020-08-28 12:44:35.000000000","updated":"2020-09-03 23:24:56.000000000","submitted":"2020-09-03 20:48:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748660-1599166095410-7e3b2230","meta_rev_id":"57e3da159e3f8e6c6d38d7e619358ea0bf10be57","_number":748660,"virtual_id_number":748660,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-03 17:46:22.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-09-03 19:37:24.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 20:48:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-03 20:28:57.000000000","_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"date":"2020-09-03 18:44:28.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-03 17:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-03 17:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-03 17:42:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-28 23:35:12.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-29 21:18:38.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-09-03 00:07:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 17:42:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-09-03 17:46:22.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 18:44:28.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-09-03 19:37:24.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 20:28:57.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-09-03 20:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42748e275fe38a0d1603d02a13f068c41b6e2874","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-28 12:44:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0a900826b3320a1d8ba77f98f5399b34da3e8b5","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-08-28 13:57:29.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : FAILURE in 1h 09m 33s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"238aef5119f1b3671b9a045ff649ed786a5753a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 14:00:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c026d397dab4725bb867d86339175df : SUCCESS in 1h 10m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c4bccb815934a529f1e3b5b100e0b0a : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3065604c2384e4d8deb4f8de36fe762 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5da57e493a604e1db809f962361d15a1 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76506c5dccd7470c98559e5e5fef8856 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da3918f463f4d20b5191836104a7bca : SUCCESS in 8m 38s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/4b0bec3b0cac4c32bae93ed6ea996ab7 : SUCCESS in 53m 10s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37ddb7a70f204a59921c3e7e099efeb7 : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b54fb8317700d45dbc39c795ebc507d08f6ee7d4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-08-28 14:05:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/748660/1/upstream-solidfire-check/cinder-solidfire-iSCSI/25075bf/ : FAILURE in 3m 53s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"ddfd570f3e518d12e40b2878fee6bdcc9791ae42","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-28 15:15:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/748660/1/upstream-check/cinder-cDOT-iSCSI/710996f/ : FAILURE in 5m 50s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/60/748660/1/upstream-check/cinder-cDOT-iSCSI-IPv6/aa791e8/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/748660/1/upstream-check/cinder-cDOT-NFS/5f261bd/ : FAILURE in 5m 27s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/60/748660/1/upstream-check/cinder-cDOT-NFS-IPv6/ef479fb/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/748660/1/upstream-check/cinder-cDOT-FCP/e4aa970/ : SUCCESS in 1h 14m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"16b4661f58d0649d1b4e78478811a9371cd1c519","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-28 18:32:56.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/tempest-dsvm-full-bionic/8c1da76/ : SUCCESS in 1h 43m 54s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/devstack-plugin-ceph-dsvm-bionic/7039a8a/ : SUCCESS in 1h 47m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ae3e86da4b3aaa9ecd92eadbba6b074164792571","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-28 20:27:24.000000000","message":"Patch Set 1: Code-Review+2\n\nLow risk and useful to operators.","accounts_in_message":[],"_revision_number":1},{"id":"3c2a0ae960f1c259123776ed5373df2e1f6a52b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-28 23:35:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/748660/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f75342eab4882205fdbbe1a152ef1068bf1b999c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-08-29 21:18:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cacc2fc670cc455f73d1a83a863311cd7e5adae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 21:18:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d78cce6e212e5070252a458987251f3aeac3c87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 00:08:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1990abb64b6042d38c344edc9c83ec8a : SUCCESS in 1h 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89147262c13c4ce6bd720af1603654dd : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e134943662f1470196a784b05b170527 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77c621f4f5284e2bb7400608397d9b20 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc29f41b63884e2d9ab76bc980a17c8e : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10cb93ca141a4e3a956de44227c51aeb : SUCCESS in 5m 09s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec8c68197fbf4fb4b33aa66d5f2d0dcb : FAILURE in 1h 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c4a9f10f925e520188edcd3deb048906c5ab2a40","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-31 10:19:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7dcd3e2004360a85fc70d2799eb62db53fee1477","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-08-31 10:43:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : FAILURE in 23m 38s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : FAILURE in 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b361a8d5d8e6571eceb2fc9534b25f0e113f34f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 10:49:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa1a72f103fb4dd8aded7ba2a2669cb2 : FAILURE in 17m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc9252643714bed88dbe70672a3b5fd : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb17f268d3a24537bf512f32b02e92df : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/719d6c10b80648ba904f8e374c851598 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7af4414adb3c4f87bd8e013450cbe936 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ae9acb6bd64b2b8644567e5d4724b4 : SUCCESS in 5m 38s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/d8a38ecc70fa4341b0e2a3b738bbc3c5 : FAILURE in 21m 46s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/be930b6a4b1f40bc8e1f7d55b65ef063 : FAILURE in 22m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0df530f107c2049e3979a1e49086b65a010fe41d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-31 11:05:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e34a5861fd88f1a1b59a1371ac38d8f003e5bf9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 11:33:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a01eea98102c40a3b4e59ad72d72c42f : FAILURE in 19m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11c530676f914706a9782e032b6e17ff : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db426c29e8fe40a4ab659484124d0a76 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efab87ad375a45269e44c949a9c18164 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7fa86ef8d81467cb73bbb5b6e57019c : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fa0837542a4aaab0c6e8e50b1afc1e : SUCCESS in 9m 14s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/2a6f82c6defa491b89f4ec30ee76e547 : FAILURE in 19m 43s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9693350a3f14b86a6e34879750cd3dd : FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d033b0d42dd264f1a81f55b5d95950ef391a8d4d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-31 11:36:11.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/tempest-dsvm-full-bionic/01396e8/ : FAILURE in 58m 29s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/devstack-plugin-ceph-dsvm-bionic/352fe1b/ : FAILURE in 1h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"de3e1837436c2f3069eb16a2eeca668ff54620f4","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-08-31 11:37:19.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : FAILURE in 28m 03s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : FAILURE in 25m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c6b3363fc68ed8908d05833ca1dbbacec0ca8416","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-31 16:57:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c6f35689792fcaccad683534867bfc4b17f018ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 17:18:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a329c72edfb4f268c70145f864d092a : FAILURE in 19m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14a20357c36b48da96918c78ea6ed854 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc8bdc5c6ad425cac6ae1680feef5f0 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed8c52f20ca74baa95c76335ed408ff6 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bb378881882424ab4b81089500dd07e : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2a91bef9aa447d9e97d5c1d2b471b1 : SUCCESS in 8m 01s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/99e2e968297345a89af99534db0c5fff : FAILURE in 17m 18s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a5fc03de4d7c4d32bdc55ebce1186b67 : FAILURE in 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b4e7b6d388c93d7032cf52af1fdeebcbfa735e1e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-31 19:45:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/tempest-dsvm-full-bionic/9a84560/ : FAILURE in 1h 07m 04s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/devstack-plugin-ceph-dsvm-bionic/6a9bffb/ : FAILURE in 1h 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"267fe801d641291d62fefbf978e1017d49b06548","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-01 08:29:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9213e7884070ccf3c68c633930298318833eb9a6","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-01 08:38:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/60/748660/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/5cccd21 : FAILURE in 4m 52s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/60/748660/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/c70f7de : FAILURE in 4m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"a585bbd9cf6897b9815afc19626d8a97740fbe70","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-09-01 08:59:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : FAILURE in 29m 52s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : FAILURE in 29m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f53fa726a3dce3ad92270dc3e52a6770e78c8244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 10:16:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91fc935e0de94dd6b133f12b3d3bbd92 : SUCCESS in 1h 11m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79c1dc5b18a3476db030a17da77aa510 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc96c7526a2649cda8dcddc319ab17a3 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0893cc42473e4ffc82588f42b0f1056b : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa6a87ebf9cb4ce68b4f907a577ee0a0 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9ef9261b0246ddb7c8e5043b228547 : SUCCESS in 7m 28s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/6183378dcb934a4cb765639f971bbf1e : FAILURE in 1h 05m 05s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2fe642eb50aa42cab0a17f6bee5e8eb1 : POST_FAILURE in 1h 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"30bd1abffe0c1035230e6b7f580866800020bfe5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-01 10:35:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/tempest-dsvm-full-bionic/ccfab83/ : SUCCESS in 1h 49m 55s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/devstack-plugin-ceph-dsvm-bionic/44aee1f/ : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ff9b4f026a086148cfd89cece398c2865e553af2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-01 15:36:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db28f71594abb444f14b80f9e2fe535f79063c57","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-01 15:46:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/60/748660/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/9613c45 : FAILURE in 4m 56s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/60/748660/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/46a82e8 : FAILURE in 5m 34s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"05cd46e6af8e9ec34bbdc841434fdd30ca7955a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 17:26:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c04598fc89c044ff809d62a9ddcf6efc : SUCCESS in 1h 26m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a974516646c049ac8962a2c94e5535e1 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2392e67bbb94235b1219abbc728fea6 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248c232fe9b24c518f3ca84cd9edb13c : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d4bc8bc066244bba227c844ce8a60a0 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b45834d45f4fbd9ae1e36eacd69cbd : SUCCESS in 5m 49s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/5d96e831d24a448091bd9bae1cb47f7c : FAILURE in 1h 05m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e943a7b3598147cbbefff1f52dd7a23a : POST_FAILURE in 1h 35m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d458bb5164de0b7cb781865bb505f8af33e3074e","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-09-01 18:39:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : FAILURE in 24m 16s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : FAILURE in 26m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a1ec21707da0c318f64ec5022cf8060a249bd8a1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-01 20:06:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d64095b6426d75c8a0b07dafcb5631278533da88","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-01 20:12:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/60/748660/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/c34148c : FAILURE in 4m 56s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/60/748660/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6889760 : FAILURE in 4m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6f84eaeefa97e7b2d1b054231d118131e7df350","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-09-01 20:31:35.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : FAILURE in 23m 49s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : FAILURE in 24m 24s","accounts_in_message":[],"_revision_number":1},{"id":"381dc8fbc6ad32326c836e88b82b7de9d70f2c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 21:44:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39824723a1ad41e9b8551dc39c9b21c7 : SUCCESS in 1h 21m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/151cc63a19ee4868888c77022952806f : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734787d6f4ae49e7a40a8f6eb9468910 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9667b62a9630468590e6384b9fc7c6f6 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d6f9d0dc9e9441b8e12c483c98876c0 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d7b1e84ae243ddabc9ce42728dd241 : SUCCESS in 5m 21s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/e0532d8d22f74e4e917e591157d99b0b : FAILURE in 1h 03m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/88996d94ffdc4cf7ad3d7381e2d8f5e5 : POST_FAILURE in 1h 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ea0cd2d883d4d0c06cd074c6e1ef92a0459cdd03","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-02 06:41:08.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/717b373d78ed4868b7708cfd81d4cd17 : POST_FAILURE in 1h 48m 03s\n- devstack-plugin-ceph-dsvm-bionic devstack-plugin-ceph-dsvm-bionic : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"7f141c10e7f0b815cf329d1ff4dfdb4d523a9f5d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-02 15:37:11.000000000","message":"Patch Set 1:\n\nrecheck\n\nman, this really is not happy","accounts_in_message":[],"_revision_number":1},{"id":"0430f3498214a6dd53ec3abe95a824b7245701d2","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-09-02 16:13:35.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : FAILURE in 34m 59s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : FAILURE in 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"43dff7f04f90936fca569fd22cdd20202a7d8ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 17:25:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90a504054f694afe85da73fed7553342 : SUCCESS in 1h 28m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e68c1f3b93b4394a5d5d45214d6150e : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe73625205a4487ea4cf64ef38aac897 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48650cb5fe774ae0b25f25eced28e02f : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80f934012bd04f0e8a12396176a3b955 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58ca6e6708c4009af843da94dc80b8b : SUCCESS in 7m 10s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/05e1675114ee4950ae001977b50a080f : FAILURE in 1h 11m 36s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7ae3591f79de4fd5b562d54cf6cbd45c : POST_FAILURE in 1h 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4d0f8b416d1d7975b70084397422dac19a420db5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-03 00:07:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/tempest-dsvm-full-bionic/0086359/ : SUCCESS in 1h 54m 58s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/1/check/devstack-plugin-ceph-dsvm-bionic/812f669/ : SUCCESS in 1h 53m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8d3704822361eadb465f5c41aeb715fa8cbffaa6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-03 15:56:31.000000000","message":"Patch Set 1:\n\nrecheck and you better work this time","accounts_in_message":[],"_revision_number":1},{"id":"1fc777cabbb3487d62e23c18842ebdeb74fc2e40","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-09-03 17:12:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/1/iscsi/ : SUCCESS in 1h 15m 10s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/1/smb/ : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"368dad6e959fc7371d1ada760a2add4702ca29ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:33:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81df699c37d34b5a9bf664f093b3c7c0 : SUCCESS in 1h 30m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62e7965f90c04e48a2d05fa3607a20c9 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7a7933e0e145fe97a902df18748aef : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39c5b16de0bc4e4d922d1f7f09660a27 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bd9edc693e4444ba39d282c101910f6 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d1cc30ea844b42bee14019e85dc541 : SUCCESS in 7m 14s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/fe426a01053f496ea77459950d7b295b : SUCCESS in 1h 01m 26s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98caee2a8fb04a04b0a65166bbf3d042 : FAILURE in 1h 27m 45s","accounts_in_message":[],"_revision_number":1},{"id":"badae91cc8872a0d3d318a709b5231abede7a87a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-03 17:41:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0cedcb209cb5c61dde0cb52111bd0b7778ec4a87","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-03 17:42:40.000000000","message":"Patch Set 2: Workflow+1\n\nI don\u0027t know what\u0027s going on, but other lio-barbican jobs have passed in stable/ussuri without seeing the setuptools problem.","accounts_in_message":[],"_revision_number":2},{"id":"744b3a7199806e07820a7de28cb8aa62e1d99d4f","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-03 17:46:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/60/748660/2/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/98d4f94 : FAILURE in 4m 58s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/60/748660/2/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/78c934f : FAILURE in 4m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":2},{"id":"685496b7fc3ad8527bca96f52bf16a10ef1fe074","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-09-03 18:44:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/748660/2/iscsi/ : SUCCESS in 1h 02m 58s\n- os-brick-smb http://cloudbase-ci.com/os-brick/748660/2/smb/ : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":2},{"id":"660f1bebc0fe1209007209bfcd608f34079ee72d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 18:59:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e390059469d44b9d9aa272261ae26d34 : SUCCESS in 1h 13m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ad39a3b9cc943848926b7db56ff14fe : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fbd873905fc42ab95a02accda349f57 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d91f8e1c485d4dd4af6277ecf0633e0e : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d86a1275ee34994b01ec819318a0f3f : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c2ca564aed44f885260d97556cefc8 : SUCCESS in 9m 48s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/6ea2ef2b0c0f4418b8c2267e0cdf3738 : SUCCESS in 1h 01m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/125884e1c16f40919884d911bcb74028 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":2},{"id":"37594d7f714a46e088823b8aa5419cde99d99ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 19:00:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"827eb7389a11486a00513b71c1a3c0391dafa222","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-03 19:37:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/748660/2/upstream-check/cinder-cDOT-iSCSI/422fd9b/ : SUCCESS in 37m 53s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/60/748660/2/upstream-check/cinder-cDOT-iSCSI-IPv6/c88aab4/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/748660/2/upstream-check/cinder-cDOT-NFS/e7458df/ : SUCCESS in 36m 41s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/60/748660/2/upstream-check/cinder-cDOT-NFS-IPv6/611b6e4/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/748660/2/upstream-check/cinder-cDOT-FCP/241d330/ : FAILURE in 9m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e0335ad5425d76ea5bd4155ea7e722e752dd8479","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-09-03 20:28:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/748660/2/upstream-solidfire-check/cinder-solidfire-iSCSI/0e27763/ : SUCCESS in 1h 29m 23s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"799691a21b547ad848b925e8bacca09c505f4fb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:48:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"948df2cf652cf5755a425024d1ce337bc9c05892","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:48:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3027217d1104d48aa6056d3d0aea770 : SUCCESS in 1h 18m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f344e376d6f04901aab57dea82fbf9f5 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c32130aa204450b948c365d8029bc92 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fa6625e7a5a45d686ef4a7a75188437 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/698851355d644d809a60ce48dd14601b : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb6499a4dcc54bd88378aac824350fa8 : SUCCESS in 5m 36s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f382faddd4a4b2b9fc7db9c6d0986a1 : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2d5d7e7858c63d4829a60b6644b842f3fba29358","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:05:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea58b2334dc4a508b4e4227d64b94fd : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2},{"id":"3f337f6164813f48a19e7e916ebca2ad2286611b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-03 23:24:56.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/2/check/tempest-dsvm-full-bionic/5c8b4fc/ : SUCCESS in 1h 51m 54s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/60/748660/2/check/devstack-plugin-ceph-dsvm-bionic/501a187/ : SUCCESS in 2h 00m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d3ee7a076be6d3397946f17f15bee4bda2968459","revisions":{"1eeffd986dd8d5a192c7af272fb5eefb0ce43da2":{"kind":"REWORK","_number":1,"created":"2020-08-28 12:44:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/748660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/60/748660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/748660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/748660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/748660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/60/748660/1"}}},"commit":{"parents":[{"commit":"65b7517db2823a267155c81b2ba8e2da1deb6acc","subject":"Merge \"iSCSI detect multipath DM with no WWN\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/65b7517db2823a267155c81b2ba8e2da1deb6acc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-22 10:07:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-08-28 12:44:15.000000000","tz":60},"subject":"rbd: Warn if ceph udev rules are not configured","message":"rbd: Warn if ceph udev rules are not configured\n\nThe LUKS encryptor feature expects devices to have a symbolic link that\nit can overwrite in order to enable transparent encryption/decryption\nfor instances [1]. This is generally the case for RBD volumes, as Ceph\nuses udev rules [2] to create a \u0027/dev/rbd/{pool}/{device}\u0027 -\u003e\n\u0027/dev/rbdN\u0027 symlink. However, in an environment where udev daemon is not\npresent or configured correctly, this symlink will never be configured.\nThis causes things to crash and burn in a rather non-obvious manner when\nlocally attaching an encrypted RBD volume:\n\n  oslo_concurrency.processutils.ProcessExecutionError: Unexpected error while running command.\n  Command: cryptsetup luksOpen --key-file\u003d- /dev/rbd/volumes/volume-foo crypt-volume-foo\n  Exit code: 4\n  Stdout: \u0027\u0027\n  Stderr: \"Device /dev/rbd/volumes/foo doesn\u0027t exist or access denied.\\n\"\n\n(\u0027foo\u0027 being a stand-in for a very long \u0027device-$UUID\u0027 name)\n\nThe long term fix here is to probably stop relying on the side effects\nof these udev rules, i.e. the symlinks, but that is a far more involved\nfix that would not be backportable. Instead, for now we simply leave a\nbreadcrumb for the user, informing them as to what\u0027s gone wrong and\nencouraging them to look at the bug report for more information.\n\n[1] https://github.com/openstack/os-brick/blob/3.1.0/os_brick/encryptors/luks.py#L191-L195\n[2] https://github.com/ceph/ceph/blob/v14.0.0/udev/50-rbd.rules\n\nChange-Id: I2775f55039695c7ec029106c0dafe4d46255b336\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nRelated-Bug: #1884114\n(cherry picked from commit ee34d925ff8a8a83345941b7876b09f2c0396864)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1eeffd986dd8d5a192c7af272fb5eefb0ce43da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1eeffd986dd8d5a192c7af272fb5eefb0ce43da2"}]},"branch":"refs/heads/stable/ussuri"},"d3ee7a076be6d3397946f17f15bee4bda2968459":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-03 17:41:09.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/60/748660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/60/748660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/748660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/748660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/748660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/60/748660/2"}}},"commit":{"parents":[{"commit":"86b2eb3dd4efe14919f003757200aad405dbad04","subject":"Merge \"Leverage the iSCSI mpath to get the WWN\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/86b2eb3dd4efe14919f003757200aad405dbad04"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-22 10:07:19.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-09-03 17:41:09.000000000","tz":0},"subject":"rbd: Warn if ceph udev rules are not configured","message":"rbd: Warn if ceph udev rules are not configured\n\nThe LUKS encryptor feature expects devices to have a symbolic link that\nit can overwrite in order to enable transparent encryption/decryption\nfor instances [1]. This is generally the case for RBD volumes, as Ceph\nuses udev rules [2] to create a \u0027/dev/rbd/{pool}/{device}\u0027 -\u003e\n\u0027/dev/rbdN\u0027 symlink. However, in an environment where udev daemon is not\npresent or configured correctly, this symlink will never be configured.\nThis causes things to crash and burn in a rather non-obvious manner when\nlocally attaching an encrypted RBD volume:\n\n  oslo_concurrency.processutils.ProcessExecutionError: Unexpected error while running command.\n  Command: cryptsetup luksOpen --key-file\u003d- /dev/rbd/volumes/volume-foo crypt-volume-foo\n  Exit code: 4\n  Stdout: \u0027\u0027\n  Stderr: \"Device /dev/rbd/volumes/foo doesn\u0027t exist or access denied.\\n\"\n\n(\u0027foo\u0027 being a stand-in for a very long \u0027device-$UUID\u0027 name)\n\nThe long term fix here is to probably stop relying on the side effects\nof these udev rules, i.e. the symlinks, but that is a far more involved\nfix that would not be backportable. Instead, for now we simply leave a\nbreadcrumb for the user, informing them as to what\u0027s gone wrong and\nencouraging them to look at the bug report for more information.\n\n[1] https://github.com/openstack/os-brick/blob/3.1.0/os_brick/encryptors/luks.py#L191-L195\n[2] https://github.com/ceph/ceph/blob/v14.0.0/udev/50-rbd.rules\n\nChange-Id: I2775f55039695c7ec029106c0dafe4d46255b336\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nRelated-Bug: #1884114\n(cherry picked from commit ee34d925ff8a8a83345941b7876b09f2c0396864)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d3ee7a076be6d3397946f17f15bee4bda2968459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d3ee7a076be6d3397946f17f15bee4bda2968459"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
